)]}'
{"id":"openstack%2Fneutron-tempest-plugin~836455","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I6e0a805520948086379f0e5537e12dedcd093e50","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-04-05 11:17:22.000000000","reason":"Rodolfo Alonso replied on the change"},"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2022-04-06 09:47:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6e0a805520948086379f0e5537e12dedcd093e50","subject":"[sfc] Remove unused check_remote_connectivity methods","status":"MERGED","created":"2022-04-05 08:51:47.000000000","updated":"2022-04-06 09:47:50.000000000","submitted":"2022-04-06 09:47:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836455-tempest-scenario-manager-cleanup","meta_rev_id":"0f5b33ece7e568bb66e26c3a5434730f12abab99","_number":836455,"virtual_id_number":836455,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-06 09:47:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-04-05 11:17:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-04-06 08:16:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-04-06 08:16:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 08:55:49.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-04-05 10:34:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-06 08:16:26.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"c714bf2ecce23c7ac4f144ed455b6ed071aa4d75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-05 08:51:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d23d8ace1655b08a05645fb6ee214457b7aad4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 10:34:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730e4c6600d2443f90ece77c380a25fd : SUCCESS in 3m 52s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/01c508af7ccf45898993937349290b7b : SUCCESS in 56m 38s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/1341e94101094ac18808c740f6a443e0 : SUCCESS in 52m 25s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/34e0b9f0816f4af2a08c306b24b3f0ba : SUCCESS in 52m 41s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/30036d93c9e640dba0d58fe2d4fdb921 : SUCCESS in 53m 54s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/7ac59725fbf64a8f9818be0e86788606 : SUCCESS in 46m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8b3f980385d4236ad886065d2efedd9 : SUCCESS in 3m 44s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/c297981734c743dbabf9301d37abd8c5 : SUCCESS in 1h 35m 11s\n- neutron-tempest-plugin-sfc-victoria https://zuul.opendev.org/t/openstack/build/3bc4a0b33fc84af0b185fa4ec0550bcd : SUCCESS in 1h 29m 16s\n- neutron-tempest-plugin-sfc-wallaby https://zuul.opendev.org/t/openstack/build/293cbae3eba74a8eb75ccd1304d62f01 : SUCCESS in 1h 09m 15s\n- neutron-tempest-plugin-sfc-xena https://zuul.opendev.org/t/openstack/build/d17e095499f745ac98efe8aa7ab31fe1 : SUCCESS in 1h 00m 49s\n- neutron-tempest-plugin-sfc-yoga https://zuul.opendev.org/t/openstack/build/d6e8a9e3cd2f470abab3e651d9cfc7f7 : SUCCESS in 1h 08m 15s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/0602960775754de695fcf79ecd5a051d : SUCCESS in 22m 19s","accounts_in_message":[],"_revision_number":1},{"id":"10e44889fdd8e2df12e27f5b110f9692ea137f00","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-05 11:17:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3de307a620ec5cd1e7ec9820cc383dceb28b2b36","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-06 08:16:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f18c3c8a026a5676c0e4ef6073d712d7bd9196f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 08:17:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a74a2e11de2d83ef8f8c36cdd81f3675e0beaf8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 09:47:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5185565d05944779f047d205de01632 : SUCCESS in 4m 12s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/83fed18323e34bd4b15a004411ee3350 : SUCCESS in 36m 38s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/91d6af84bc4a4ba0a7040dc40db673e7 : SUCCESS in 42m 02s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/e679fb38b81745b59d636dc3b305ee16 : SUCCESS in 43m 44s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/abfa73663e2a44408e09a87f8db89859 : SUCCESS in 37m 33s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/a361742742664c57b036e0009b01cf8a : SUCCESS in 40m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68c92e6ccab24c36884e4b5b3c38017a : SUCCESS in 4m 03s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/0109dc117d7446c59c62973a8b8e1314 : SUCCESS in 1h 15m 58s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/a78f7a75ca424507a61eb0842f4226de : SUCCESS in 28m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0f5b33ece7e568bb66e26c3a5434730f12abab99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 09:47:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ee1c152f5c374a171bee9ee0e3493e26c314dda","revisions":{"1ee1c152f5c374a171bee9ee0e3493e26c314dda":{"kind":"REWORK","_number":1,"created":"2022-04-05 08:51:47.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/55/836455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/55/836455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/55/836455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/55/836455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/55/836455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/55/836455/1"}}},"commit":{"parents":[{"commit":"f0e89f4a5303625f23b9534fb0b8d62a10abb5db","subject":"[sfc] Remove unused floating ip methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f0e89f4a5303625f23b9534fb0b8d62a10abb5db"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-05 08:50:34.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-05 08:50:34.000000000","tz":120},"subject":"[sfc] Remove unused check_remote_connectivity methods","message":"[sfc] Remove unused check_remote_connectivity methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest\n27.0.0[1] it can be now reused in plugins.\n\nRemoved methods:\n    * _check_remote_connectivity\n    * check_remote_connectivity\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I6e0a805520948086379f0e5537e12dedcd093e50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1ee1c152f5c374a171bee9ee0e3493e26c314dda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1ee1c152f5c374a171bee9ee0e3493e26c314dda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
