)]}'
{"id":"openstack%2Fneutron-tempest-plugin~836491","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I44d2ebf2bd74e42474774c4ff72e3a3021a82983","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2022-04-13 06:20:25.000000000","reason":"Removed by Roman Popelka by clicking the attention icon"}},"hashtags":[],"change_id":"I44d2ebf2bd74e42474774c4ff72e3a3021a82983","subject":"[sfc] Remove _create_port \u0026 create_keypair","status":"MERGED","created":"2022-04-05 09:11:04.000000000","updated":"2022-04-22 11:09:43.000000000","submitted":"2022-04-22 11:09:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836491-tempest-scenario-manager-cleanup","meta_rev_id":"d3383d41fb62179a119bc2602e92644729bfe99c","_number":836491,"virtual_id_number":836491,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-22 11:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-04-07 08:34:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-04-11 11:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-04-11 11:11:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 10:21:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-07 08:34:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-04-11 11:11:12.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"c8600a6d5c7d081283a7e70553d7fdbc6ba7c6a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-05 09:11:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c25316c71ae98f3a905140168c35667080adfe8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 10:21:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c54107cfbd42ca8d530eb468ae6618 : SUCCESS in 5m 43s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/85cd9f9590d741448a5d2287d9df402d : SUCCESS in 46m 15s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/e22d69d5586f4891a82ac64bc50a8c03 : SUCCESS in 46m 24s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/ec0b57e03141456985c2b08af227515f : SUCCESS in 46m 04s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/23bd750535fe45228360719d5a3c57b9 : SUCCESS in 43m 29s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/148ebbe2ac65451f8bd3d347fe032677 : SUCCESS in 49m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19586d4d0abc45d88a185f57dc599542 : SUCCESS in 6m 44s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/5768fb81329b41289c0f6121dae53804 : FAILURE in 28m 15s\n- neutron-tempest-plugin-sfc-victoria https://zuul.opendev.org/t/openstack/build/3589203209844fb788f5016bffa05eed : FAILURE in 42m 34s\n- neutron-tempest-plugin-sfc-wallaby https://zuul.opendev.org/t/openstack/build/5da0384fa0db48d1ae32508bfe14c269 : FAILURE in 34m 16s\n- neutron-tempest-plugin-sfc-xena https://zuul.opendev.org/t/openstack/build/ad0b8656d9cf40a0bf00b85eedd5bf55 : FAILURE in 28m 02s\n- neutron-tempest-plugin-sfc-yoga https://zuul.opendev.org/t/openstack/build/052cbfad5bc84733ad72f66e40a09cc8 : FAILURE in 33m 14s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/c6446fe3b3c84fbfab706eea56755f38 : SUCCESS in 32m 26s","accounts_in_message":[],"_revision_number":1},{"id":"aca8f11ec96d6555c789c39f197f867d54edc22e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-05 10:32:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4ef5522c1220c6fc932fa30d51db05a965aaebb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 12:24:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe65912d93844a6cb4ac107309c4f37e : SUCCESS in 4m 40s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b79a49748a7d4d21a31a94e350b50a66 : SUCCESS in 42m 46s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/ac21376d99bb4e0eb3d86a1e5f0f9fa8 : SUCCESS in 44m 40s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/d42bc228fbc54d84b3ff08b6d1183ded : SUCCESS in 43m 25s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/a044ff623fb549a398254ae776ce47be : SUCCESS in 47m 51s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/157765faa4ab4c919d9379702d650619 : SUCCESS in 46m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d083417277884d7a85fef756c4fd26d1 : SUCCESS in 4m 34s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/11f6e5b27a8d483a94249e6fbac0b74a : SUCCESS in 1h 29m 36s\n- neutron-tempest-plugin-sfc-victoria https://zuul.opendev.org/t/openstack/build/5e7a164032364d7788bedbcb501c7904 : SUCCESS in 1h 19m 09s\n- neutron-tempest-plugin-sfc-wallaby https://zuul.opendev.org/t/openstack/build/53a920a00a9d4f319c7bd1d7baed2735 : SUCCESS in 1h 45m 14s\n- neutron-tempest-plugin-sfc-xena https://zuul.opendev.org/t/openstack/build/f328a02a94394afeb1f86457612929bb : SUCCESS in 1h 22m 04s\n- neutron-tempest-plugin-sfc-yoga https://zuul.opendev.org/t/openstack/build/0f27ec5504dd41e688b1b0261584e251 : SUCCESS in 1h 30m 42s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/b55a7a1a3ac64eb78422266d0da055b3 : SUCCESS in 30m 27s","accounts_in_message":[],"_revision_number":2},{"id":"de23aca50f0c7ad6421baedce7497df5bdaeb6bc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-07 08:34:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"83d52f1f7318f4d1de905f2dfd92be5b2fda9f2b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-11 11:11:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a9843cd8aa9ece037412643abdf7af1467441d01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 09:47:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cc4f191ab39e1c6ab0c056669e3332addd329a34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 11:09:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f74876e6e32491b91e1dc661d989444 : SUCCESS in 4m 09s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c5846896fa4f4fc798eee6bcffe9403e : SUCCESS in 44m 32s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/d62ae7dce7cd40feaf4e656bf17520b8 : SUCCESS in 47m 41s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/6d286bd66e7f47738d1a33b9f110ba6a : SUCCESS in 46m 37s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/ff7a9f3d3b9f40b897e9608258d2f0f6 : SUCCESS in 44m 04s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/0f08193ea31748978e55b9168091790e : SUCCESS in 41m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55a8f18c452b4f7099eb1428e432626c : SUCCESS in 4m 29s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/6eddb1f43db64fe5814c35ff01c63f53 : SUCCESS in 1h 17m 13s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/6504be47bae34da08387c1279b3b48fb : SUCCESS in 29m 29s","accounts_in_message":[],"_revision_number":2},{"id":"d3383d41fb62179a119bc2602e92644729bfe99c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 11:09:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e3c390835af2a097e4d6f5113387cee4f1a402d4","revisions":{"5c4b01814b5299eb92e1ca1417d0fa2f21751a80":{"kind":"REWORK","_number":1,"created":"2022-04-05 09:11:04.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/91/836491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/91/836491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/836491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/836491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/836491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/836491/1"}}},"commit":{"parents":[{"commit":"fb2dd32da43b08cd1c9b86212ee973ba2ccd8d00","subject":"[sfc] Remove unused router related methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fb2dd32da43b08cd1c9b86212ee973ba2ccd8d00"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-05 09:09:27.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-05 09:09:27.000000000","tz":120},"subject":"[sfc] Remove _create_port \u0026 create_keypair","message":"[sfc] Remove _create_port \u0026 create_keypair\n\nAs tempest.scenario.manager was announced stable interface in Tempest\n27.0.0[1] it can be now reused in plugins.\n\nRemoved methods:\n   * _create_port\n   * create_keypair\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I44d2ebf2bd74e42474774c4ff72e3a3021a82983\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5c4b01814b5299eb92e1ca1417d0fa2f21751a80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5c4b01814b5299eb92e1ca1417d0fa2f21751a80"}]},"branch":"refs/heads/master"},"e3c390835af2a097e4d6f5113387cee4f1a402d4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-05 10:32:27.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/91/836491/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/91/836491/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/836491/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/836491/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/836491/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/91/836491/2"}}},"commit":{"parents":[{"commit":"e800f7be3f204a98f2ac7adc51b2f750faa3c661","subject":"[sfc] Remove unused router related methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e800f7be3f204a98f2ac7adc51b2f750faa3c661"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-05 09:09:27.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-05 10:32:27.000000000","tz":0},"subject":"[sfc] Remove _create_port \u0026 create_keypair","message":"[sfc] Remove _create_port \u0026 create_keypair\n\nAs tempest.scenario.manager was announced stable interface in Tempest\n27.0.0[1] it can be now reused in plugins.\n\nRemoved methods:\n   * _create_port\n   * create_keypair\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I44d2ebf2bd74e42474774c4ff72e3a3021a82983\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e3c390835af2a097e4d6f5113387cee4f1a402d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e3c390835af2a097e4d6f5113387cee4f1a402d4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
