)]}'
{"id":"openstack%2Fneutron-tempest-plugin~836582","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ib01030cf424f96fbcaaa173f0f755e80057c2cdf","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-05-03 15:08:15.000000000","reason":"Rodolfo Alonso replied on the change"},"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2022-05-03 16:35:21.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-05-03 16:35:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib01030cf424f96fbcaaa173f0f755e80057c2cdf","subject":"[sfc] Remove _get_network_by_name \u0026 check_floating_ip_status","status":"MERGED","created":"2022-04-05 09:30:45.000000000","updated":"2022-05-03 16:35:21.000000000","submitted":"2022-05-03 16:35:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"836582-tempest-scenario-manager-cleanup","meta_rev_id":"b6ac08701f185afc82a5811271e4d628e5cadc3a","_number":836582,"virtual_id_number":836582,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-03 16:35:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-05-03 15:08:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-05-03 15:01:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-05-03 15:08:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 10:01:25.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-04-05 10:31:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-03 15:01:13.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"f9fa0a8de52680b3d59beba6cd9569eea2b2e208","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-05 09:30:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"646113337810bc7a5fd260febacd807e3adc5a64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 10:31:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6441cba6583a412ab309a48d0f9ac4a5 : SUCCESS in 5m 47s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c9409eb3cbf34af68ecef5a82e85d251 : SUCCESS in 43m 47s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/cfad060504da42398907834e88654e1e : SUCCESS in 49m 00s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/1f341bf80c4d4dffac6920cea2c856ef : SUCCESS in 39m 11s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/ab05041586694a76be375a1f57bb3da6 : SUCCESS in 39m 32s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/13f7032e469742e9b041b2e625ffb385 : SUCCESS in 42m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b8365a2ed734b6c8e4e046ced1e3db1 : SUCCESS in 4m 56s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/0966b4fba31346e1b3ce3ad80dc57d4c : FAILURE in 31m 42s\n- neutron-tempest-plugin-sfc-victoria https://zuul.opendev.org/t/openstack/build/5bc981d816ea4634b269c5fd45b92ba0 : FAILURE in 35m 12s\n- neutron-tempest-plugin-sfc-wallaby https://zuul.opendev.org/t/openstack/build/7a200bc88b434a1890c8eda22346f05c : FAILURE in 33m 41s\n- neutron-tempest-plugin-sfc-xena https://zuul.opendev.org/t/openstack/build/d8d6dd20037644f494a223176b020397 : FAILURE in 39m 56s\n- neutron-tempest-plugin-sfc-yoga https://zuul.opendev.org/t/openstack/build/de57d43c96994c5c9ed77a73c040ea4f : FAILURE in 32m 40s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/979820c6db334c65ae8683f8922340b1 : SUCCESS in 35m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9beb7586e824f1202eabce0c0a4d127ceeb9d97b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-05 10:32:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f021f1ff10390b32807985fb7de07d0b0c478c84","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-05 11:24:50.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa7f230d3016f42e007d685a0c8e53ba7d84582b","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-05 12:12:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9ea0e5129a2fe1a1fae692341c734d4fa3dcc6db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 12:17:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90fe8652e89c4da88c064a5052a856c5 : SUCCESS in 4m 43s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d7fd5f018ebc4f8e89c16cdf7e02af7e : SUCCESS in 36m 39s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/c64e3f3048894fc0ae2c75552e6b6596 : SUCCESS in 40m 46s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/77a9aaacf812403988f539bba735ccba : SUCCESS in 38m 06s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/c2de5c5e23aa447fa75b0263e7a8c56f : SUCCESS in 39m 12s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/2c87de9cae884dd88c922798a6ba276b : SUCCESS in 39m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a28c064cb564d06b2111d079351795b : SUCCESS in 4m 39s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/18502af5d4904933b40af30ef01d76da : SUCCESS in 1h 12m 17s\n- neutron-tempest-plugin-sfc-victoria https://zuul.opendev.org/t/openstack/build/29269e446a2d463a930d47a948cc132e : SUCCESS in 1h 04m 31s\n- neutron-tempest-plugin-sfc-wallaby https://zuul.opendev.org/t/openstack/build/d9d2101131644da389918287fea436bd : SUCCESS in 1h 26m 07s\n- neutron-tempest-plugin-sfc-xena https://zuul.opendev.org/t/openstack/build/d2ba35bc880b499e93f3e27d485e1806 : SUCCESS in 1h 02m 32s\n- neutron-tempest-plugin-sfc-yoga https://zuul.opendev.org/t/openstack/build/c04ea95a7ba245d7a261443ea81cd63d : SUCCESS in 1h 36m 05s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/72b4e48c8f4b4ef28f3b178ec326aea3 : SUCCESS in 37m 30s","accounts_in_message":[],"_revision_number":2},{"id":"f2667e9d154c39b92ba6291bb94f75aa7f295a09","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-03 15:01:13.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fc86a12562b8f0cccb5ff86e6b5932c74aba76c4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-03 15:08:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6a46f8e437cf810a4f4ef2ad03c487a283e70e85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 15:08:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7665d9705ebba479664427934131a4e9c5ec4dd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 16:35:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ffdfbb4be77471da90e5c279d9e26f2 : SUCCESS in 3m 41s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/33fb01dd4da94447bcb2283cc352f5ea : SUCCESS in 42m 17s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/4d86b9d1501a43a8aa9e00cea280018a : SUCCESS in 45m 37s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/f4ad418f58e84f51a49ff59dec98313e : SUCCESS in 51m 32s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/06fe553237ec40b08cd20127700ced43 : SUCCESS in 48m 23s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/8d7fd98f2c72485cb59b829441cf9b2f : SUCCESS in 45m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8fc32ca06b94601938d9f4fec6fc500 : SUCCESS in 3m 47s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/98ecc491649f4ca8842a690617ff7212 : SUCCESS in 1h 22m 18s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/335163a1b48d41c3b92c927e0bd8224d : SUCCESS in 38m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b6ac08701f185afc82a5811271e4d628e5cadc3a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 16:35:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9aa4811fb1e16b4f3c7e18d974dfd028d90be940","revisions":{"2ebc4fe777b73663bb49ab7d62c6c2b990b0a0db":{"kind":"REWORK","_number":1,"created":"2022-04-05 09:30:45.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/82/836582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/82/836582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/82/836582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/82/836582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/82/836582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/82/836582/1"}}},"commit":{"parents":[{"commit":"cca21179540e6897a1ccb4a97e724bda216cdd46","subject":"[sfc] Remove _log_console_output \u0026 ping_ip_address","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cca21179540e6897a1ccb4a97e724bda216cdd46"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-05 09:28:06.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-05 09:28:06.000000000","tz":120},"subject":"[sfc] Remove _get_network_by_name \u0026 check_floating_ip_status","message":"[sfc] Remove _get_network_by_name \u0026 check_floating_ip_status\n\nAs tempest.scenario.manager was announced stable interface in Tempest\n27.0.0[1] it can be now reused in plugins.\n\nRemoved methods:\n    * _get_network_by_name\n    * check_floating_ip_status\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ib01030cf424f96fbcaaa173f0f755e80057c2cdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2ebc4fe777b73663bb49ab7d62c6c2b990b0a0db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2ebc4fe777b73663bb49ab7d62c6c2b990b0a0db"}]},"branch":"refs/heads/master"},"9aa4811fb1e16b4f3c7e18d974dfd028d90be940":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-05 10:32:48.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/82/836582/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/82/836582/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/82/836582/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/82/836582/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/82/836582/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/82/836582/2"}}},"commit":{"parents":[{"commit":"96e90907b6a019574bc1714b776dee5acc2f38cf","subject":"[sfc] Remove _log_console_output \u0026 ping_ip_address","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/96e90907b6a019574bc1714b776dee5acc2f38cf"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-05 09:28:06.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-05 10:32:48.000000000","tz":0},"subject":"[sfc] Remove _get_network_by_name \u0026 check_floating_ip_status","message":"[sfc] Remove _get_network_by_name \u0026 check_floating_ip_status\n\nAs tempest.scenario.manager was announced stable interface in Tempest\n27.0.0[1] it can be now reused in plugins.\n\nRemoved methods:\n    * _get_network_by_name\n    * check_floating_ip_status\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ib01030cf424f96fbcaaa173f0f755e80057c2cdf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9aa4811fb1e16b4f3c7e18d974dfd028d90be940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9aa4811fb1e16b4f3c7e18d974dfd028d90be940"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
