)]}'
{"id":"openstack%2Fneutron-tempest-plugin~837511","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ia2f8eed6da612bc88c9b9eb48873dcc11b309f42","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2022-05-03 00:11:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia2f8eed6da612bc88c9b9eb48873dcc11b309f42","subject":"[bgpvpn] Make _create_loginable_secgroup_rule public","status":"MERGED","created":"2022-04-12 09:33:45.000000000","updated":"2022-05-03 00:11:50.000000000","submitted":"2022-05-03 00:11:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837511-tempest-scenario-manager-cleanup","meta_rev_id":"a17d970302732aa0676e0f3a18dc2459fcbc44c9","_number":837511,"virtual_id_number":837511,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-03 00:11:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2022-05-02 22:36:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2022-05-02 11:38:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2022-05-02 22:36:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-12 13:23:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-02 11:38:58.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-05-02 22:36:14.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"ef2a970cc715e37ae4960bc497d66ad0bdc49f99","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-12 09:33:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19fa69d6305ac8766de0f9af0b97ea7855251c0f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-12 09:36:27.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5756360d7ed4580a043d2f71cb34dcdcbe220a17","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-12 10:17:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1057ac3b5b4235bde8f170e505aab4c0374fc50f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 13:23:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/465164df21a24c79a40b32225a01cbfd : SUCCESS in 4m 17s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/2e7e1ff18e6f42faa4dc34d68e7f100b : SUCCESS in 50m 11s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/551e8acc05d447ac95a5a28588708e78 : SUCCESS in 43m 32s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/07a19c0c14f34793b36a9be85a3f10ba : SUCCESS in 59m 31s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/961ca7bd10cc4f9cb8a43bef84232806 : SUCCESS in 41m 23s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/c5cf46e05ae44800963b5a4f413b894a : SUCCESS in 39m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b032e8d96b041288820de36cb64b16c : SUCCESS in 3m 40s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/0283bad92094437db48b1c80fe642f73 : SUCCESS in 1h 16m 26s\n- neutron-tempest-plugin-bgpvpn-bagpipe-victoria https://zuul.opendev.org/t/openstack/build/c977d034aeeb43c486e8f2601ce72118 : SUCCESS in 1h 15m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe-wallaby https://zuul.opendev.org/t/openstack/build/fe0d1cc17eac45c387ddeed980f7d736 : SUCCESS in 1h 21m 39s\n- neutron-tempest-plugin-bgpvpn-bagpipe-xena https://zuul.opendev.org/t/openstack/build/c1292cdf98ad418f967faa504ac52214 : SUCCESS in 1h 10m 55s\n- neutron-tempest-plugin-bgpvpn-bagpipe-yoga https://zuul.opendev.org/t/openstack/build/6fc5992e3b5b4c3094590592a7f81737 : SUCCESS in 1h 04m 54s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/5db469b8827c40b39e87b23311a854a3 : SUCCESS in 35m 17s","accounts_in_message":[],"_revision_number":3},{"id":"b241cda5aaef628a948321fd8edd42abc7c211ee","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-20 07:40:15.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"192a361522c6ba4eefc94a4057ba5082b6f3b26a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-05-02 11:38:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8efd41b485c5f59e4975ed4d8f0fef213def043d","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-05-02 22:36:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"eb452d22c36d7af96cc68b6fc9c70bf9a2295259","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 22:38:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2ca487fa750a70eb84f0f5f51cf00fbf3f35b460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 00:11:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad181df404e4208b766b5faa08dc633 : SUCCESS in 5m 12s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/346c28c1e9da4d4fa68a36272773ba3f : SUCCESS in 1h 01m 55s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/78c5e369469a4db985209c1b0e2e465f : SUCCESS in 55m 05s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/0c722c84aab74d8f8163a5a29e57f496 : SUCCESS in 53m 23s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/7383860dafd34e9f9b4b120f1f043947 : SUCCESS in 43m 26s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/e4e3990664644f36915794f7b9f1e1c1 : SUCCESS in 44m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dd62ea332054ddfb0468043fd418934 : SUCCESS in 3m 34s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/593702b7adc04922af9e5652b81962a8 : SUCCESS in 1h 28m 50s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/3ab046f1c657401f8140c82bc0d543e7 : SUCCESS in 25m 16s","accounts_in_message":[],"_revision_number":3},{"id":"a17d970302732aa0676e0f3a18dc2459fcbc44c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 00:11:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8235e3b5198c3af40cf2c7c797c8beb028814e80","revisions":{"d79c7ac15f2201e4bc3fb4bc0db6b75a3aee1551":{"kind":"REWORK","_number":1,"created":"2022-04-12 09:33:45.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/11/837511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/11/837511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/837511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/837511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/837511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/837511/1"}}},"commit":{"parents":[{"commit":"40b7c5ec74c7bb63eeeae0b2fb4bc5f2776ff021","subject":"[bgpvpn] Remove _get_server_port_id_and_ip4 \u0026 check_floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/40b7c5ec74c7bb63eeeae0b2fb4bc5f2776ff021"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-12 09:29:17.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-12 09:29:17.000000000","tz":120},"subject":"Make _create_loginable_secgroup_rule public","message":"Make _create_loginable_secgroup_rule public\n\nAs tempest.scenario.manager was announced stable interface in Tempest\n27.0.0[1] it can be now reused in plugins.\n\n_create_loginable_secgroup method isn\u0027t ready to be deleted yet, because its rulesets differs from tempest\u0027s.\nThis method will be public now, so replaced methods will use this method over the tempest\u0027s one.\nTempest didn\u0027t allow to modify rulesets, but patch that allows this is already merged[2].\nIn some time, this plugin\u0027s implementation will be replaced\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n[2] https://review.opendev.org/c/openstack/tempest/+/835018\n\nChange-Id: Ia2f8eed6da612bc88c9b9eb48873dcc11b309f42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d79c7ac15f2201e4bc3fb4bc0db6b75a3aee1551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d79c7ac15f2201e4bc3fb4bc0db6b75a3aee1551"}]},"branch":"refs/heads/master"},"5388796030f0983f5be29bf1d9f4cb71c7958e80":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-04-12 09:36:27.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/11/837511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/11/837511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/837511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/837511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/837511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/837511/2"}}},"commit":{"parents":[{"commit":"40b7c5ec74c7bb63eeeae0b2fb4bc5f2776ff021","subject":"[bgpvpn] Remove _get_server_port_id_and_ip4 \u0026 check_floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/40b7c5ec74c7bb63eeeae0b2fb4bc5f2776ff021"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-12 09:29:17.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-12 09:36:04.000000000","tz":120},"subject":"[bgpvpn] Make _create_loginable_secgroup_rule public","message":"[bgpvpn] Make _create_loginable_secgroup_rule public\n\nAs tempest.scenario.manager was announced stable interface in Tempest\n27.0.0[1] it can be now reused in plugins.\n\n_create_loginable_secgroup method isn\u0027t ready to be deleted yet, because its rulesets differs from tempest\u0027s.\nThis method will be public now, so replaced methods will use this method over the tempest\u0027s one.\nTempest didn\u0027t allow to modify rulesets, but patch that allows this is already merged[2].\nIn some time, this plugin\u0027s implementation will be replaced\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n[2] https://review.opendev.org/c/openstack/tempest/+/835018\n\nChange-Id: Ia2f8eed6da612bc88c9b9eb48873dcc11b309f42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5388796030f0983f5be29bf1d9f4cb71c7958e80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5388796030f0983f5be29bf1d9f4cb71c7958e80"}]},"branch":"refs/heads/master"},"8235e3b5198c3af40cf2c7c797c8beb028814e80":{"kind":"NO_CHANGE","_number":3,"created":"2022-04-12 10:17:41.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/11/837511/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/11/837511/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/837511/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/837511/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/837511/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/11/837511/3"}}},"commit":{"parents":[{"commit":"1d9e28bd2aaac8abb89708d7a3b1760daed3432b","subject":"[bgpvpn] Remove _get_server_port_id_and_ip4 \u0026 create_floating_ip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1d9e28bd2aaac8abb89708d7a3b1760daed3432b"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-12 09:29:17.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-12 10:17:41.000000000","tz":0},"subject":"[bgpvpn] Make _create_loginable_secgroup_rule public","message":"[bgpvpn] Make _create_loginable_secgroup_rule public\n\nAs tempest.scenario.manager was announced stable interface in Tempest\n27.0.0[1] it can be now reused in plugins.\n\n_create_loginable_secgroup method isn\u0027t ready to be deleted yet, because its rulesets differs from tempest\u0027s.\nThis method will be public now, so replaced methods will use this method over the tempest\u0027s one.\nTempest didn\u0027t allow to modify rulesets, but patch that allows this is already merged[2].\nIn some time, this plugin\u0027s implementation will be replaced\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n[2] https://review.opendev.org/c/openstack/tempest/+/835018\n\nChange-Id: Ia2f8eed6da612bc88c9b9eb48873dcc11b309f42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8235e3b5198c3af40cf2c7c797c8beb028814e80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8235e3b5198c3af40cf2c7c797c8beb028814e80"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
