)]}'
{"id":"openstack%2Fneutron-tempest-plugin~838460","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Iea2655a4dae2c8e260fb1dff9fdd9f426fe3c1b3","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"improve-neutron-ci","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-05-02 23:59:00.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-05-02 23:59:00.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-05-02 13:18:17.000000000","reason":"Lajos Katona replied on the change"}},"hashtags":[],"change_id":"Iea2655a4dae2c8e260fb1dff9fdd9f426fe3c1b3","subject":"Make test test_remove_associated_ports to be compatible with L3HA","status":"MERGED","created":"2022-04-19 09:33:49.000000000","updated":"2022-05-02 23:59:00.000000000","submitted":"2022-05-02 23:59:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"838460-improve-neutron-ci","meta_rev_id":"d0d14d4faec6f3c4b2b0db263771ff03d6a6e41c","_number":838460,"virtual_id_number":838460,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-02 23:59:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2022-05-02 22:20:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2022-05-02 23:59:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2022-05-02 22:20:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-19 10:37:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-20 15:42:11.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2022-04-20 16:08:39.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-04-20 23:54:50.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"e32a2acbd220711690e5b036895080cf738026c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-19 09:33:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2930e1799800b5cdb7597b2cd401d1f4350d46c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 10:37:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6408124b6a1046ecbb321908ba53cf21 : SUCCESS in 3m 56s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/8f7c3a7cc6ac44dda2916d019c778e30 : SUCCESS in 36m 39s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/6a99686d0e2b4e269fc1fa535f92f0cc : SUCCESS in 54m 52s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/56ef549af0e54129933f9c1c32f6a9de : SUCCESS in 59m 11s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/42094840d65543b4a09828d1b685ee71 : SUCCESS in 42m 30s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/7cb746776e48427986fafa3310dd8e13 : SUCCESS in 42m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fabd8c3af6c4c96bfccbc2362def4dc : SUCCESS in 4m 19s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/dcf805a136f64ca5ba24331a52340611 : SUCCESS in 32m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a455cf032803bd9d6312a69df53a3c4ce1074913","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-20 15:42:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fcfb7a2419ace3f82dba02e737175f2bcebdb594","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-04-20 16:08:39.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"64e57dfa88e89d89a1a3b1bf8c3fd9ca48da19d7","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-04-20 23:54:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc3d6d1816f7f8e77b5aa77e442891f5e3d021b9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-21 07:32:35.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b47148d471bc3e03f2011824b645038d75f50b97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 08:30:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0140d6af29774301b26f9fd22956742f : SUCCESS in 4m 15s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/b3720674bba44066b08c246feea74c49 : SUCCESS in 51m 47s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/17edfad338084d82b9519a8587d94886 : SUCCESS in 51m 55s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/31bdbff6f80945bebc69e0c653ca7745 : SUCCESS in 48m 44s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/98c3529c2a20453788f50490812926af : SUCCESS in 45m 42s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/0b7edc3d74744a97bb9308e807ce1e1d : SUCCESS in 48m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b882aa0f015448c2966bc1927e03ebe1 : SUCCESS in 3m 22s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/39d82bd706d84241bb928edbf973e715 : SUCCESS in 35m 40s","accounts_in_message":[],"_revision_number":1},{"id":"9513b344617078e82a2ce91eb8ae801151350ce1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-25 12:50:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea6380b1f07a2e0af2ae0dcb0e4bcef42ed7a620","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-02 13:18:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e9b14979b91316f71c1b597e1c581b91312639e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 13:19:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4108d32e5fd9a395fb3930a4ebe0937f79bc2f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-05-02 13:19:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"23f93e5036da126dee52593f0d9b4689d922ffa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 14:28:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d08f1f0d73c4be99e7c13818a6f4acd : SUCCESS in 3m 30s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fb48342ccaa7435eb7084f0015ba64b6 : SUCCESS in 47m 48s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/e96198f80f59430985d4be3724a6199d : SUCCESS in 51m 26s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/ec17219ebc944cb1b5f17a5ba0344155 : SUCCESS in 1h 02m 17s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/21a6382e9a724f93aa168f7fa1baf5cb : SUCCESS in 43m 40s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/d72c818313f54ddea354c1ff94c4cb80 : SUCCESS in 45m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e379bf9ea714574bb3c170a049ba959 : SUCCESS in 4m 05s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/f877347581d446039e9290377f057548 : SUCCESS in 36m 35s","accounts_in_message":[],"_revision_number":2},{"id":"c1e8fdac841c0be9b9e4bf2a784a0eba831ce95b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-05-02 22:20:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d27168b639e1800b332425b986570258d7b41e3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 22:21:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7e6917d28b95e283a84c4d02a6a568ba7e04cc53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 23:59:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/245a609a8c9841a189cec469c31f2fd2 : SUCCESS in 3m 52s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/e8e4635695584d189fbb90d6b6b3ccd4 : SUCCESS in 55m 04s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/117a04cc39284e97b4905d4c7224f4ba : SUCCESS in 46m 03s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/70f26cc5996842f98eb53ad5c4d6f17a : SUCCESS in 52m 15s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/d8b6caf9f93641b2ae2316529e104e38 : SUCCESS in 49m 30s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/3afbb210d4d34edd9295d942c0d6f6ae : SUCCESS in 49m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23a7e7ffa9ce462b9216b91bd87aac49 : SUCCESS in 3m 24s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/d9fe5c2a749f4ef2abf59ed2761f05cb : SUCCESS in 37m 53s","accounts_in_message":[],"_revision_number":2},{"id":"8e536cf07b99f6d2cd579c5e17e1a23884ce0de5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 23:59:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7c09d72b7e0113e5cd8392f7c7d0c1eee672657f","revisions":{"210af9fdf71661ca4027801bbb0a75235e23f51d":{"kind":"REWORK","_number":1,"created":"2022-04-19 09:33:49.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/60/838460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/60/838460/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/60/838460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/60/838460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/60/838460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/60/838460/1"}}},"commit":{"parents":[{"commit":"36a231e229a0d178403d6aadf37567868bedf4d3","subject":"Merge \"[sfc] Remove unused sec groups methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/36a231e229a0d178403d6aadf37567868bedf4d3"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-19 09:30:39.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-19 09:33:45.000000000","tz":120},"subject":"Make test test_remove_associated_ports to be compatible with L3HA","message":"Make test test_remove_associated_ports to be compatible with L3HA\n\nNegative API test\nneutron_tempest_plugin.api.test_routers_negative.RoutersNegativePolicyTest.\\\ntest_remove_associated_ports\n\nwas trying to remove all ports which belongs to the specic device\n(router) and expected Conflict response for each of them.\nIn L3HA environment that\u0027s not true and router ha interfaces can be\nremoved without any problems. That caused failure of the test when L3_ha\nwas enabled.\nThis patch fixes it by trying to delete different types of port than ha\ninterfaces.\n\nChange-Id: Iea2655a4dae2c8e260fb1dff9fdd9f426fe3c1b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/210af9fdf71661ca4027801bbb0a75235e23f51d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/210af9fdf71661ca4027801bbb0a75235e23f51d"}]},"branch":"refs/heads/master"},"7c09d72b7e0113e5cd8392f7c7d0c1eee672657f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-05-02 13:19:41.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/60/838460/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/60/838460/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/60/838460/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/60/838460/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/60/838460/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/60/838460/2"}}},"commit":{"parents":[{"commit":"1a263ccccb23316f5744957e1b868dab11a8432c","subject":"Merge \"Collect ovn/ovs logs and ovn dbs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1a263ccccb23316f5744957e1b868dab11a8432c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-19 09:30:39.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-05-02 13:19:41.000000000","tz":0},"subject":"Make test test_remove_associated_ports to be compatible with L3HA","message":"Make test test_remove_associated_ports to be compatible with L3HA\n\nNegative API test\nneutron_tempest_plugin.api.test_routers_negative.RoutersNegativePolicyTest.\\\ntest_remove_associated_ports\n\nwas trying to remove all ports which belongs to the specic device\n(router) and expected Conflict response for each of them.\nIn L3HA environment that\u0027s not true and router ha interfaces can be\nremoved without any problems. That caused failure of the test when L3_ha\nwas enabled.\nThis patch fixes it by trying to delete different types of port than ha\ninterfaces.\n\nChange-Id: Iea2655a4dae2c8e260fb1dff9fdd9f426fe3c1b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7c09d72b7e0113e5cd8392f7c7d0c1eee672657f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7c09d72b7e0113e5cd8392f7c7d0c1eee672657f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
