)]}'
{"id":"openstack%2Fneutron-tempest-plugin~839954","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ic23cbd154271988adad8d172c7668a918655c4e2","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-06-15 14:58:05.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ic23cbd154271988adad8d172c7668a918655c4e2","subject":"DNM Just test API with enforced new defaults","status":"ABANDONED","created":"2022-04-29 14:21:51.000000000","updated":"2022-06-15 14:58:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9e54041dd68fe782acd9f8546e9c9bdc34bc8e99","_number":839954,"virtual_id_number":839954,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-29 15:12:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 15:12:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b7c7c5a772f586717608f2df7241c33d257ed5cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-29 14:21:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9251b472c5ac47fd3e23df49bcfe56c62e9d41e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 15:12:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79685d92b7984ae49bc2279f5d634915 : SUCCESS in 4m 41s\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1312afbe17e04bd6b1291480b808e998 : FAILURE in 38m 35s\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/ef31c29b2d3f4c30983de5c82c6bd3bf : SUCCESS in 43m 13s\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/5c2c436087c0498db01eceb92a1ce26d : SUCCESS in 38m 02s\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/3951263aa990489dbbf56ce42156d23b : SUCCESS in 46m 50s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/9c36e0655d1d4ab2be5a70a1e892a0e2 : FAILURE in 32m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e9e1c13ef96413bbfc7bfe221592cf9 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"9e54041dd68fe782acd9f8546e9c9bdc34bc8e99","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-06-15 14:58:05.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"110b876ed1a7a6ad9c2b5c010bd28e46cd15f38b","revisions":{"110b876ed1a7a6ad9c2b5c010bd28e46cd15f38b":{"kind":"REWORK","_number":1,"created":"2022-04-29 14:21:51.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/54/839954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/54/839954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/54/839954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/54/839954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/54/839954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/54/839954/1"}}},"commit":{"parents":[{"commit":"b02a11f3f7b7eb126eb57ab6d3260d650ef24211","subject":"Merge \"[fwaas] Remove secgroup creation related methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b02a11f3f7b7eb126eb57ab6d3260d650ef24211"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-29 14:21:02.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-29 14:21:48.000000000","tz":120},"subject":"DNM Just test API with enforced new defaults","message":"DNM Just test API with enforced new defaults\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/839952\nChange-Id: Ic23cbd154271988adad8d172c7668a918655c4e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/110b876ed1a7a6ad9c2b5c010bd28e46cd15f38b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/110b876ed1a7a6ad9c2b5c010bd28e46cd15f38b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
