)]}'
{"id":"openstack%2Fneutron-tempest-plugin~848198","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I1d4ed4172031f0c2aae4886ca0daad1b90e55ee0","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"default-acl-drop","attention_set":{},"removed_from_attention_set":{"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-06-17 09:47:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1d4ed4172031f0c2aae4886ca0daad1b90e55ee0","subject":"Use OVN version 22.06","status":"ABANDONED","created":"2022-06-29 23:54:20.000000000","updated":"2023-06-17 09:47:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d35556e51ec7a3ec97ca83463d1c9d793c938655","_number":848198,"virtual_id_number":848198,"owner":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-29 19:00:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-30 01:20:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-18 10:58:55.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-07-21 15:47:05.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-07-29 07:24:08.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"35c6253ec6adfc91de6a1362423a979ac33d17e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-06-29 23:54:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e981f84589ef51a89d147e5847024992bbbaa1f","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-06-30 00:01:01.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"3220795196e6fbabc23ad3a439ef2936f69aa26d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 01:20:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5b11dcb24c1481480fc9dcd9e713ef2 : SUCCESS in 3m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8e57765c06ca4f99a499876db6a072b8 : FAILURE in 1h 23m 28s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/6feaf90dd9cd4c6ab0d3da4af408eab7 : SUCCESS in 1h 18m 46s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/1706be2d2d3448caab427b9b8b6b1d07 : SUCCESS in 1h 09m 34s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/aafc58a2c4f04a06b5c7cce795153aaa : FAILURE in 1h 20m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/148118a12d9144b1ad5ea58afbdc8520 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"31e84138961593acbf1520f6271a1fb7e374125a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-07-07 23:15:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"659754aef4d6e978e67c9dcf7e9a088349dc442b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-07-07 23:16:13.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"921115db719d71afd13a3d373cdde8a0cf05254d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 00:30:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ebcbb5cd9114bb78e18a01cc20bd712\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8f7d5635a2d4db782ed07e4c109e9e2 : SUCCESS in 3m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1ed26ea88fbd436abbdfceba73e78ade : SUCCESS in 1h 05m 01s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/fb9017e494784abc830056d1e9ea37cd : SUCCESS in 1h 13m 10s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/c9aa3d7c9ba744a582a215fc2210646b : SUCCESS in 51m 55s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/c9f1d2ba0e1447d2853457d4ff1bc1f0 : SUCCESS in 58m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7f584d26fea409b8fe385b636f0c2ab : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":2},{"id":"a078fa05435b76ccb98feb04f2f7ad612e5d84d9","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-07-11 21:25:22.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"2f945915745357b8520df5b633b8bde0d1f8abf2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-18 10:58:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a86664c0ec4e14f29289ed0b2bce8bae959562b2","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-07-21 15:47:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fb29537615654af732ca7cc929b8cc484d5763ab","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-07-29 07:24:08.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7052a9d7e6cec9b45a3399c5fddc328665cafa19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-01-31 01:18:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"687cdfb13d06ad62b67c2b09e832ce1e4d9ec6b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 02:58:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d92637003d954f889df8d4ec1b007cc9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d57dff4190e14272ac90336856e14c7a : SUCCESS in 3m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/90326ecc50f8442cae4990acd6a18250 : SUCCESS in 1h 33m 29s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/0c9737ae44474f17be4627cc7926dc9b : FAILURE in 17m 06s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/307368ca5a6a485c997b89c9f78968e3 : FAILURE in 16m 56s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/cf175fe424554d9eaa1af7986ddf8eaa : FAILURE in 15m 44s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/730ae4242f5f4db28b775bc0d9640ae6 : SUCCESS in 1h 37m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9323a8d1bdd492392c4ff89db65a85a : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":3},{"id":"08dd4a6ada28cefd845d1f459318e9fc25fb77d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-02-13 04:20:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5778e20b55ba1bcd5e403cda513a0bdde88dbdd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 05:40:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/32ff0d0098944868bcc0ec1b37079a8a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/312d4495951d493c8a66052ab09ba3da : SUCCESS in 3m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/052b0838bf35472d99f8f673f3a8cfcd : SUCCESS in 1h 17m 35s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/96f4175b33074eedbae0961cc57a93fa : FAILURE in 16m 59s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/79a72f6de8ec4cf68d84f3017ba0e319 : FAILURE in 15m 21s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/8b35ed2768484a968623522283b57cd3 : SUCCESS in 1h 12m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d141b35df004f8f8630bb664ee16477 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":4},{"id":"a2da699ced5c8b3761ddcb7b5c8eddbf08bf77a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-02-14 00:40:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"688c59e0c22845cc8bd623c21b93a87fe522c2e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 02:21:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd790c631dda44b7a186b41031c378ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/003bf2d716f74b80ad174d7802738edc : SUCCESS in 3m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5d0dc44953f442eb9eda80c5e136aedf : SUCCESS in 1h 26m 15s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/72d0d2e6ad4d4690aad35e87b9c31a1d : FAILURE in 14m 55s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/d84183c2c24e4c00ac1c73e726c521c6 : FAILURE in 17m 26s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/0e4be1a46305406e899ed63fe91adb8c : SUCCESS in 1h 38m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d67cf9ffc31b48a4af419fb7230d963e : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":5},{"id":"a58022fe51fb0ec34e737462806af43f9a5ac24c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-17 08:02:46.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":5},{"id":"52a5828b24aba3dea2d497eac343203688ceea1d","tag":"autogenerated:gerrit:restore","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-04-28 23:26:29.000000000","message":"Restored","accounts_in_message":[],"_revision_number":5},{"id":"3fa6eab16c4dc7dc80cb8dab0e0b7fc75ef2854f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-29 01:28:38.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02e0e3f14ab348a8a5bfc777f5f57a0d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98895858a758486fa932b13ea942bd44 : SUCCESS in 3m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b96bab1c154a406bbb267b0197517572 : SUCCESS in 1h 39m 37s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/d431283dc41f44109c33a38afb9739b5 : FAILURE in 15m 52s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/8ed0a4e0c973463badae5f9136540591 : FAILURE in 16m 17s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/e671dbff01a743d484235465d265c7c6 : SUCCESS in 2h 00m 16s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/00164f23815b451e9ad2c23ad8484745 : SUCCESS in 1h 33m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52040fbb2a4045dfb6f66794fbb167b5 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":5},{"id":"8cdb03efddc5b1d518c5feedd5a40a66964c83c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-04-29 16:57:41.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0ba67ad324001cc368a6b7fcb95140544a175e00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-29 19:00:48.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8577be8b6bfc4c2aa4760d60ff0f733c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f4d969eac5f4e6185e3c7df3d5b4c1e : SUCCESS in 3m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8d75e79741a743e5b177bcbfcf019fd8 : SUCCESS in 1h 56m 39s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/a6870e2b4b214c77b774e02020fda58a : NODE_FAILURE Node request 300-0021098250 failed in 0s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/08a66d63306846fdb36f3a272ee2e61c : FAILURE in 21m 12s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/d1d4f7f7341a4a4fb1ea2f03ca332aff : SUCCESS in 1h 51m 07s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/bcdcd7d1136343268426d037796e4e02 : SUCCESS in 1h 47m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48c934d8daf9454ba04ae1cc46525121 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":6},{"id":"d35556e51ec7a3ec97ca83463d1c9d793c938655","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-06-17 09:47:38.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a44372f605dffac72d63933b6decac7cb158c94d","revisions":{"2e4885e20423c18fcd2d8331566c391172c0faa3":{"kind":"REWORK","_number":1,"created":"2022-06-29 23:54:20.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/98/848198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/98/848198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/1"}}},"commit":{"parents":[{"commit":"318ec127c548d4f3cba1ceb9edb11c5d250b1786","subject":"Merge \"Add ndp proxy API tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/318ec127c548d4f3cba1ceb9edb11c5d250b1786"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2022-06-29 23:45:09.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2022-06-29 23:45:09.000000000","tz":-300},"subject":"Use OVN version 22.06","message":"Use OVN version 22.06\n\n[0] changes the default firewall policy in ML2 / OVN based on the new\ndefault_acl_drop option in upstream OVN [1]. This new option is included\nin version 22.06 of OVN. This change updates CI jobs definition to\nrequire OVN version 22.06\n\n[0] https://review.opendev.org/c/openstack/neutron/+/839066\n[1] https://patchwork.ozlabs.org/project/ovn/patch/20220419133746.17712-1-dceara@redhat.com/\n\nChange-Id: I1d4ed4172031f0c2aae4886ca0daad1b90e55ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2e4885e20423c18fcd2d8331566c391172c0faa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2e4885e20423c18fcd2d8331566c391172c0faa3"}]},"branch":"refs/heads/master"},"b7bea70f82d44f39b715ae252234b802c1104657":{"kind":"REWORK","_number":2,"created":"2022-07-07 23:15:51.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/98/848198/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/98/848198/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/2"}}},"commit":{"parents":[{"commit":"7b2f5c38a1b5483c0cb8a767e74ae12e3df6c63b","subject":"Merge \"Add a test for removing security group from ACTIVE instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7b2f5c38a1b5483c0cb8a767e74ae12e3df6c63b"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2022-06-29 23:45:09.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2022-07-07 23:15:40.000000000","tz":-300},"subject":"Use OVN version 22.06","message":"Use OVN version 22.06\n\n[0] changes the default firewall policy in ML2 / OVN based on the new\ndefault_acl_drop option in upstream OVN [1]. This new option is included\nin version 22.06 of OVN. This change updates CI jobs definition to\nrequire OVN version 22.06\n\n[0] https://review.opendev.org/c/openstack/neutron/+/839066\n[1] https://patchwork.ozlabs.org/project/ovn/patch/20220419133746.17712-1-dceara@redhat.com/\n\nChange-Id: I1d4ed4172031f0c2aae4886ca0daad1b90e55ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b7bea70f82d44f39b715ae252234b802c1104657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b7bea70f82d44f39b715ae252234b802c1104657"}]},"branch":"refs/heads/master"},"f3bd780048fcf1308f4f202b72e5a22d4006ed69":{"kind":"REWORK","_number":3,"created":"2023-01-31 01:18:01.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/98/848198/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/98/848198/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/3"}}},"commit":{"parents":[{"commit":"a20bcddf514ea2fc3ee445d8547cbf520c32bdfb","subject":"Stateful SG tests without ingress metadata rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a20bcddf514ea2fc3ee445d8547cbf520c32bdfb"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2022-06-29 23:45:09.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2023-01-31 01:17:30.000000000","tz":-360},"subject":"Use OVN version 22.06","message":"Use OVN version 22.06\n\n[0] changes the default firewall policy in ML2 / OVN based on the new\ndefault_acl_drop option in upstream OVN [1]. This new option is included\nin version 22.06 of OVN. This change updates CI jobs definition to\nrequire OVN version 22.06\n\n[0] https://review.opendev.org/c/openstack/neutron/+/839066\n[1] https://patchwork.ozlabs.org/project/ovn/patch/20220419133746.17712-1-dceara@redhat.com/\n\nChange-Id: I1d4ed4172031f0c2aae4886ca0daad1b90e55ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f3bd780048fcf1308f4f202b72e5a22d4006ed69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f3bd780048fcf1308f4f202b72e5a22d4006ed69"}]},"branch":"refs/heads/master"},"bc237b3ef8ea83280bf5fb67de7e915b33da241f":{"kind":"REWORK","_number":4,"created":"2023-02-13 04:20:40.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/98/848198/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/98/848198/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/4"}}},"commit":{"parents":[{"commit":"481f19fb786d2f7a8420f6f061791b9c9f91482a","subject":"Do not use ubuntu minimal image in ussuri scenario jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/481f19fb786d2f7a8420f6f061791b9c9f91482a"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2022-06-29 23:45:09.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2023-02-13 04:19:16.000000000","tz":-360},"subject":"Use OVN version 22.06","message":"Use OVN version 22.06\n\n[0] changes the default firewall policy in ML2 / OVN based on the new\ndefault_acl_drop option in upstream OVN [1]. This new option is included\nin OVN\u0027s latest versions. This change updates CI jobs definition to\nrequire OVN branch main.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/839066\n[1] https://patchwork.ozlabs.org/project/ovn/patch/20220419133746.17712-1-dceara@redhat.com/\n\nChange-Id: I1d4ed4172031f0c2aae4886ca0daad1b90e55ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bc237b3ef8ea83280bf5fb67de7e915b33da241f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bc237b3ef8ea83280bf5fb67de7e915b33da241f"}]},"branch":"refs/heads/master"},"461d599b761a9984aaa1a9782afbead8adf6af9b":{"kind":"REWORK","_number":5,"created":"2023-02-14 00:40:28.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/98/848198/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/98/848198/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/5"}}},"commit":{"parents":[{"commit":"481f19fb786d2f7a8420f6f061791b9c9f91482a","subject":"Do not use ubuntu minimal image in ussuri scenario jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/481f19fb786d2f7a8420f6f061791b9c9f91482a"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2022-06-29 23:45:09.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2023-02-14 00:39:35.000000000","tz":-360},"subject":"Use OVN version 22.06","message":"Use OVN version 22.06\n\n[0] changes the default firewall policy in ML2 / OVN based on the new\ndefault_acl_drop option in upstream OVN [1]. This new option is included\nin version 22.06 of OVN. This change updates CI jobs definition to\nrequire OVN version 22.06.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/839066\n[1] https://patchwork.ozlabs.org/project/ovn/patch/20220419133746.17712-1-dceara@redhat.com/\n\nChange-Id: I1d4ed4172031f0c2aae4886ca0daad1b90e55ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/461d599b761a9984aaa1a9782afbead8adf6af9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/461d599b761a9984aaa1a9782afbead8adf6af9b"}]},"branch":"refs/heads/master"},"a44372f605dffac72d63933b6decac7cb158c94d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-04-29 16:57:41.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/98/848198/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/98/848198/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/98/848198/6"}}},"commit":{"parents":[{"commit":"4330ca08687772aaa50867cc48668e2a17098aaf","subject":"Don\u0027t run stadium related jobs when scenario tests are changed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4330ca08687772aaa50867cc48668e2a17098aaf"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2022-06-29 23:45:09.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2023-04-28 23:52:51.000000000","tz":-300},"subject":"Use OVN version 22.06","message":"Use OVN version 22.06\n\n[0] changes the default firewall policy in ML2 / OVN based on the new\ndefault_acl_drop option in upstream OVN [1]. This new option is included\nin version 22.06 of OVN. This change updates CI jobs definition to\nrequire OVN version 22.06.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/839066\n[1] https://patchwork.ozlabs.org/project/ovn/patch/20220419133746.17712-1-dceara@redhat.com/\n\nChange-Id: I1d4ed4172031f0c2aae4886ca0daad1b90e55ee0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a44372f605dffac72d63933b6decac7cb158c94d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a44372f605dffac72d63933b6decac7cb158c94d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
