)]}'
{"id":"openstack%2Fneutron-tempest-plugin~848934","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I6e361126f77df1b033001f9e6c97f86ea9e76a98","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"test-igmp","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2022-07-07 12:35:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6e361126f77df1b033001f9e6c97f86ea9e76a98","subject":"DO NOT REVIEW, TEST IGMP","status":"ABANDONED","created":"2022-07-07 09:27:22.000000000","updated":"2022-07-07 12:35:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8fc924e2913deab8fa6a1d9ca5c20d7320090eb1","_number":848934,"virtual_id_number":848934,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-07 11:14:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-07 09:52:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1145cd39ed94c8237876f78723461da418c72791","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-07-07 09:27:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"113e7ee727b008ebb43e032b6e8bd53027101cf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 09:52:43.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"bca5becd0ea5d03bfb2e703be71ca0dc5ce170f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-07-07 09:53:13.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7d18d3a31853427c87f896c604b25341848f0860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 11:14:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cb1299b6eb148a9a78168ff81dc2b48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96ecec2b4ed24346a45354c4a33019d9 : SUCCESS in 4m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b3191bfb933c4b5798b6785ada55763a : SUCCESS in 1h 03m 55s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/d4aba6c8db764f38bf58b5af504a9661 : SUCCESS in 1h 19m 34s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/1db25b7d4dd442f898d4540265b9383b : SUCCESS in 58m 20s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/21eb1b8e8ef7474bab5573a90eb7f6f0 : SUCCESS in 52m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4efc90b9e5b64d7aac752e8ffab3b6fc : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"8fc924e2913deab8fa6a1d9ca5c20d7320090eb1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-07-07 12:35:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7f2db1f768c0838ab3eb1fd7a590beffdd5d69f5","revisions":{"123955f0758801e053744bc451b18316fa445eba":{"kind":"REWORK","_number":1,"created":"2022-07-07 09:27:22.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/34/848934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/34/848934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/848934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/848934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/848934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/848934/1"}}},"commit":{"parents":[{"commit":"7b2f5c38a1b5483c0cb8a767e74ae12e3df6c63b","subject":"Merge \"Add a test for removing security group from ACTIVE instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7b2f5c38a1b5483c0cb8a767e74ae12e3df6c63b"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2022-07-07 09:26:46.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2022-07-07 09:26:46.000000000","tz":60},"subject":"DO NOT REVIEW, TEST IGMP","message":"DO NOT REVIEW, TEST IGMP\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/848933\nChange-Id: I6e361126f77df1b033001f9e6c97f86ea9e76a98\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/123955f0758801e053744bc451b18316fa445eba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/123955f0758801e053744bc451b18316fa445eba"}]},"branch":"refs/heads/master"},"7f2db1f768c0838ab3eb1fd7a590beffdd5d69f5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-07 09:53:13.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/34/848934/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/34/848934/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/848934/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/848934/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/848934/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/848934/2"}}},"commit":{"parents":[{"commit":"7b2f5c38a1b5483c0cb8a767e74ae12e3df6c63b","subject":"Merge \"Add a test for removing security group from ACTIVE instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7b2f5c38a1b5483c0cb8a767e74ae12e3df6c63b"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2022-07-07 09:26:46.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2022-07-07 09:53:03.000000000","tz":60},"subject":"DO NOT REVIEW, TEST IGMP","message":"DO NOT REVIEW, TEST IGMP\n\ntest\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/848933\nChange-Id: I6e361126f77df1b033001f9e6c97f86ea9e76a98\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7f2db1f768c0838ab3eb1fd7a590beffdd5d69f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7f2db1f768c0838ab3eb1fd7a590beffdd5d69f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
