)]}'
{"id":"openstack%2Fneutron-tempest-plugin~856985","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I87f2a891ae518a79ca8502ed576b1dea27ea25b2","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"test-cirros","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-10-16 07:12:40.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I87f2a891ae518a79ca8502ed576b1dea27ea25b2","subject":"DNM: Test Ubuntu 22.04 as advanced image","status":"ABANDONED","created":"2022-09-12 04:42:28.000000000","updated":"2022-10-16 07:12:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"95c5076174dcffa7ac88514d1a068ee936ace447","_number":856985,"virtual_id_number":856985,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-12 06:49:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 06:49:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-16 07:12:40.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"57f33337f49a102b2ce31a10d9398e3842809562","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-09-12 04:42:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c8957ebe0f56b35660de8ed6ca28141d68d6a22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 06:49:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fce939be548d4015a1610adebcbab3ae\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/66246ff280e24fec98f669d43422e2f0 : SUCCESS in 1h 40m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1676ce26cb154e27a0f9bbbd46de0410 : SUCCESS in 1h 33m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/795efee44c4d4908b01178eb70b0f31f : SUCCESS in 1h 20m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7567327d3a734aca83015debf3b03e82 : SUCCESS in 1h 12m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f9c0ed19533f4df1b89002928ea23f39 : SUCCESS in 33m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7007ac8595fd40709cb5ba3949719a9f : SUCCESS in 2m 51s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/73f20912b7d341259c0061124916de24 : FAILURE in 1h 11m 19s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/8e1c477e9d434778be07cca54e8a54e0 : TIMED_OUT in 2h 05m 05s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/7ec1a2483f1a4814991eae4fa3149dba : SUCCESS in 37m 21s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/2aa9407e412741c8b26ab52066e43991 : FAILURE in 34m 25s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/b55fd84ff52b47cb9b4d0ee4b058dfa9 : SUCCESS in 35m 04s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/d3cfe7ffcaed4f41baf7aae1d6b41318 : SUCCESS in 1h 04m 14s","accounts_in_message":[],"_revision_number":1},{"id":"95c5076174dcffa7ac88514d1a068ee936ace447","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-10-16 07:12:40.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c3a252d9ce16dd74390cf0db66871b1943d8b51","revisions":{"0c3a252d9ce16dd74390cf0db66871b1943d8b51":{"kind":"REWORK","_number":1,"created":"2022-09-12 04:42:28.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/85/856985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/85/856985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/85/856985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/85/856985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/85/856985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/85/856985/1"}}},"commit":{"parents":[{"commit":"9f11643b41abf625c2b49f8328503fd178b58be5","subject":"Merge \"Stein: fix failing jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9f11643b41abf625c2b49f8328503fd178b58be5"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-08-28 12:53:44.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-09-12 04:42:02.000000000","tz":120},"subject":"DNM: Test Ubuntu 22.04 as advanced image","message":"DNM: Test Ubuntu 22.04 as advanced image\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I87f2a891ae518a79ca8502ed576b1dea27ea25b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0c3a252d9ce16dd74390cf0db66871b1943d8b51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0c3a252d9ce16dd74390cf0db66871b1943d8b51"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
