)]}'
{"id":"openstack%2Fneutron-tempest-plugin~856999","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I592bebfdbcf51e6a2273a2e7676365b0a790a818","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"test-cirros","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-10-16 07:12:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I592bebfdbcf51e6a2273a2e7676365b0a790a818","subject":"DNM: Test with qemu instead of kvm","status":"ABANDONED","created":"2022-09-12 07:24:58.000000000","updated":"2022-10-16 07:12:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ecaa8534e547077b2811db345878183e222f6a9e","_number":856999,"virtual_id_number":856999,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-14 15:05:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-12 10:12:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-16 07:12:30.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"00f3ae6d659f788ebd873c9f375f8105c54dcb51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-09-12 07:24:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbf4c9595a6fb75f05e2c7725d6973f3df8606c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 10:12:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ef04263831a46b28509174adeba1911\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/32a0f1b333aa4b6fb5590254ac48a8bb : FAILURE in 1h 19m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e2f65e0c83bd44c39453bcb742aab330 : FAILURE in 1h 47m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b0207601d7f3434db0438a676422fcdc : FAILURE in 2h 05m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0c87eca29beb4a89a07b50438a116866 : FAILURE in 2h 06m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/67b8a1c08109400b98fd38383034ced1 : SUCCESS in 52m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89a10bb0114e462d8a87f7095f742b18 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"57bf78f0ca2a9415b2a94d28179ed71999307980","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-09-14 11:45:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0b5010c0a14add34765f7e3d75117927cd9027fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 15:05:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a90acb373c24dd2ae499fd9a9884406\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d12833b6dd4e4e3fb0b9bcbbb209de4a : FAILURE in 1h 36m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/159e7b071d674daca305cc3b777240b9 : FAILURE in 1h 45m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8518b87f8a534ce5a7515150a04f19c9 : FAILURE in 1h 43m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/34260fcfe5ce4e7181a07a6c3460ca45 : FAILURE in 2h 46m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9bedfdd3cede4ea489b65a229258535d : SUCCESS in 37m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14cc03216ca741bc9c1603547f925c5e : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"ecaa8534e547077b2811db345878183e222f6a9e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-10-16 07:12:30.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5475ca8cf7c6c2aca65e66054a6de5000cc5b6f3","revisions":{"8062318d7340cd2377832e439993efa0d72dab6c":{"kind":"REWORK","_number":1,"created":"2022-09-12 07:24:58.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/99/856999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/99/856999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/99/856999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/99/856999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/99/856999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/99/856999/1"}}},"commit":{"parents":[{"commit":"350caeef62126fec427f5252325ba082979e36c7","subject":"DNM: Test custom cirros image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/350caeef62126fec427f5252325ba082979e36c7"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-09-12 07:24:00.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-09-12 07:24:00.000000000","tz":120},"subject":"DNM: Test with qemu instead of kvm","message":"DNM: Test with qemu instead of kvm\n\nIt looks like nested kvm is unstable on some clouds.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I592bebfdbcf51e6a2273a2e7676365b0a790a818\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8062318d7340cd2377832e439993efa0d72dab6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8062318d7340cd2377832e439993efa0d72dab6c"}]},"branch":"refs/heads/master"},"5475ca8cf7c6c2aca65e66054a6de5000cc5b6f3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-09-14 11:45:25.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/99/856999/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/99/856999/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/99/856999/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/99/856999/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/99/856999/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/99/856999/2"}}},"commit":{"parents":[{"commit":"9400b61397de867aa7f8caa91bad5cb4a3ebb47c","subject":"DNM: Test custom cirros image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9400b61397de867aa7f8caa91bad5cb4a3ebb47c"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-09-12 07:24:00.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-09-14 11:45:25.000000000","tz":0},"subject":"DNM: Test with qemu instead of kvm","message":"DNM: Test with qemu instead of kvm\n\nIt looks like nested kvm is unstable on some clouds.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I592bebfdbcf51e6a2273a2e7676365b0a790a818\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5475ca8cf7c6c2aca65e66054a6de5000cc5b6f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5475ca8cf7c6c2aca65e66054a6de5000cc5b6f3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
