)]}'
{"id":"openstack%2Fneutron-tempest-plugin~867601","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I5f18e33d254a0e214c386451329206d677dcc612","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"dnm","hashtags":[],"change_id":"I5f18e33d254a0e214c386451329206d677dcc612","subject":"[DNM] reproduce #1999249","status":"ABANDONED","created":"2022-12-14 05:25:17.000000000","updated":"2023-10-06 14:40:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"483bf9578ada455eea0782c0e0f066ba0636cc83","_number":867601,"virtual_id_number":867601,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-12-14 08:44:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-14 07:03:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0cbc27c4cb2bda5426db19d691108a302a62942","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-12-14 05:25:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d8a4fcfa77b703f6dc8437e6245a9995895e8f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-12-14 05:46:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c643539e3f7c0fd18c207e9ded87c84ed101e82a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 07:03:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1a13a582f0d475fa0ca38fc606e9f3c\n\n- neutron-tempest-plugin-ovn-1 https://zuul.opendev.org/t/openstack/build/162c917fa6ab4db3bce04494b348861b : SUCCESS in 1h 15m 08s\n- neutron-tempest-plugin-ovn-2 https://zuul.opendev.org/t/openstack/build/b69732ea724248a5af55276666deebfa : SUCCESS in 1h 02m 17s\n- neutron-tempest-plugin-ovn-3 https://zuul.opendev.org/t/openstack/build/1d35d8d7bd77409da3f1d1e1305fd83c : SUCCESS in 1h 03m 19s\n- neutron-tempest-plugin-ovn-4 https://zuul.opendev.org/t/openstack/build/031bff3d73ba4ca4b9d947e5c3060319 : SUCCESS in 1h 04m 24s","accounts_in_message":[],"_revision_number":2},{"id":"0f48411368a6c1cb649a41616f58bd2767458b62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-12-14 07:04:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6672311a02a0e63e91ba00107de4b7b4f7264613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 08:44:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1de03e0024dd4a948e39082575c65c38\n\n- neutron-tempest-plugin-ovn-1 https://zuul.opendev.org/t/openstack/build/73d879845d394acdbabfd880fc79b80f : SUCCESS in 1h 31m 42s\n- neutron-tempest-plugin-ovn-2 https://zuul.opendev.org/t/openstack/build/85285e8c739e4a788f217fc8347d07b9 : SUCCESS in 1h 28m 31s\n- neutron-tempest-plugin-ovn-3 https://zuul.opendev.org/t/openstack/build/c99d3b2910ab422a8c57032c60df91bc : SUCCESS in 1h 31m 36s\n- neutron-tempest-plugin-ovn-4 https://zuul.opendev.org/t/openstack/build/20d0bd5c35424afebaf402823e53a04d : SUCCESS in 1h 33m 52s\n- neutron-tempest-plugin-ovn-5 https://zuul.opendev.org/t/openstack/build/90d77cca03ae44fca98d12d81cfbdbd5 : SUCCESS in 1h 23m 09s\n- neutron-tempest-plugin-ovn-6 https://zuul.opendev.org/t/openstack/build/ab3947836fd04ef5b9acb305d790d4a7 : SUCCESS in 1h 19m 57s\n- neutron-tempest-plugin-ovn-7 https://zuul.opendev.org/t/openstack/build/269744e67f3a4cfa890ec606a19cf059 : SUCCESS in 1h 26m 34s\n- neutron-tempest-plugin-ovn-8 https://zuul.opendev.org/t/openstack/build/3ff06357341f425b96894d220979b4e5 : SUCCESS in 1h 30m 40s\n- neutron-tempest-plugin-ovn-9 https://zuul.opendev.org/t/openstack/build/701e66ad6d304204a024b8e2b4d60908 : SUCCESS in 1h 29m 14s\n- neutron-tempest-plugin-ovn-10 https://zuul.opendev.org/t/openstack/build/d1e816f9613247a7b26409a2faae9dfc : SUCCESS in 1h 11m 52s","accounts_in_message":[],"_revision_number":3},{"id":"483bf9578ada455eea0782c0e0f066ba0636cc83","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-06 14:40:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"50e5ac4196d139e03fee320c522e3af094ae5a80","revisions":{"a6c3b9aa79ff56d28882d440cfde53e9a8a52ba8":{"kind":"REWORK","_number":1,"created":"2022-12-14 05:25:17.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/01/867601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/01/867601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/01/867601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/01/867601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/01/867601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/01/867601/1"}}},"commit":{"parents":[{"commit":"f8eda8fdacd3e2ca6a7e42262467bf974c321319","subject":"Merge \"Update nested-virt testing for the 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f8eda8fdacd3e2ca6a7e42262467bf974c321319"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-14 05:24:54.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-14 05:25:03.000000000","tz":330},"subject":"[DNM] reproduce #1999249","message":"[DNM] reproduce #1999249\n\nChange-Id: I5f18e33d254a0e214c386451329206d677dcc612\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a6c3b9aa79ff56d28882d440cfde53e9a8a52ba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a6c3b9aa79ff56d28882d440cfde53e9a8a52ba8"}]},"branch":"refs/heads/master"},"ce1824cffdea3e51458e9d72cd0a19c964783d7e":{"kind":"REWORK","_number":2,"created":"2022-12-14 05:46:46.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/01/867601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/01/867601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/01/867601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/01/867601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/01/867601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/01/867601/2"}}},"commit":{"parents":[{"commit":"f8eda8fdacd3e2ca6a7e42262467bf974c321319","subject":"Merge \"Update nested-virt testing for the 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f8eda8fdacd3e2ca6a7e42262467bf974c321319"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-14 05:24:54.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-14 05:46:22.000000000","tz":330},"subject":"[DNM] reproduce #1999249","message":"[DNM] reproduce #1999249\n\nChange-Id: I5f18e33d254a0e214c386451329206d677dcc612\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ce1824cffdea3e51458e9d72cd0a19c964783d7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ce1824cffdea3e51458e9d72cd0a19c964783d7e"}]},"branch":"refs/heads/master"},"50e5ac4196d139e03fee320c522e3af094ae5a80":{"kind":"REWORK","_number":3,"created":"2022-12-14 07:04:32.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/01/867601/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/01/867601/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/01/867601/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/01/867601/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/01/867601/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/01/867601/3"}}},"commit":{"parents":[{"commit":"f8eda8fdacd3e2ca6a7e42262467bf974c321319","subject":"Merge \"Update nested-virt testing for the 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f8eda8fdacd3e2ca6a7e42262467bf974c321319"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-14 05:24:54.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-14 07:03:02.000000000","tz":330},"subject":"[DNM] reproduce #1999249","message":"[DNM] reproduce #1999249\n\nChange-Id: I5f18e33d254a0e214c386451329206d677dcc612\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/50e5ac4196d139e03fee320c522e3af094ae5a80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/50e5ac4196d139e03fee320c522e3af094ae5a80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
