)]}'
{"id":"openstack%2Fneutron-tempest-plugin~867934","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ib242dfeca4f348a30f31729a7c7584af0c50edf8","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"dnm","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-04-17 08:03:32.000000000","reason":"Change was abandoned"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-04-17 08:03:32.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib242dfeca4f348a30f31729a7c7584af0c50edf8","subject":"[DNM] Test tb-cache workaround for jammy","status":"ABANDONED","created":"2022-12-16 06:38:15.000000000","updated":"2023-04-17 08:03:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"94acac966a9481955c36cc72b9f0f99cada0b232","_number":867934,"virtual_id_number":867934,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-03 18:58:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-16 09:46:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-03 15:39:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-04-17 08:03:32.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"b8f043f4cfe3148e8da75790423e74c495bc6f41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-12-16 06:38:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"622a33eeee4ec0fffbf900dc7513f7cb4954db9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 09:46:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3b02271830b4e0396ae84382039d7f7\n\n- neutron-tempest-plugin-ovn-1 https://zuul.opendev.org/t/openstack/build/5281df9526704fe0bf3a2f9525459fd1 : TIMED_OUT in 3h 04m 13s\n- neutron-tempest-plugin-ovn-2 https://zuul.opendev.org/t/openstack/build/f12c12437115405c8b12edca969f9d69 : TIMED_OUT in 3h 03m 39s\n- neutron-tempest-plugin-ovn-3 https://zuul.opendev.org/t/openstack/build/fcbe800a833542ae9d0881cadb08db64 : SUCCESS in 2h 30m 23s\n- neutron-tempest-plugin-ovn-4 https://zuul.opendev.org/t/openstack/build/17775dc97316478aafb66c6beb307de3 : TIMED_OUT in 3h 03m 14s\n- neutron-tempest-plugin-ovn-5 https://zuul.opendev.org/t/openstack/build/ea396cd746e04842ba11e1b2a787756c : FAILURE in 2h 58m 03s\n- neutron-tempest-plugin-openvswitch-1 https://zuul.opendev.org/t/openstack/build/1674c518ad304e75bde09c1d904397f4 : SUCCESS in 2h 19m 51s\n- neutron-tempest-plugin-openvswitch-2 https://zuul.opendev.org/t/openstack/build/e2cee323c9324558bca5e19c4f4ab05e : TIMED_OUT in 2h 49m 42s\n- neutron-tempest-plugin-openvswitch-3 https://zuul.opendev.org/t/openstack/build/2340a115ebbd427d9d13c8ddd8aad26a : TIMED_OUT in 2h 50m 54s\n- neutron-tempest-plugin-openvswitch-4 https://zuul.opendev.org/t/openstack/build/7bdec6c57ea54d7cb07912b044580190 : TIMED_OUT in 2h 50m 11s\n- neutron-tempest-plugin-openvswitch-5 https://zuul.opendev.org/t/openstack/build/a1c1dffdc0d04d52af71da89f210b776 : TIMED_OUT in 2h 49m 45s\n- neutron-tempest-plugin-linuxbridge-1 https://zuul.opendev.org/t/openstack/build/e6f0bedd54a640cd97d3ea44bbb55aa4 : TIMED_OUT in 2h 51m 04s\n- neutron-tempest-plugin-linuxbridge-2 https://zuul.opendev.org/t/openstack/build/b16a94fd6f544500a3bd1d9850f3710b : TIMED_OUT in 2h 49m 59s\n- neutron-tempest-plugin-linuxbridge-3 https://zuul.opendev.org/t/openstack/build/a43623d00bf6482d9b8024524a30883c : TIMED_OUT in 2h 50m 25s\n- neutron-tempest-plugin-linuxbridge-4 https://zuul.opendev.org/t/openstack/build/5bcfc4a278c74b9b9f86272802cfec3e : TIMED_OUT in 2h 50m 50s\n- neutron-tempest-plugin-linuxbridge-5 https://zuul.opendev.org/t/openstack/build/ecb4c07a4e394befa2d94a3a6e0ebdd0 : TIMED_OUT in 2h 50m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b128dd161161e8327572be8ed02a5e162083b548","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-12-16 10:32:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"beae0628cb7059682849dc1b83a404a56eed3cf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 13:44:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6da9f7cb59964131810ca6324df2890c\n\n- neutron-tempest-plugin-ovn-1 https://zuul.opendev.org/t/openstack/build/ec85425804a542e2af2de369fdc6bc4e : TIMED_OUT in 3h 03m 07s\n- neutron-tempest-plugin-ovn-2 https://zuul.opendev.org/t/openstack/build/0b94b8716ee24152befecf924662d98e : TIMED_OUT in 3h 03m 11s\n- neutron-tempest-plugin-ovn-3 https://zuul.opendev.org/t/openstack/build/50fad310f86346b9875a01db54924ead : TIMED_OUT in 3h 03m 30s\n- neutron-tempest-plugin-ovn-4 https://zuul.opendev.org/t/openstack/build/9f385fcf038b49c79474be99f40f8d05 : FAILURE in 2h 35m 28s\n- neutron-tempest-plugin-ovn-5 https://zuul.opendev.org/t/openstack/build/755827e223e541d68755efac19b38764 : FAILURE in 2h 46m 24s\n- neutron-tempest-plugin-openvswitch-1 https://zuul.opendev.org/t/openstack/build/94862c90ddb04ff7915048f0a9b1f9d4 : TIMED_OUT in 2h 50m 44s\n- neutron-tempest-plugin-openvswitch-2 https://zuul.opendev.org/t/openstack/build/ae30a33d7c354d9c9e542f6ea2e390e7 : TIMED_OUT in 2h 49m 56s\n- neutron-tempest-plugin-openvswitch-3 https://zuul.opendev.org/t/openstack/build/cc9126eecf2f40fa8903f7901868a8b5 : TIMED_OUT in 2h 50m 14s\n- neutron-tempest-plugin-openvswitch-4 https://zuul.opendev.org/t/openstack/build/8b2fe4b17fd142528501e7236fe73dd8 : TIMED_OUT in 2h 50m 02s\n- neutron-tempest-plugin-openvswitch-5 https://zuul.opendev.org/t/openstack/build/0c0f88902a0e4eb69454b14435bd096c : TIMED_OUT in 2h 49m 54s\n- neutron-tempest-plugin-linuxbridge-1 https://zuul.opendev.org/t/openstack/build/e15a3abf458148379425e1597fa63b7d : TIMED_OUT in 2h 50m 37s\n- neutron-tempest-plugin-linuxbridge-2 https://zuul.opendev.org/t/openstack/build/a2e3ca869088439284a5da629ba4133d : TIMED_OUT in 2h 50m 02s\n- neutron-tempest-plugin-linuxbridge-3 https://zuul.opendev.org/t/openstack/build/248fff8b65644ea2b16c2edd6b190060 : TIMED_OUT in 2h 50m 30s\n- neutron-tempest-plugin-linuxbridge-4 https://zuul.opendev.org/t/openstack/build/9f9264dcb23d4be289a59eea5c1fbac5 : TIMED_OUT in 2h 50m 06s\n- neutron-tempest-plugin-linuxbridge-5 https://zuul.opendev.org/t/openstack/build/e2f76cd9f17b4ff1a4e269d1c9d21ae2 : TIMED_OUT in 2h 49m 57s","accounts_in_message":[],"_revision_number":2},{"id":"e470f1e8283f23c8e30cbb94be43d0d24fd46050","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-12-16 15:46:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7f570fc22cccefcde4e1076c07798ec386d9440b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 18:41:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/15dfa778715e46cba0fcebf11d243e74\n\n- neutron-tempest-plugin-ovn-1 https://zuul.opendev.org/t/openstack/build/2a242662595a4e63b7e9c8f55f793261 : SUCCESS in 1h 47m 52s\n- neutron-tempest-plugin-ovn-2 https://zuul.opendev.org/t/openstack/build/b227d951ecf3442cad343f77fbf08f7e : FAILURE in 2h 16m 21s\n- neutron-tempest-plugin-ovn-3 https://zuul.opendev.org/t/openstack/build/65a3e894be69486fb51f50f8733c2f8a : SUCCESS in 2h 05m 26s\n- neutron-tempest-plugin-ovn-4 https://zuul.opendev.org/t/openstack/build/ac0d56fec66b47bd99d8b1cd274350b1 : SUCCESS in 1h 39m 57s\n- neutron-tempest-plugin-ovn-5 https://zuul.opendev.org/t/openstack/build/b1944e2f4bff4b13880a32078c88483c : SUCCESS in 2h 29m 30s\n- neutron-tempest-plugin-openvswitch-1 https://zuul.opendev.org/t/openstack/build/47fd7697e3ef4ac0aa555040fbe2b9d7 : SUCCESS in 2h 25m 10s\n- neutron-tempest-plugin-openvswitch-2 https://zuul.opendev.org/t/openstack/build/0847b9c949544d97ac7c26b526182fa9 : SUCCESS in 2h 18m 03s\n- neutron-tempest-plugin-openvswitch-3 https://zuul.opendev.org/t/openstack/build/d40ef528ed854382b16750c468d61d68 : SUCCESS in 1h 51m 06s\n- neutron-tempest-plugin-openvswitch-4 https://zuul.opendev.org/t/openstack/build/a94a3d388c44485c9aa738c1cea3f527 : SUCCESS in 2h 09m 30s\n- neutron-tempest-plugin-openvswitch-5 https://zuul.opendev.org/t/openstack/build/19eb95a3290a4cb9a7ab0d5a24b22658 : SUCCESS in 1h 42m 26s\n- neutron-tempest-plugin-linuxbridge-1 https://zuul.opendev.org/t/openstack/build/cb5aa9023df547f3bb6e63586f1f49e7 : SUCCESS in 2h 51m 02s\n- neutron-tempest-plugin-linuxbridge-2 https://zuul.opendev.org/t/openstack/build/26ad250128cf4edca5ecd0ff8c8c9aee : SUCCESS in 2h 41m 11s\n- neutron-tempest-plugin-linuxbridge-3 https://zuul.opendev.org/t/openstack/build/c582d21ec7fc4816965bb593277ce25f : TIMED_OUT in 2h 49m 45s\n- neutron-tempest-plugin-linuxbridge-4 https://zuul.opendev.org/t/openstack/build/c5e7a784cfb449f2809ab1db1a25ccca : SUCCESS in 2h 10m 28s\n- neutron-tempest-plugin-linuxbridge-5 https://zuul.opendev.org/t/openstack/build/3f1c387fd8104a00a27299aec82fc7c7 : FAILURE in 2h 02m 36s","accounts_in_message":[],"_revision_number":3},{"id":"8cca57c4b71a2a85d82c017e62b13d5817bf2282","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-12-22 15:02:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2e050919067efe7332b2a0df6bf83e8a8c8c7149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 18:11:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/18dfaa24bc414b94b41d4994ce42c434\n\n- neutron-tempest-plugin-ovn-1 https://zuul.opendev.org/t/openstack/build/e7d3e7405deb48e587ace0238727cb5e : FAILURE in 2h 58m 32s\n- neutron-tempest-plugin-ovn-2 https://zuul.opendev.org/t/openstack/build/36fa4d8b15e6448f8e49146bc580ff6c : FAILURE in 2h 52m 21s\n- neutron-tempest-plugin-ovn-3 https://zuul.opendev.org/t/openstack/build/9e9433cb5b0c4f1db68fc948e6453f63 : TIMED_OUT in 3h 03m 20s\n- neutron-tempest-plugin-ovn-4 https://zuul.opendev.org/t/openstack/build/09791e1c4c9f4b688d33b7359624570e : FAILURE in 2h 56m 22s\n- neutron-tempest-plugin-ovn-5 https://zuul.opendev.org/t/openstack/build/11b1287b829b4e86a7f2b9371f091e9b : TIMED_OUT in 3h 03m 24s\n- neutron-tempest-plugin-openvswitch-1 https://zuul.opendev.org/t/openstack/build/c156857e5fa042348d7bc98145ffa6b2 : TIMED_OUT in 2h 49m 49s\n- neutron-tempest-plugin-openvswitch-2 https://zuul.opendev.org/t/openstack/build/9a050036ebb745beb592e9682c31da3b : TIMED_OUT in 2h 49m 41s\n- neutron-tempest-plugin-openvswitch-3 https://zuul.opendev.org/t/openstack/build/edafe300b4564c9e8851c775e3f100df : TIMED_OUT in 2h 49m 38s\n- neutron-tempest-plugin-openvswitch-4 https://zuul.opendev.org/t/openstack/build/7a20dc03f93d46b28c194aa8b7b3119c : TIMED_OUT in 2h 50m 12s\n- neutron-tempest-plugin-openvswitch-5 https://zuul.opendev.org/t/openstack/build/8abe4c63247b49ebb3f630b643af7deb : TIMED_OUT in 2h 49m 56s\n- neutron-tempest-plugin-linuxbridge-1 https://zuul.opendev.org/t/openstack/build/e4a439669a5844aebfcb2bd1967ee414 : TIMED_OUT in 2h 50m 01s\n- neutron-tempest-plugin-linuxbridge-2 https://zuul.opendev.org/t/openstack/build/27ec891dd654449cbdf1b8a1ba4227ca : TIMED_OUT in 2h 50m 14s\n- neutron-tempest-plugin-linuxbridge-3 https://zuul.opendev.org/t/openstack/build/e7cbaeb2021540e785404c4e00b32f23 : TIMED_OUT in 2h 50m 27s\n- neutron-tempest-plugin-linuxbridge-4 https://zuul.opendev.org/t/openstack/build/1592885ad5e74f0484f846f05973c2a3 : TIMED_OUT in 2h 50m 01s\n- neutron-tempest-plugin-linuxbridge-5 https://zuul.opendev.org/t/openstack/build/ea0d46622d2b448daeda83386f16de65 : SUCCESS in 2h 30m 33s","accounts_in_message":[],"_revision_number":4},{"id":"40bec819e1d7b9780873bc11ea644abb36e943e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-12-29 17:07:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"750760f53f6dc21a0208aa1e79e9b4cdf4f7c88d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-29 20:16:24.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d7d23e00aa34545ba3dae4953dffdee\n\n- neutron-tempest-plugin-ovn-1 https://zuul.opendev.org/t/openstack/build/ffb26b920cc34fe5b29cff2c5099bc39 : TIMED_OUT in 3h 03m 07s\n- neutron-tempest-plugin-ovn-2 https://zuul.opendev.org/t/openstack/build/a86d7badfb71495d80a69acc713e57fc : TIMED_OUT in 3h 03m 20s\n- neutron-tempest-plugin-ovn-3 https://zuul.opendev.org/t/openstack/build/f5ec546602b64bfabd957998f75b86c6 : FAILURE in 2h 04m 37s\n- neutron-tempest-plugin-ovn-4 https://zuul.opendev.org/t/openstack/build/a6bdfede2d84495c9c098db5f74c394c : SUCCESS in 1h 01m 58s\n- neutron-tempest-plugin-ovn-5 https://zuul.opendev.org/t/openstack/build/30e199cb4ce045b18e6de0a23e2248ae : SUCCESS in 48m 35s\n- neutron-tempest-plugin-openvswitch-1 https://zuul.opendev.org/t/openstack/build/635df4d5b7e34164a10c0175f9a66ca3 : TIMED_OUT in 2h 49m 31s\n- neutron-tempest-plugin-openvswitch-2 https://zuul.opendev.org/t/openstack/build/de2146b2b12744639debbd1b31b01bbd : TIMED_OUT in 2h 49m 45s\n- neutron-tempest-plugin-openvswitch-3 https://zuul.opendev.org/t/openstack/build/3b98240772f745489a6f02995e412c33 : TIMED_OUT in 2h 50m 03s\n- neutron-tempest-plugin-openvswitch-4 https://zuul.opendev.org/t/openstack/build/b23ad2c3e6e445d7a2f743ad0591242f : SUCCESS in 1h 00m 02s\n- neutron-tempest-plugin-openvswitch-5 https://zuul.opendev.org/t/openstack/build/50d74f67c8cb4d96b9fe3c3153f2c972 : SUCCESS in 58m 37s\n- neutron-tempest-plugin-linuxbridge-1 https://zuul.opendev.org/t/openstack/build/4da8baf2dab24cfd8b3c9dc79a949aff : TIMED_OUT in 2h 50m 02s\n- neutron-tempest-plugin-linuxbridge-2 https://zuul.opendev.org/t/openstack/build/86fc2fb40c434972acb7f17a407afa86 : TIMED_OUT in 2h 49m 55s\n- neutron-tempest-plugin-linuxbridge-3 https://zuul.opendev.org/t/openstack/build/8a2178c76f36426e9ed6aa02f21543ff : SUCCESS in 2h 49m 26s\n- neutron-tempest-plugin-linuxbridge-4 https://zuul.opendev.org/t/openstack/build/9e635e2d44174f058cd31c4d316c9f54 : SUCCESS in 1h 08m 10s\n- neutron-tempest-plugin-linuxbridge-5 https://zuul.opendev.org/t/openstack/build/f655542844404b629c9408b503c4b371 : SUCCESS in 52m 20s","accounts_in_message":[],"_revision_number":5},{"id":"0b720d61e3894e172f84a44a0287b2dc2208f6e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-12-30 14:12:27.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3fc3ea8186bd8b0817080d22f6f0dc6d85d50612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-30 17:21:33.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a019b6b8c2ce4709acbcd761f207e139\n\n- neutron-tempest-plugin-ovn-1 https://zuul.opendev.org/t/openstack/build/556685d4552b4c3e9ce5c132c0c0e36e : TIMED_OUT in 3h 03m 30s\n- neutron-tempest-plugin-ovn-2 https://zuul.opendev.org/t/openstack/build/48ddd065351c4444bce55aa32b953c09 : FAILURE in 2h 46m 44s\n- neutron-tempest-plugin-ovn-3 https://zuul.opendev.org/t/openstack/build/711b674528274ce88d4f8bde43a2bb70 : SUCCESS in 2h 27m 14s\n- neutron-tempest-plugin-ovn-4 https://zuul.opendev.org/t/openstack/build/d3d68dc6f5f14ac28362cb22eb8fd918 : SUCCESS in 33m 18s\n- neutron-tempest-plugin-ovn-5 https://zuul.opendev.org/t/openstack/build/eeeac22527f249479261ef4d8f273fa2 : SUCCESS in 33m 05s\n- neutron-tempest-plugin-openvswitch-1 https://zuul.opendev.org/t/openstack/build/b2cbe4f49d7e43ee82de10b7d2a401e8 : SUCCESS in 2h 34m 41s\n- neutron-tempest-plugin-openvswitch-2 https://zuul.opendev.org/t/openstack/build/c0f7bd87190a4bf19cf27e41bb565807 : SUCCESS in 3h 02m 25s\n- neutron-tempest-plugin-openvswitch-3 https://zuul.opendev.org/t/openstack/build/218150668592422696281dec5884d783 : SUCCESS in 1h 58m 21s\n- neutron-tempest-plugin-openvswitch-4 https://zuul.opendev.org/t/openstack/build/25d045ad11bb4e59b01602e16be4ff31 : SUCCESS in 41m 38s\n- neutron-tempest-plugin-openvswitch-5 https://zuul.opendev.org/t/openstack/build/ec5cadf5a57845fd94e809e3da945413 : SUCCESS in 1h 02m 11s\n- neutron-tempest-plugin-linuxbridge-1 https://zuul.opendev.org/t/openstack/build/db8f3b2a1d144ae4a1439ed857a6857c : SUCCESS in 2h 40m 56s\n- neutron-tempest-plugin-linuxbridge-2 https://zuul.opendev.org/t/openstack/build/074cf5cb0197411593ca76ea3abb23d0 : TIMED_OUT in 3h 03m 28s\n- neutron-tempest-plugin-linuxbridge-3 https://zuul.opendev.org/t/openstack/build/3298745e2e8e49b3896fd95a3bde7c7d : SUCCESS in 2h 05m 49s\n- neutron-tempest-plugin-linuxbridge-4 https://zuul.opendev.org/t/openstack/build/f3eeda0acdcf489ea54fce655b8e0391 : SUCCESS in 42m 33s\n- neutron-tempest-plugin-linuxbridge-5 https://zuul.opendev.org/t/openstack/build/755b03892abb4f5eb9b23f19b0f90639 : SUCCESS in 40m 31s","accounts_in_message":[],"_revision_number":6},{"id":"3000ae49931ae4eaae9b76f4341ad34ba354d6b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-01-01 13:51:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fcee55da69ca6ad6e3a4fe71a9d36aa94761675b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-01 17:00:25.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb37ea112c184af7a8b26f7cee8d1464\n\n- neutron-tempest-plugin-ovn-1 https://zuul.opendev.org/t/openstack/build/1c449e91e30644618a9feb32d4c7b46b : TIMED_OUT in 3h 02m 40s\n- neutron-tempest-plugin-ovn-2 https://zuul.opendev.org/t/openstack/build/b3feed2fb6fe4b259d5d6a5a08d76e09 : TIMED_OUT in 3h 03m 01s\n- neutron-tempest-plugin-ovn-3 https://zuul.opendev.org/t/openstack/build/c3bc14b0ce024524845e67bdbf8d1740 : FAILURE in 2h 34m 21s\n- neutron-tempest-plugin-ovn-4 https://zuul.opendev.org/t/openstack/build/ffb438ad02904c2b9251138f8de69851 : SUCCESS in 46m 48s\n- neutron-tempest-plugin-ovn-5 https://zuul.opendev.org/t/openstack/build/131abe576e574fd6a529f00b5a1d438d : FAILURE in 3h 03m 12s\n- neutron-tempest-plugin-openvswitch-1 https://zuul.opendev.org/t/openstack/build/552965ad028b49adaaa94c067d3c5d12 : SUCCESS in 2h 02m 38s\n- neutron-tempest-plugin-openvswitch-2 https://zuul.opendev.org/t/openstack/build/72c88ccf3ea84cbfac894caf6b720b44 : SUCCESS in 2h 54m 42s\n- neutron-tempest-plugin-openvswitch-3 https://zuul.opendev.org/t/openstack/build/b6233f7d3d9c461bb07d3a55f66740f4 : TIMED_OUT in 3h 02m 33s\n- neutron-tempest-plugin-openvswitch-4 https://zuul.opendev.org/t/openstack/build/005abe99b531465399668f722e32a972 : SUCCESS in 1h 00m 20s\n- neutron-tempest-plugin-openvswitch-5 https://zuul.opendev.org/t/openstack/build/6a94c8eb7f9b47cca297322bf43b98f8 : TIMED_OUT in 3h 03m 06s\n- neutron-tempest-plugin-linuxbridge-1 https://zuul.opendev.org/t/openstack/build/86eb3026989f46d499f7641c6e16fbe5 : SUCCESS in 2h 52m 57s\n- neutron-tempest-plugin-linuxbridge-2 https://zuul.opendev.org/t/openstack/build/443de004ae0f402b82672a64979185a1 : SUCCESS in 2h 56m 42s\n- neutron-tempest-plugin-linuxbridge-3 https://zuul.opendev.org/t/openstack/build/9b31c9ad57e74627837189949515fb54 : TIMED_OUT in 3h 03m 15s\n- neutron-tempest-plugin-linuxbridge-4 https://zuul.opendev.org/t/openstack/build/8670de34b3d74c39a56101c9ffabf34a : SUCCESS in 57m 08s\n- neutron-tempest-plugin-linuxbridge-5 https://zuul.opendev.org/t/openstack/build/2d6c03a60e4e40d585c5119c19772a5d : TIMED_OUT in 3h 03m 26s","accounts_in_message":[],"_revision_number":7},{"id":"388cb5d1fa4b687b0312016f9a7af74700c438ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-01-03 15:52:15.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e5b81f2e4f1d42bc024b580ae77a5bf52bb4e507","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 18:58:05.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1923268527a4a24992e64163f547a03\n\n- neutron-tempest-plugin-ovn-1 https://zuul.opendev.org/t/openstack/build/a6ac36b2d086496a8cf8e022cd12a2b3 : SUCCESS in 2h 28m 31s\n- neutron-tempest-plugin-ovn-2 https://zuul.opendev.org/t/openstack/build/c0be0cf999b54f19a21dfbf5e665e2c0 : SUCCESS in 2h 15m 03s\n- neutron-tempest-plugin-ovn-3 https://zuul.opendev.org/t/openstack/build/aa7a1a9d1b7a484094471a93204a2f95 : SUCCESS in 2h 36m 13s\n- neutron-tempest-plugin-ovn-4 https://zuul.opendev.org/t/openstack/build/0dd46ebe9bf9476eaabe23eb9bc36845 : SUCCESS in 1h 00m 37s\n- neutron-tempest-plugin-ovn-5 https://zuul.opendev.org/t/openstack/build/f8b0df71f13a43d6a6d1becb32be2909 : TIMED_OUT in 3h 03m 33s\n- neutron-tempest-plugin-openvswitch-1 https://zuul.opendev.org/t/openstack/build/8c418a4038ce40849edf0cc2b27d2b4c : SUCCESS in 2h 09m 34s\n- neutron-tempest-plugin-openvswitch-2 https://zuul.opendev.org/t/openstack/build/8464a06c78664889943ce6d0cd763306 : SUCCESS in 2h 18m 44s\n- neutron-tempest-plugin-openvswitch-3 https://zuul.opendev.org/t/openstack/build/9e25fb054f6e4106a3b1c29eab98c4c4 : SUCCESS in 2h 08m 44s\n- neutron-tempest-plugin-openvswitch-4 https://zuul.opendev.org/t/openstack/build/983cecc36a814f9282252e0572525b4f : SUCCESS in 1h 01m 34s\n- neutron-tempest-plugin-openvswitch-5 https://zuul.opendev.org/t/openstack/build/cba75fdfed174341aede6e4a7bc59d7f : SUCCESS in 3h 01m 40s\n- neutron-tempest-plugin-linuxbridge-1 https://zuul.opendev.org/t/openstack/build/3b8ac41573b74bf08aef5d30bcdca67d : TIMED_OUT in 3h 03m 24s\n- neutron-tempest-plugin-linuxbridge-2 https://zuul.opendev.org/t/openstack/build/882686eeebf34760a53cdbf2eb891d32 : SUCCESS in 2h 47m 35s\n- neutron-tempest-plugin-linuxbridge-3 https://zuul.opendev.org/t/openstack/build/f4444365222244cc8aeb0dd15ab0de1a : SUCCESS in 2h 48m 37s\n- neutron-tempest-plugin-linuxbridge-4 https://zuul.opendev.org/t/openstack/build/8dd91858e44647e3bab47ad079193bba : SUCCESS in 53m 46s\n- neutron-tempest-plugin-linuxbridge-5 https://zuul.opendev.org/t/openstack/build/36bdf940df5f4c53a4a2ee60b59ac354 : SUCCESS in 2h 49m 00s","accounts_in_message":[],"_revision_number":8},{"id":"94acac966a9481955c36cc72b9f0f99cada0b232","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-17 08:03:32.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"a13e9d35cc0590029a7871ab6541cd54649b6110","revisions":{"7bbeffe102e2e63c553f09ca7d78a45f59b52e86":{"kind":"REWORK","_number":1,"created":"2022-12-16 06:38:15.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/34/867934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/34/867934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/1"}}},"commit":{"parents":[{"commit":"f8eda8fdacd3e2ca6a7e42262467bf974c321319","subject":"Merge \"Update nested-virt testing for the 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f8eda8fdacd3e2ca6a7e42262467bf974c321319"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-13 07:09:50.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-16 06:38:17.000000000","tz":330},"subject":"[DNM] Test lower concurrency on non nested provider","message":"[DNM] Test lower concurrency on non nested provider\n\nworkaround https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/867320\n\nChange-Id: Ib242dfeca4f348a30f31729a7c7584af0c50edf8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7bbeffe102e2e63c553f09ca7d78a45f59b52e86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7bbeffe102e2e63c553f09ca7d78a45f59b52e86"}]},"branch":"refs/heads/master"},"39b26e9a16e0df763de733ee189ca2cfe26e9d29":{"kind":"REWORK","_number":2,"created":"2022-12-16 10:32:18.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/34/867934/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/34/867934/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/2"}}},"commit":{"parents":[{"commit":"f8eda8fdacd3e2ca6a7e42262467bf974c321319","subject":"Merge \"Update nested-virt testing for the 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f8eda8fdacd3e2ca6a7e42262467bf974c321319"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-13 07:09:50.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-16 10:31:53.000000000","tz":330},"subject":"[DNM] Test lower concurrency on non nested provider","message":"[DNM] Test lower concurrency on non nested provider\n\nCheck if zram helps, it will create half the\nswap of total memory.\n\nChange-Id: Ib242dfeca4f348a30f31729a7c7584af0c50edf8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/39b26e9a16e0df763de733ee189ca2cfe26e9d29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/39b26e9a16e0df763de733ee189ca2cfe26e9d29"}]},"branch":"refs/heads/master"},"a82c8fa7a874006550b680399f61d6a217b2568a":{"kind":"REWORK","_number":3,"created":"2022-12-16 15:46:49.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/34/867934/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/34/867934/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/3"}}},"commit":{"parents":[{"commit":"f8eda8fdacd3e2ca6a7e42262467bf974c321319","subject":"Merge \"Update nested-virt testing for the 2023.1 cycle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f8eda8fdacd3e2ca6a7e42262467bf974c321319"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-13 07:09:50.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-16 15:42:05.000000000","tz":330},"subject":"[DNM] Test lower concurrency on non nested provider","message":"[DNM] Test lower concurrency on non nested provider\n\nChange-Id: Ib242dfeca4f348a30f31729a7c7584af0c50edf8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a82c8fa7a874006550b680399f61d6a217b2568a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a82c8fa7a874006550b680399f61d6a217b2568a"}]},"branch":"refs/heads/master"},"5aa7a97deec0c83cc56ecc2c204a40f541ecd75d":{"kind":"REWORK","_number":4,"created":"2022-12-22 15:02:17.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/34/867934/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/34/867934/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/4"}}},"commit":{"parents":[{"commit":"021ce916c6b72f15fa1b873b2c01c1409ca0b37d","subject":"Merge \"Revert \"Update nested-virt testing for the 2023.1 cycle\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/021ce916c6b72f15fa1b873b2c01c1409ca0b37d"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-13 07:09:50.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-22 15:01:58.000000000","tz":330},"subject":"[DNM] Test tb-cache workaround for jammy","message":"[DNM] Test tb-cache workaround for jammy\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/868419\nChange-Id: Ib242dfeca4f348a30f31729a7c7584af0c50edf8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5aa7a97deec0c83cc56ecc2c204a40f541ecd75d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5aa7a97deec0c83cc56ecc2c204a40f541ecd75d"}]},"branch":"refs/heads/master"},"ce4ba0f0ce1918de1a551fcf2a6ff30069607ab2":{"kind":"REWORK","_number":5,"created":"2022-12-29 17:07:18.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/34/867934/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/34/867934/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/5"}}},"commit":{"parents":[{"commit":"021ce916c6b72f15fa1b873b2c01c1409ca0b37d","subject":"Merge \"Revert \"Update nested-virt testing for the 2023.1 cycle\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/021ce916c6b72f15fa1b873b2c01c1409ca0b37d"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-13 07:09:50.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-29 17:06:37.000000000","tz":330},"subject":"[DNM] Test tb-cache workaround for jammy","message":"[DNM] Test tb-cache workaround for jammy\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/868419\nChange-Id: Ib242dfeca4f348a30f31729a7c7584af0c50edf8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ce4ba0f0ce1918de1a551fcf2a6ff30069607ab2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ce4ba0f0ce1918de1a551fcf2a6ff30069607ab2"}]},"branch":"refs/heads/master"},"b65e2ed75b40739b8d1cc345256b61d60c1d40c2":{"kind":"REWORK","_number":6,"created":"2022-12-30 14:12:27.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/34/867934/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/34/867934/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/6"}}},"commit":{"parents":[{"commit":"021ce916c6b72f15fa1b873b2c01c1409ca0b37d","subject":"Merge \"Revert \"Update nested-virt testing for the 2023.1 cycle\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/021ce916c6b72f15fa1b873b2c01c1409ca0b37d"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-13 07:09:50.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-30 14:12:36.000000000","tz":330},"subject":"[DNM] Test tb-cache workaround for jammy","message":"[DNM] Test tb-cache workaround for jammy\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/868419\nChange-Id: Ib242dfeca4f348a30f31729a7c7584af0c50edf8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b65e2ed75b40739b8d1cc345256b61d60c1d40c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b65e2ed75b40739b8d1cc345256b61d60c1d40c2"}]},"branch":"refs/heads/master"},"44b2a739b2bc9bfa776fe55f66f6e471c6d392f0":{"kind":"REWORK","_number":7,"created":"2023-01-01 13:51:20.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/34/867934/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/34/867934/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/7"}}},"commit":{"parents":[{"commit":"021ce916c6b72f15fa1b873b2c01c1409ca0b37d","subject":"Merge \"Revert \"Update nested-virt testing for the 2023.1 cycle\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/021ce916c6b72f15fa1b873b2c01c1409ca0b37d"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-13 07:09:50.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-01-01 13:50:49.000000000","tz":330},"subject":"[DNM] Test tb-cache workaround for jammy","message":"[DNM] Test tb-cache workaround for jammy\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/868419\nChange-Id: Ib242dfeca4f348a30f31729a7c7584af0c50edf8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/44b2a739b2bc9bfa776fe55f66f6e471c6d392f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/44b2a739b2bc9bfa776fe55f66f6e471c6d392f0"}]},"branch":"refs/heads/master"},"a13e9d35cc0590029a7871ab6541cd54649b6110":{"kind":"REWORK","_number":8,"created":"2023-01-03 15:52:15.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/34/867934/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/34/867934/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/34/867934/8"}}},"commit":{"parents":[{"commit":"021ce916c6b72f15fa1b873b2c01c1409ca0b37d","subject":"Merge \"Revert \"Update nested-virt testing for the 2023.1 cycle\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/021ce916c6b72f15fa1b873b2c01c1409ca0b37d"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-13 07:09:50.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-01-03 15:52:27.000000000","tz":330},"subject":"[DNM] Test tb-cache workaround for jammy","message":"[DNM] Test tb-cache workaround for jammy\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/868419\nChange-Id: Ib242dfeca4f348a30f31729a7c7584af0c50edf8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a13e9d35cc0590029a7871ab6541cd54649b6110"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a13e9d35cc0590029a7871ab6541cd54649b6110"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
