)]}'
{"id":"openstack%2Fneutron-tempest-plugin~868948","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I65681721ebdda011672862f729023961aecba047","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"dnm","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-04-17 08:03:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I65681721ebdda011672862f729023961aecba047","subject":"[DNM] Test with default tb-cache","status":"ABANDONED","created":"2023-01-01 13:54:27.000000000","updated":"2023-04-17 08:03:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d670338d87d7c4b8da1aea3e69b314af22f48a40","_number":868948,"virtual_id_number":868948,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-01 17:04:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-01 17:04:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-17 08:03:37.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"4f102569e42d46d1064a5ff73f8e5dfd6cce51a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-01-01 13:54:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22e725278007c63b89ee7f1ab97aa3d72c6d150b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-01 17:04:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1fb3fcfd2f004af1ac5ef71bc7b09703\n\n- neutron-tempest-plugin-ovn-1 https://zuul.opendev.org/t/openstack/build/893a863fd4924042abb0894c40c6a843 : TIMED_OUT in 3h 02m 31s\n- neutron-tempest-plugin-ovn-2 https://zuul.opendev.org/t/openstack/build/51aee0ac4968467681c435d2c9258a9c : FAILURE in 2h 05m 43s\n- neutron-tempest-plugin-ovn-3 https://zuul.opendev.org/t/openstack/build/b6cf1bab165945b4832d8271fa02f01f : SUCCESS in 2h 22m 29s\n- neutron-tempest-plugin-ovn-4 https://zuul.opendev.org/t/openstack/build/3cd9947c2e6e45f3b92159a282c39c83 : SUCCESS in 44m 05s\n- neutron-tempest-plugin-ovn-5 https://zuul.opendev.org/t/openstack/build/f22f40042c174f8299c18ad0430a147f : TIMED_OUT in 3h 03m 08s\n- neutron-tempest-plugin-openvswitch-1 https://zuul.opendev.org/t/openstack/build/8437b435b18946e68be9f1f21b086784 : FAILURE in 2h 42m 14s\n- neutron-tempest-plugin-openvswitch-2 https://zuul.opendev.org/t/openstack/build/1402fae43b6542c19a7fc560d65ed365 : TIMED_OUT in 3h 02m 49s\n- neutron-tempest-plugin-openvswitch-3 https://zuul.opendev.org/t/openstack/build/3e730f6df00f4a7d8efbbdcdc10271a5 : FAILURE in 2h 11m 51s\n- neutron-tempest-plugin-openvswitch-4 https://zuul.opendev.org/t/openstack/build/6409a713ec884bcbbcb2c7d26b4fa819 : SUCCESS in 51m 05s\n- neutron-tempest-plugin-openvswitch-5 https://zuul.opendev.org/t/openstack/build/f8f1af89d5224379969a9e7de2743d9d : FAILURE in 1h 53m 28s\n- neutron-tempest-plugin-linuxbridge-1 https://zuul.opendev.org/t/openstack/build/8907c7c3dc3a45218b2550fe467194a5 : FAILURE in 2h 49m 07s\n- neutron-tempest-plugin-linuxbridge-2 https://zuul.opendev.org/t/openstack/build/ed2db1681adf4a769e731341a10ba099 : FAILURE in 2h 38m 53s\n- neutron-tempest-plugin-linuxbridge-3 https://zuul.opendev.org/t/openstack/build/83a6eff774594afb9a4986880c6ce4af : FAILURE in 2h 33m 18s\n- neutron-tempest-plugin-linuxbridge-4 https://zuul.opendev.org/t/openstack/build/f7d060051cfc4f35b8415f7f29b97af1 : SUCCESS in 49m 12s\n- neutron-tempest-plugin-linuxbridge-5 https://zuul.opendev.org/t/openstack/build/59049115c72942c79acb7da77e93c635 : FAILURE in 2h 25m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d670338d87d7c4b8da1aea3e69b314af22f48a40","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-17 08:03:37.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2901d55d614cf73c43a4bb6666759e49745853c5","revisions":{"2901d55d614cf73c43a4bb6666759e49745853c5":{"kind":"REWORK","_number":1,"created":"2023-01-01 13:54:27.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/48/868948/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/48/868948/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/48/868948/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/48/868948/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/48/868948/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/48/868948/1"}}},"commit":{"parents":[{"commit":"021ce916c6b72f15fa1b873b2c01c1409ca0b37d","subject":"Merge \"Revert \"Update nested-virt testing for the 2023.1 cycle\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/021ce916c6b72f15fa1b873b2c01c1409ca0b37d"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-12-13 07:09:50.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-01-01 13:54:37.000000000","tz":330},"subject":"[DNM] Test with default tb-cache","message":"[DNM] Test with default tb-cache\n\nChange-Id: I65681721ebdda011672862f729023961aecba047\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2901d55d614cf73c43a4bb6666759e49745853c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2901d55d614cf73c43a4bb6666759e49745853c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
