)]}'
{"id":"openstack%2Fneutron-tempest-plugin~882818","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Iadf69c167cdda0017084e482a58116520a1ea80f","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2018585","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-05-11 02:58:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iadf69c167cdda0017084e482a58116520a1ea80f","subject":"[S-RBAC] Update DSCP marking rule create API test","status":"MERGED","created":"2023-05-10 10:52:34.000000000","updated":"2023-05-11 02:58:28.000000000","submitted":"2023-05-11 02:58:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882818-bug/2018585","meta_rev_id":"14bf84672bbc338cc20c768fcb1439e535c6ec7d","_number":882818,"virtual_id_number":882818,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-11 02:58:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-05-10 13:10:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-05-10 23:27:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-05-10 23:27:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-05-10 13:10:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-10 13:09:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-10 13:10:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-05-10 23:27:52.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"4fdfb70c892077f2d32aa610bae8146f004875c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-10 10:52:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b279fd7c0c443c3aeb13438e61b517ca4b2bd79d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 13:09:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1050f6189fef413ab5f3ebc8bee5e6bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0813fa9c387144d7bccebb1e039c631e : SUCCESS in 3m 53s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/1ab6e171432e4fec9d4d721ee1db0260 : SUCCESS in 1h 48m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0b21a8802180425d98ea48989d06f6a5 : SUCCESS in 1h 54m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ba7f3067127944999fee630089c44794 : SUCCESS in 2h 13m 53s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b80a5de4471543928b853fc42fa1f23c : SUCCESS in 1h 41m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2bc8f83f06414d68a664e0ddc4b1898b : SUCCESS in 1h 54m 21s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/5213e71aba3947ebaf4fe6c7a5c1726e : SUCCESS in 50m 30s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/8a9975053f3d4c17987316f9ceabce17 : SUCCESS in 2h 13m 39s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/fbd8331e18964cbd9a67f3dca52f9742 : SUCCESS in 1h 30m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/ddb2704019244c2aac131ce414a3e7b6 : SUCCESS in 1h 45m 57s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/48c48d17d3764c7891c8b7605de93e73 : SUCCESS in 1h 27m 34s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/dec7a78c5ee842e2aac3ce7a34a51ee3 : SUCCESS in 1h 40m 45s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/92df1b7d75de44b89ac53413755fefa7 : SUCCESS in 1h 51m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/a69677dc9dbb41dda6b4cacf64114b86 : SUCCESS in 2h 14m 21s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/491dfb238762467b9d4a7bea03f9c742 : SUCCESS in 2h 02m 09s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/7d36aa636c7f43e0a0449d85377d1c45 : SUCCESS in 2h 09m 20s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/25505f9f481f482b929188db0d0aa201 : SUCCESS in 1h 37m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/555db9407cb34ec7a7ec7610332e0ff5 : SUCCESS in 1h 35m 38s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/db54d07e89594fb996c38ca571c7a754 : SUCCESS in 1h 47m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/166400fcdc7b4f8a80d43c9276ab8555 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"1a09401f0986c24304c6ec229d786df448461f45","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-10 13:10:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c583457e35eba3eb7c1de28634783a51340d1f26","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-10 13:10:27.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"ac5759683b4e678fec90a041f5d83e91ba30587b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-05-10 23:27:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aafa038a8b4496dd86ccce19a796c53912b3ab68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 23:28:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d60172f68783f66b07d069a9363e0c8534ec96bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 02:58:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50c67c14bfcf47dd9d4036390522c940\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd6007d057c949088792d5277ade2e04 : SUCCESS in 3m 32s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/8430b8a045ea4523a217fa6f0b368efd : SUCCESS in 2h 05m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/123926d908634ab3b68b14c67e281d19 : SUCCESS in 2h 03m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0e7a1473d3544d90b489ea8af9fe69d3 : SUCCESS in 2h 06m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e838b59031364441ab11883fd5f077a9 : SUCCESS in 2h 20m 46s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e19557826dca462ba5cfda44ad72092d : SUCCESS in 1h 45m 16s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/81f4cf10753a4aa693fc45416451a74a : SUCCESS in 49m 44s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/883ee515c78141e19c5663d85e00782a : SUCCESS in 2h 14m 42s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/143971b357744e7497137447d5e2def9 : SUCCESS in 1h 43m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cc6d155428b442eba0b1a1b06077305 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":1},{"id":"14bf84672bbc338cc20c768fcb1439e535c6ec7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 02:58:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fdfb4741f60a0309939f82f7aa44a11c5f2bbd31","revisions":{"fdfb4741f60a0309939f82f7aa44a11c5f2bbd31":{"kind":"REWORK","_number":1,"created":"2023-05-10 10:52:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/18/882818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/18/882818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/18/882818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/18/882818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/18/882818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/18/882818/1"}}},"commit":{"parents":[{"commit":"3e247e5c769d6759aa56ba959970f3ef2688c49a","subject":"Merge \"Get nc client\u0027s command based on the nc on correct host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3e247e5c769d6759aa56ba959970f3ef2688c49a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-10 10:48:49.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-10 10:52:31.000000000","tz":120},"subject":"[S-RBAC] Update DSCP marking rule create API test","message":"[S-RBAC] Update DSCP marking rule create API test\n\nIn the test_qos_dscp_create_and_update API test, qos policy was made as\nadmin user (which is correct) but was also owned by admin project. And\nlater, to check if DSCP marking rule was created in that policy\nproperly, regular client is used instead.\nThe problem is that with new S-RBAC API policies, rules are visible to\nowners of the policy, not to all users. And due to that this test is\nfailing with new S-RBAC policies enforced.\n\nThis patch fixes it by changing owner of the qos policy to the regular\nclient\u0027s project.\n\nRelated-Bug: #2018727\nChange-Id: Iadf69c167cdda0017084e482a58116520a1ea80f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fdfb4741f60a0309939f82f7aa44a11c5f2bbd31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fdfb4741f60a0309939f82f7aa44a11c5f2bbd31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
