)]}'
{"id":"openstack%2Fneutron-tempest-plugin~883826","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Id328446be8591fac3affe2cf8adeaeee09697085","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-06-05 09:06:52.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"30380":{"account":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"last_update":"2023-06-06 23:34:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id328446be8591fac3affe2cf8adeaeee09697085","subject":"[fwaas]Disassociate all ports before deleting firewall group","status":"MERGED","created":"2023-05-25 00:55:47.000000000","updated":"2023-06-06 23:34:07.000000000","submitted":"2023-06-06 23:34:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883826","meta_rev_id":"532860d35993009cccf6476891bf5495b0534281","_number":883826,"virtual_id_number":883826,"owner":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-06 23:34:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2023-06-06 21:37:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-06-05 09:06:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2023-06-06 21:37:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-25 02:00:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-25 14:18:04.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-06-06 21:37:26.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"af4653f98d202fb2bf968cf6917d56eea098c9ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2023-05-25 00:55:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1996496cac9a3e93d45b455cf7f0b553115f54f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2023-05-25 01:06:13.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b39ce1679cb914fa7828f0c3b8f1a9238827d0c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 02:00:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cdce35c672345aa841236f3005548e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4d86231be6468f9fcd665af2a3a56b : SUCCESS in 3m 00s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/cc5293216ef54e319d792a01aa4f5321 : SUCCESS in 42m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8656dd0454ae47b8b03e3579078a6424 : SUCCESS in 2m 48s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/6bdbfc803c864e14a394427c8edb7cea : SUCCESS in 29m 48s\n- neutron-tempest-plugin-fwaas-zed https://zuul.opendev.org/t/openstack/build/705fa235b3b747738a466a6d7fe69e22 : SUCCESS in 49m 05s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/7324ad032ab148f5a80ed8dc30a493b2 : FAILURE in 29m 17s","accounts_in_message":[],"_revision_number":2},{"id":"1051869afeac4c6a2e2247b7ecdc726ec3f7f528","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2023-05-25 02:14:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1d0913a570a57ed0cc12ccaa2b41b05043bd7109","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 03:44:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac122a707c7f4799a5806dbff42e6f09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0857210fa7e2499a8bd1b02bebd7051b : SUCCESS in 3m 55s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/05c7b71573dc40af81be77ade2682e80 : SUCCESS in 1h 08m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eec728bc30d04eeabf90699993fe3099 : SUCCESS in 3m 08s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/72c45836f73d4225af33972870df101a : FAILURE in 59m 31s\n- neutron-tempest-plugin-fwaas-zed https://zuul.opendev.org/t/openstack/build/c73877d2eba643b1a09a42871435e713 : FAILURE in 43m 34s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/8f1dc801365e4c9bb8f960d4bf9e9118 : FAILURE in 48m 09s","accounts_in_message":[],"_revision_number":3},{"id":"065ff115e2f16c5bc841da82f95ce1ddf8ced57c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2023-05-25 05:53:22.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"92c8033e5587d44d19b3cfe5ea558571be3f56fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 06:51:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26dedbec0ea84562b237b915f472c873\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43ac6450cc7344c6bb2cbe82900acda5 : SUCCESS in 3m 26s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/6e5d936aee204626bdd4f1f991b96fbc : SUCCESS in 39m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ba8642791e7416fbefdcae80351e435 : SUCCESS in 3m 01s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/bede4ad54a044b6f8249828ade4eea30 : SUCCESS in 46m 48s\n- neutron-tempest-plugin-fwaas-zed https://zuul.opendev.org/t/openstack/build/839b29e0fdd147df8cc3d611a3cf4b60 : SUCCESS in 43m 55s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/0e31cb3bf28f419a87287ce5616d35e0 : SUCCESS in 50m 40s","accounts_in_message":[],"_revision_number":4},{"id":"04f0558a2b662e141ab8c9a373c595e464f1449f","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2023-05-25 07:05:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"48b3c8ec5caf2091875dda8b951f05f87f68bf42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2023-05-25 07:57:43.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"8d57996f54188019e1e351d049191174ad022c7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 08:59:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31d8d881d3cb4575ad0551cb865bf5ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0cd7f8616444c6d8e771bff93ba52a3 : SUCCESS in 3m 00s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/a99922bc37534c9e8f89204bad8e8464 : SUCCESS in 58m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d957beb897f4bd497dbda185be27d78 : SUCCESS in 2m 38s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/678b3307a72c4d4c9909916d1ca4de52 : SUCCESS in 56m 39s\n- neutron-tempest-plugin-fwaas-zed https://zuul.opendev.org/t/openstack/build/07616898517a42d6ba5ccf915d269cd8 : SUCCESS in 40m 28s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/30c32ddfadf4409f9b1f6eadef1eac8b : SUCCESS in 53m 12s","accounts_in_message":[],"_revision_number":5},{"id":"702fb630e2f0569be45af7769eb0641af12353e2","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2023-05-25 09:09:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f838a14a9c086903ee22ad75e2380ce9601efb62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 10:03:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4813317615c84b629c715ef390ec8c90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b7a97b5131c4315b52b19fafc8d11eb : SUCCESS in 2m 59s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/f517979a62e9431f883f73ec2f2ed0b7 : SUCCESS in 51m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6d29cab65b14e829fcf3ec6d4b2c4a9 : SUCCESS in 2m 53s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/44c06ca65c654638bd264885e85e4cda : SUCCESS in 39m 22s\n- neutron-tempest-plugin-fwaas-zed https://zuul.opendev.org/t/openstack/build/e2d7ffcb7d3f41feb9f5ceffc073286c : FAILURE in 41m 59s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/2a68383f778f472aa045209bcf5e57a9 : SUCCESS in 48m 51s","accounts_in_message":[],"_revision_number":5},{"id":"bd061ffddec1121d9b8fbc45800284550f2a1459","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2023-05-25 11:30:44.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"fdf8f10a635c17bf0a8c13b8364a910e0782b6c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 12:32:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b58c5765186048d8b88a2c6c4bec5738\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28ccd181af0a4ff1a7d9b0ef4b08b11e : SUCCESS in 3m 59s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/7b1c7fffae5049b3bd421bcb3cfc557b : SUCCESS in 56m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1791f0a064634ab4822bba0cfc7b7707 : SUCCESS in 3m 27s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/79d5f9f1d99d4e0c93bb464c64297fbe : SUCCESS in 40m 49s\n- neutron-tempest-plugin-fwaas-zed https://zuul.opendev.org/t/openstack/build/f6420d0c3c6d433f8650cf49c84ef29a : SUCCESS in 36m 26s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/89bfcccc45a04c38b351822b97e3ac85 : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":6},{"id":"465859b7388e3db4a08219e3001e4da9ffd033bb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-25 14:18:04.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e21f7243f8764f2b269f0f53470c10e0e3320939","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2023-05-26 00:17:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"55743b106ca223fc2b73575fede440f3b664f33a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 01:36:52.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e499d854d8a4253807123295f8cbbb1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dcf29520932427e8776bc0afbd38058 : SUCCESS in 3m 51s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/965e5785bcb1407ca3db40ca6a740a37 : SUCCESS in 1h 14m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18d8c98926274c4f9717e1855b263b54 : SUCCESS in 3m 05s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/6a7727d392654444a09db4ddf7df7951 : FAILURE in 44m 36s\n- neutron-tempest-plugin-fwaas-zed https://zuul.opendev.org/t/openstack/build/fd031c0317c843339aaf0c708cfc6d62 : SUCCESS in 34m 12s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/ae8107591cf546358b1d84491e998f39 : SUCCESS in 38m 25s","accounts_in_message":[],"_revision_number":6},{"id":"7932a70058c68af31448aedebeb11d0a4ccf95be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2023-05-29 08:03:31.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"917f7d20a1e0068a042e69c4794502987af69647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 09:24:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efbea9f765584166acaa8e86fa8fea34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1a8f298caa94313ac98d544781927b8 : SUCCESS in 3m 01s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/044844f8f18743d484a530cad3321f3d : SUCCESS in 55m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41decfec70e24f4a9a881e6edaae0034 : SUCCESS in 2m 52s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/fe7629d99d1a40c3882f7bb849b0cfe5 : SUCCESS in 48m 12s\n- neutron-tempest-plugin-fwaas-zed https://zuul.opendev.org/t/openstack/build/c82e6cc2f6414ff7847685e1c6760fa1 : SUCCESS in 49m 23s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/55849ed3397f4589b711f14e203f4a45 : SUCCESS in 46m 53s","accounts_in_message":[],"_revision_number":7},{"id":"74dcc932f077ce91b443bae56a1afc5b8455333c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-06-05 09:06:52.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"6bfe3f11170b794e19658befec470dbc6c0cac56","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-06-06 21:37:26.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"25008c9d40f3065ad67505ce90339d70393765c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 21:37:57.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"7f4a505d0096c0caaa74da75edea98f3ccf2658a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 23:34:07.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c02c202d513c4e8087100452db4a5124\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98d652bc136a488d89341d1e4a4b0ab6 : SUCCESS in 4m 34s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/8fca30c1b0df48159b92612586a23bde : SUCCESS in 47m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b46c9041e4944d4bfb41486d573c2c7 : SUCCESS in 3m 24s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/612a81c9c6044b589baef53aa940bccc : SUCCESS in 31m 48s","accounts_in_message":[],"_revision_number":7},{"id":"532860d35993009cccf6476891bf5495b0534281","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 23:34:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"7b1f9b11e66b81eabdc8361c89f8ccfdb3448f9a","revisions":{"c60c755417c39347b3eb27d38b4e57e287d496b4":{"kind":"REWORK","_number":1,"created":"2023-05-25 00:55:47.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/26/883826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/883826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/1"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2023-05-25 00:42:36.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2023-05-25 00:46:41.000000000","tz":480},"subject":"Disassociate all ports before delete firewall group","message":"Disassociate all ports before delete firewall group\n\nIf the firewall group has associated ports, the firewall group may be in an\nactive state. Firewall group is active, deletion is not allowed.\n\nCloses-Bug: #2018967\n\nChange-Id: Id328446be8591fac3affe2cf8adeaeee09697085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c60c755417c39347b3eb27d38b4e57e287d496b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c60c755417c39347b3eb27d38b4e57e287d496b4"}]},"branch":"refs/heads/master"},"eb82c95d39ef78296f63051e2dc4f9d194f60424":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-05-25 01:06:13.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/26/883826/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/883826/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/2"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2023-05-25 00:42:36.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2023-05-25 01:05:23.000000000","tz":480},"subject":"[fwaas]Disassociate all ports before deleting firewall group","message":"[fwaas]Disassociate all ports before deleting firewall group\n\nIf the firewall group has associated ports, the firewall group may be in\nan active state. Firewall group is active, deletion is not allowed.\n\nCloses-Bug: #2018967\n\nChange-Id: Id328446be8591fac3affe2cf8adeaeee09697085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/eb82c95d39ef78296f63051e2dc4f9d194f60424"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/eb82c95d39ef78296f63051e2dc4f9d194f60424"}]},"branch":"refs/heads/master"},"6c26d9ab4ff5c4a1cda5cb7f691c066c3c1bafae":{"kind":"REWORK","_number":3,"created":"2023-05-25 02:14:35.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/26/883826/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/883826/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/3"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2023-05-25 00:42:36.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2023-05-25 02:14:17.000000000","tz":480},"subject":"[fwaas]Disassociate all ports before deleting firewall group","message":"[fwaas]Disassociate all ports before deleting firewall group\n\nIf the firewall group has associated ports, the firewall group may be in\nan active state. Firewall group is active, deletion is not allowed.\n\nCloses-Bug: #2018967\n\nChange-Id: Id328446be8591fac3affe2cf8adeaeee09697085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6c26d9ab4ff5c4a1cda5cb7f691c066c3c1bafae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6c26d9ab4ff5c4a1cda5cb7f691c066c3c1bafae"}]},"branch":"refs/heads/master"},"af1bfba2e9e58e8332e6c6ad6e20de52a0227f83":{"kind":"REWORK","_number":4,"created":"2023-05-25 05:53:22.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/26/883826/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/883826/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/4"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2023-05-25 00:42:36.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2023-05-25 05:50:43.000000000","tz":480},"subject":"[fwaas]Disassociate all ports before deleting firewall group","message":"[fwaas]Disassociate all ports before deleting firewall group\n\nIf the firewall group has associated ports, the firewall group may be in\nan active state. Firewall group is active, deletion is not allowed.\n\nCloses-Bug: #2018967\n\nChange-Id: Id328446be8591fac3affe2cf8adeaeee09697085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/af1bfba2e9e58e8332e6c6ad6e20de52a0227f83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/af1bfba2e9e58e8332e6c6ad6e20de52a0227f83"}]},"branch":"refs/heads/master"},"865226c80b4e4e0e93f3ce5a417891a35cbfb43f":{"kind":"REWORK","_number":5,"created":"2023-05-25 07:57:43.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/26/883826/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/883826/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/5"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2023-05-25 00:42:36.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2023-05-25 07:54:40.000000000","tz":480},"subject":"[fwaas]Disassociate all ports before deleting firewall group","message":"[fwaas]Disassociate all ports before deleting firewall group\n\nIf the firewall group has associated ports, the firewall group may be in\nan active state. Firewall group is active, deletion is not allowed.\n\nCloses-Bug: #2018967\n\nChange-Id: Id328446be8591fac3affe2cf8adeaeee09697085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/865226c80b4e4e0e93f3ce5a417891a35cbfb43f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/865226c80b4e4e0e93f3ce5a417891a35cbfb43f"}]},"branch":"refs/heads/master"},"6bc7b1374ab2d1befce24a62709358506f6cff5c":{"kind":"REWORK","_number":6,"created":"2023-05-25 11:30:44.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/26/883826/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/883826/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/6"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2023-05-25 00:42:36.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2023-05-25 11:30:10.000000000","tz":480},"subject":"[fwaas]Disassociate all ports before deleting firewall group","message":"[fwaas]Disassociate all ports before deleting firewall group\n\nIf the firewall group has associated ports, the firewall group may be in\nan active state. Firewall group is active, deletion is not allowed.\n\nCloses-Bug: #2018967\n\nChange-Id: Id328446be8591fac3affe2cf8adeaeee09697085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6bc7b1374ab2d1befce24a62709358506f6cff5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6bc7b1374ab2d1befce24a62709358506f6cff5c"}]},"branch":"refs/heads/master"},"7b1f9b11e66b81eabdc8361c89f8ccfdb3448f9a":{"kind":"REWORK","_number":7,"created":"2023-05-29 08:03:31.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/26/883826/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/26/883826/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/26/883826/7"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2023-05-25 00:42:36.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2023-05-29 07:56:59.000000000","tz":480},"subject":"[fwaas]Disassociate all ports before deleting firewall group","message":"[fwaas]Disassociate all ports before deleting firewall group\n\nIf the firewall group has associated ports, the firewall group may be in\nan active state. Firewall group is active, deletion is not allowed.\nThis patch cannot solve the bug[1] at the moment. Because there are\nother bugs[2] present.\nWe plan to avoid bugs[2] through patches[3] now. To solve the bug[1].\nThere are other patches[1] here to try and solve the bug[2].\n\nPartial-Bug: #2018967\n\n[1] https://bugs.launchpad.net/neutron/+bug/2018967\n[2] https://bugs.launchpad.net/neutron/+bug/2021457\n[3] https://review.opendev.org/c/openstack/neutron-fwaas/+/884333\n[4] https://review.opendev.org/c/openstack/neutron-fwaas/+/884335\n\nChange-Id: Id328446be8591fac3affe2cf8adeaeee09697085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7b1f9b11e66b81eabdc8361c89f8ccfdb3448f9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7b1f9b11e66b81eabdc8361c89f8ccfdb3448f9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
