)]}'
{"id":"openstack%2Fneutron-tempest-plugin~883838","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I7a9a1a7a440cf0b6be28abc877b1d895f69e30c9","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2019097","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-06-01 12:14:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7a9a1a7a440cf0b6be28abc877b1d895f69e30c9","subject":"Adapt bgpvpn scenarios to recent sRBAC changes","status":"MERGED","created":"2023-05-22 13:06:31.000000000","updated":"2023-06-01 12:14:13.000000000","submitted":"2023-06-01 12:14:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"883838-bug/2019097","meta_rev_id":"796d62e057705e7b59e8559a895c791db3792c21","_number":883838,"virtual_id_number":883838,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-01 12:14:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-06-01 10:21:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-05-31 13:30:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-06-01 10:21:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2023-05-31 18:26:13.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-22 14:06:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-31 13:30:28.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-05-31 15:11:16.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-06-01 10:21:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"e025d4bddfc260fa870f7349eea90a2e0cd27448","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-22 13:06:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe0e6d60783b6ca9d2aae0b46293d59526f2b057","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-22 13:10:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8954156b3a9968ef5aceccf93100ee96f7f0b72e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 14:06:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c456b1187814487a28d8d2f219d9459\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df0a0dfc8fed43548c099485c629fb94 : SUCCESS in 4m 14s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/d4a728d4b4bd436e8b86499f3c24ac12 : SUCCESS in 54m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c27c4faa34b94b98b7eff1b4a0d87efc : SUCCESS in 3m 19s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4956546aa0664ebaab98d83aed9c59b1 : FAILURE in 37m 47s\n- neutron-tempest-plugin-bgpvpn-bagpipe-yoga https://zuul.opendev.org/t/openstack/build/cdf882600f5247f0a107df296b24df32 : FAILURE in 34m 22s\n- neutron-tempest-plugin-bgpvpn-bagpipe-zed https://zuul.opendev.org/t/openstack/build/5351338494254ba2ab856ee8aafbbfd9 : FAILURE in 35m 04s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/1d4c560f6be14fd18a270d4f105a9135 : FAILURE in 41m 49s","accounts_in_message":[],"_revision_number":2},{"id":"59e38d11ca8e443aa299fad14b0a93b585f37deb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-22 17:27:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"fcea89107dc0d807797e5d5b3cd7ebf2705d8509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 19:34:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f81625b18e8e42b99872d02bff361ffa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c356ca71374d2595d7681cd5e3f84e : SUCCESS in 3m 00s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/5511d578ca874613a3c68ef3346e86ff : SUCCESS in 43m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25b27901c43d4860b7d315383e1210c7 : SUCCESS in 3m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e7852cbf63294ad1a67aa30d7c29ef10 : FAILURE in 58m 02s\n- neutron-tempest-plugin-bgpvpn-bagpipe-yoga https://zuul.opendev.org/t/openstack/build/97b6dcc106f643a09630f4f3a22fec03 : FAILURE in 36m 00s\n- neutron-tempest-plugin-bgpvpn-bagpipe-zed https://zuul.opendev.org/t/openstack/build/e3980749a583452e9f71d3cace76a0cd : FAILURE in 45m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/0747b232128647279c0b78fbe8a199de : FAILURE in 42m 03s","accounts_in_message":[],"_revision_number":3},{"id":"4acf42713120360263e0f998972fe0e1eb6b3cbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-23 09:26:23.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b56fd67815cad399491f669878e526f6e55f621f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 10:26:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf96b5a591c145ea809fd17449c23eb0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cac0f44c5074247ba8846641ea3dea4 : SUCCESS in 4m 19s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/d07403c6acce419a93fb410e23fedc7c : SUCCESS in 48m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f4dc6ed37f64329a86f50e4e2ff6b73 : SUCCESS in 3m 16s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/898c669d0b3d43d3879c521ce1cc42e0 : FAILURE in 55m 01s\n- neutron-tempest-plugin-bgpvpn-bagpipe-yoga https://zuul.opendev.org/t/openstack/build/562090e33ab54529a5ef14c13ec2a338 : FAILURE in 30m 54s\n- neutron-tempest-plugin-bgpvpn-bagpipe-zed https://zuul.opendev.org/t/openstack/build/90b0b23cd7d245f188a64d1f27c1d150 : FAILURE in 51m 39s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/76a9c6d14ea042479164696934226da5 : FAILURE in 46m 21s","accounts_in_message":[],"_revision_number":4},{"id":"9a5adff7d329a6db72252d29bb4460f7aa22d1f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-23 11:45:15.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"0ab56c3968271b41d82fb6cf48f432488f5ed748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 12:46:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/edced22480f549cbbcba6d12c671fe17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea08b077331a493f8bc65961f7d5bf1d : SUCCESS in 3m 25s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/ae4e083a801841fc807e2ffc58ce673f : SUCCESS in 52m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24a8de8faa834f0d984cfc8514c2fe3b : SUCCESS in 3m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d9e0bcc00fb941fcb1624e2c6f85265b : FAILURE in 55m 09s\n- neutron-tempest-plugin-bgpvpn-bagpipe-yoga https://zuul.opendev.org/t/openstack/build/935dc0d27da94d4991ff87341f03676c : FAILURE in 50m 52s\n- neutron-tempest-plugin-bgpvpn-bagpipe-zed https://zuul.opendev.org/t/openstack/build/22ca3c5042d14dabb5ce7288aedbb180 : FAILURE in 52m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/3b168a731db94174a753a6040e873b2a : FAILURE in 52m 48s","accounts_in_message":[],"_revision_number":5},{"id":"f159d129986881ba129a5ebeae13d9f204cabeaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-25 15:56:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"1b9b683b76f423a906e6359fa52549f8daeb429f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 18:10:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3580c6b14d940b08bde196e829d78b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ff709d36dc54614be4118aec8fb1a20 : SUCCESS in 3m 10s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/53315db85e0a41629234ee3d9d0bc617 : SUCCESS in 46m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd6480ceed8f4e91b14172155d1215ef : SUCCESS in 3m 32s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c84870531fc742e68073c989a48a799a : FAILURE in 1h 53m 10s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe-yoga https://zuul.opendev.org/t/openstack/build/643d8869727d4a93863c313e9c9359fa : FAILURE in 1h 31m 38s\n- neutron-tempest-plugin-bgpvpn-bagpipe-zed https://zuul.opendev.org/t/openstack/build/fd94dc70b9424c38a7f8f0bd4c59957c : FAILURE in 1h 13m 05s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/fc53eb0403434d559b8bb5ca8f7c9b48 : FAILURE in 1h 24m 06s","accounts_in_message":[],"_revision_number":6},{"id":"e5f5de3f7086b7fd1d2029226782ff4a47e0702e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-26 09:42:20.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"46caf0458c54f1e49c99bcf8042f781dd3694794","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-26 09:45:35.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4310f2253a7de18f7c32967e1341f65e2ea3275c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 11:49:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cd6002d75dc4e9b8a141b3a2b44f77a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0272951687be411b9c3cb39d028bdb86 : SUCCESS in 3m 28s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/f3d97ae2a0ec43249ad2650198e52947 : SUCCESS in 55m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e56d47b480bf41bcb168e4460ceacabb : SUCCESS in 2m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/cdf5915e04284a299e5ca1758e92d2c7 : TIMED_OUT in 2h 03m 05s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe-yoga https://zuul.opendev.org/t/openstack/build/b687c78cce8643aea400cf2e24bac21b : SUCCESS in 1h 17m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe-zed https://zuul.opendev.org/t/openstack/build/b96146e1437b4c77820f65b7e6428633 : SUCCESS in 1h 41m 45s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/ab847abcf13d4871960d1c0eced898e2 : SUCCESS in 1h 31m 30s","accounts_in_message":[],"_revision_number":8},{"id":"ac124ff6b9066cb41589926ac77e97b3cbc45f01","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-26 13:05:17.000000000","message":"Patch Set 8: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a84de993231e1196aa0294ef45994b0e09fb91fe","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-31 13:30:28.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"35838ca81c48ce67f5de15e0c4ef70d3da7dbc94","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-31 15:11:16.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a9a260f99a29506af5c0b64af9175a105f13b82a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-31 18:26:13.000000000","message":"Patch Set 8: -Workflow","accounts_in_message":[],"_revision_number":8},{"id":"d97e75845ebafbfb44b93c0f9e21b6a40bf160c6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-01 10:21:52.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"1674d9212bca9524fbf80e21b82985b96c9297fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 10:22:21.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"9f4f728a732d1f1dfd342772b5374091615a649e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 12:14:12.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a14b9a8a15c4875bcef5a08aeb8322d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/830e71ab782e4487b942faa6a9b1dd93 : SUCCESS in 3m 51s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/f15b932184a246a69f134030bed1e92f : SUCCESS in 47m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01035bc6d196420d85f42c8cc666f85b : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":8},{"id":"796d62e057705e7b59e8559a895c791db3792c21","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 12:14:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"8c7c5e31f19ed70786d15091f8271ca947778c9c","revisions":{"a3ef555a46635b9df398518de1a18ab96ec5da82":{"kind":"REWORK","_number":1,"created":"2023-05-22 13:06:31.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/38/883838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/38/883838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/1"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-22 13:05:36.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-22 13:05:36.000000000","tz":120},"subject":"Adopt bgpvpn scenarios to recent sRBAC changes","message":"Adopt bgpvpn scenarios to recent sRBAC changes\n\nChange-Id: I7a9a1a7a440cf0b6be28abc877b1d895f69e30c9\nRelated-Bug: 2019097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a3ef555a46635b9df398518de1a18ab96ec5da82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a3ef555a46635b9df398518de1a18ab96ec5da82"}]},"branch":"refs/heads/master"},"04cee11d2ecf25112a2fdcf47d951c713b0118a9":{"kind":"REWORK","_number":2,"created":"2023-05-22 13:10:12.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/38/883838/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/38/883838/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/2"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-22 13:05:36.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-22 13:09:58.000000000","tz":120},"subject":"Adopt bgpvpn scenarios to recent sRBAC changes","message":"Adopt bgpvpn scenarios to recent sRBAC changes\n\nChange-Id: I7a9a1a7a440cf0b6be28abc877b1d895f69e30c9\nRelated-Bug: 2019097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/04cee11d2ecf25112a2fdcf47d951c713b0118a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/04cee11d2ecf25112a2fdcf47d951c713b0118a9"}]},"branch":"refs/heads/master"},"c1b26ca36877c4fdfbbef59b51aa7e1b676b2bb2":{"kind":"REWORK","_number":3,"created":"2023-05-22 17:27:31.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/38/883838/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/38/883838/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/3"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-22 13:05:36.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-22 17:27:19.000000000","tz":120},"subject":"Adopt bgpvpn scenarios to recent sRBAC changes","message":"Adopt bgpvpn scenarios to recent sRBAC changes\n\nChange-Id: I7a9a1a7a440cf0b6be28abc877b1d895f69e30c9\nRelated-Bug: 2019097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c1b26ca36877c4fdfbbef59b51aa7e1b676b2bb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c1b26ca36877c4fdfbbef59b51aa7e1b676b2bb2"}]},"branch":"refs/heads/master"},"ac7f1af7f16fd6dd4d982c669d2edc6a2266737f":{"kind":"REWORK","_number":4,"created":"2023-05-23 09:26:23.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/38/883838/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/38/883838/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/4"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-22 13:05:36.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-23 09:26:13.000000000","tz":120},"subject":"Adopt bgpvpn scenarios to recent sRBAC changes","message":"Adopt bgpvpn scenarios to recent sRBAC changes\n\nChange-Id: I7a9a1a7a440cf0b6be28abc877b1d895f69e30c9\nRelated-Bug: 2019097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ac7f1af7f16fd6dd4d982c669d2edc6a2266737f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ac7f1af7f16fd6dd4d982c669d2edc6a2266737f"}]},"branch":"refs/heads/master"},"e289b0c04c9b2d389f99744b4924e33038cf74f6":{"kind":"REWORK","_number":5,"created":"2023-05-23 11:45:15.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/38/883838/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/38/883838/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/5"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-22 13:05:36.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-23 11:45:04.000000000","tz":120},"subject":"Adopt bgpvpn scenarios to recent sRBAC changes","message":"Adopt bgpvpn scenarios to recent sRBAC changes\n\nChange-Id: I7a9a1a7a440cf0b6be28abc877b1d895f69e30c9\nRelated-Bug: 2019097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e289b0c04c9b2d389f99744b4924e33038cf74f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e289b0c04c9b2d389f99744b4924e33038cf74f6"}]},"branch":"refs/heads/master"},"ed03320215d5d13689f1494bbff69842aa46d92d":{"kind":"REWORK","_number":6,"created":"2023-05-25 15:56:20.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/38/883838/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/38/883838/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/6"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-22 13:05:36.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-25 15:45:27.000000000","tz":120},"subject":"Adopt bgpvpn scenarios to recent sRBAC changes","message":"Adopt bgpvpn scenarios to recent sRBAC changes\n\nChange-Id: I7a9a1a7a440cf0b6be28abc877b1d895f69e30c9\nRelated-Bug: 2019097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ed03320215d5d13689f1494bbff69842aa46d92d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ed03320215d5d13689f1494bbff69842aa46d92d"}]},"branch":"refs/heads/master"},"6c0a0785703d27790e3cc29785087f4ea5d1d17f":{"kind":"REWORK","_number":7,"created":"2023-05-26 09:42:20.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/38/883838/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/38/883838/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/7"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-22 13:05:36.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-26 07:17:14.000000000","tz":120},"subject":"Adapt bgpvpn scenarios to recent sRBAC changes","message":"Adapt bgpvpn scenarios to recent sRBAC changes\n\nChange-Id: I7a9a1a7a440cf0b6be28abc877b1d895f69e30c9\nRelated-Bug: 2019097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6c0a0785703d27790e3cc29785087f4ea5d1d17f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6c0a0785703d27790e3cc29785087f4ea5d1d17f"}]},"branch":"refs/heads/master"},"8c7c5e31f19ed70786d15091f8271ca947778c9c":{"kind":"REWORK","_number":8,"created":"2023-05-26 09:45:35.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/38/883838/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/38/883838/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/38/883838/8"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-22 13:05:36.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-26 09:45:25.000000000","tz":120},"subject":"Adapt bgpvpn scenarios to recent sRBAC changes","message":"Adapt bgpvpn scenarios to recent sRBAC changes\n\nChange-Id: I7a9a1a7a440cf0b6be28abc877b1d895f69e30c9\nRelated-Bug: 2019097\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8c7c5e31f19ed70786d15091f8271ca947778c9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8c7c5e31f19ed70786d15091f8271ca947778c9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
