)]}'
{"id":"openstack%2Fneutron-tempest-plugin~884270","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I6426a60e01f757f1e97ac698c5cc7abf2b7924d4","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2020698","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-05-25 09:47:22.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-05-25 09:47:22.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-05-25 07:41:40.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-05-25 09:47:22.000000000","reason":"Change was submitted"},"30380":{"account":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"last_update":"2023-05-25 09:47:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6426a60e01f757f1e97ac698c5cc7abf2b7924d4","subject":"Make neutron-tempest-plugin-bgpvpn-bagpipe non-voting","status":"MERGED","created":"2023-05-24 18:23:08.000000000","updated":"2023-06-08 16:53:01.000000000","submitted":"2023-05-25 09:47:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884270-bug/2020698","meta_rev_id":"ee8a02870894a8ede910f2bf1cb5cc776fe911c4","_number":884270,"virtual_id_number":884270,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-25 09:47:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2023-05-25 03:39:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2023-05-25 07:43:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2023-05-25 07:43:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-24 18:31:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-25 03:39:41.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-05-25 03:39:41.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-05-25 03:39:41.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-05-25 03:39:41.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-05-25 07:41:40.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"state":"REVIEWER"}],"messages":[{"id":"f338a91259a8da944593d408c930a4ead0c54947","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-05-24 18:23:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac9385e9fc3c05ca0450165fbb869a240c9ed3db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-05-24 18:30:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5bb63d17c3a0137c4b951aa50f988ba6c4828bbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 18:31:26.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron-tempest-plugin on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/neutron-tempest-plugin/zuul.d/project.yaml@master\", line 196, column 17","accounts_in_message":[],"_revision_number":2},{"id":"c50e7d9c5f3678d9c71136133a184ce54952ad19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-05-24 18:33:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6ddc11afe0856d77f1918de7360bbee10a347406","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 18:33:22.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron-tempest-plugin on branch master.  The error\nwas:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][4][\u0027neutron-tempest-plugin-bgpvpn-bagpipe\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - build-openstack-docs-pti\n        - neutron-tempest-plugin-jobs\n        - neutron-tempest-plugin-jobs-yoga\n        - neutron-tempest-plugin-jobs-zed\n        - neutron-tempest-plugin-jobs-2023-1\n        - check-requirements\n        - tempest-plugin-jobs\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - neutron-tempest-plugin-sfc\n          - neutron-tempest-plugin-sfc-yoga\n          - neutron-tempest-plugin-sfc-zed\n          - neutron-tempest-plugin-sfc-2023-1\n          # Unstable, https://bugs.launchpad.net/neutron/+bug/2020698\n          - neutron-tempest-plugin-bgpvpn-bagpipe:\n            voting: false\n          - neutron-tempest-plugin-bgpvpn-bagpipe-yoga\n          - neutron-tempest-plugin-bgpvpn-bagpipe-zed\n          - neutron-tempest-plugin-bgpvpn-bagpipe-2023-1\n          - neutron-tempest-plugin-dynamic-routing\n          - neutron-tempest-plugin-dynamic-routing-yoga\n          - neutron-tempest-plugin-dynamic-routing-zed\n          - neutron-tempest-plugin-dynamic-routing-2023-1\n          - neutron-tempest-plugin-fwaas\n          - neutron-tempest-plugin-fwaas-zed\n          - neutron-tempest-plugin-fwaas-2023-1\n          - neutron-tempest-plugin-vpnaas\n          - neutron-tempest-plugin-vpnaas-yoga\n          - neutron-tempest-plugin-vpnaas-zed\n          - neutron-tempest-plugin-vpnaas-2023-1\n          - neutron-tempest-plugin-tap-as-a-service\n          - neutron-tempest-plugin-tap-as-a-service-yoga\n          - neutron-tempest-plugin-tap-as-a-service-zed\n          - neutron-tempest-plugin-tap-as-a-service-2023-1\n  \n      gate:\n        jobs:\n          - neutron-tempest-plugin-sfc\n          # Unstable, https://bugs.launchpad.net/neutron/+bug/2020698\n          # - neutron-tempest-plugin-bgpvpn-bagpipe\n          - neutron-tempest-plugin-dynamic-routing\n          - neutron-tempest-plugin-fwaas\n\n  in \"openstack/neutron-tempest-plugin/zuul.d/project.yaml@master\", line 179, column 3","accounts_in_message":[],"_revision_number":3},{"id":"1ded394fd1336662a8c52b205945f4d02aa3d678","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-05-24 18:37:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"bc16699a49933aee5430ea58a21e7628a661ee68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 21:07:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a1678ad74c44334bdd18fd8650812a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cd9e9b8b88c42e5969306e663574bb6 : SUCCESS in 3m 18s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c1b5b885e13d4caabc025e412475dc1b : SUCCESS in 2h 17m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fc565dfa7c604a59966b7777f002c993 : SUCCESS in 1h 54m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fa8da93258dc48f58598e71d3a8f7dbb : SUCCESS in 2h 17m 25s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c29ff98f0a2940a2b0d35fa84405973c : SUCCESS in 1h 17m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e4f4345b76c74c2298e369219f652805 : SUCCESS in 2h 26m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0d302bc1fdeb479bb8270b7baa5e9ea5 : SUCCESS in 47m 57s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/40f3360997464ceca35449e0ced5c0b5 : SUCCESS in 59m 22s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/abb8528c87f8458980cbf2ddf9d99d73 : SUCCESS in 1h 17m 26s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/908e381b45cb420193640c5ce0359747 : SUCCESS in 1h 03m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/800f6750cabb4bfcb041c13d2f5a577f : SUCCESS in 1h 22m 11s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/cc2a98c5f2384adb980bd8c93a057565 : SUCCESS in 1h 03m 20s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/53db930f3e71491b8a471715818e7da2 : SUCCESS in 42m 48s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/420ced890c1e4bedba6d385d6ad89999 : SUCCESS in 1h 53m 20s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/316e776cfa9540658bd16c4442dee26a : SUCCESS in 1h 53m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/33242a7a399c410abf197bb02bccd7ce : SUCCESS in 1h 39m 09s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/2642059ddd4543bcb8c09b525cabe52e : SUCCESS in 1h 14m 22s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/8f03ee5472bc44a49e7628d6c4afc29d : SUCCESS in 30m 58s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/0474389f334d437aa9ce1e9f3f92fa66 : SUCCESS in 2h 12m 11s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/cc08dd8df84f44cb8d4e5dc9d2ecb168 : SUCCESS in 1h 46m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/0f6a17aec70141ccad1d768c007d5e35 : SUCCESS in 1h 51m 14s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/5dfc8c9c3048468b890acdf0015e3b11 : SUCCESS in 1h 54m 49s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/3ff8c719501a433d8b1792228f00b25c : SUCCESS in 40m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76f971a81413455eb9d7293d0647d357 : SUCCESS in 3m 22s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/88a1b58e7727431ea2eb6b0a3563b862 : SUCCESS in 1h 38m 41s\n- neutron-tempest-plugin-sfc-yoga https://zuul.opendev.org/t/openstack/build/bb77a937eda343efa41e00943a5ed124 : SUCCESS in 1h 32m 43s\n- neutron-tempest-plugin-sfc-zed https://zuul.opendev.org/t/openstack/build/8a504c0ad0584df497ac79ba0e6dca1c : SUCCESS in 1h 32m 12s\n- neutron-tempest-plugin-sfc-2023-1 https://zuul.opendev.org/t/openstack/build/57784ae6c6664d65b3002192e0fee8aa : SUCCESS in 1h 38m 34s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d9f492f07eeb49a3a27e8780fba60156 : TIMED_OUT in 2h 03m 17s (non-voting)\n- neutron-tempest-plugin-bgpvpn-bagpipe-yoga https://zuul.opendev.org/t/openstack/build/89daf5b07a7b4ce4befdf9afc55f66c6 : SUCCESS in 1h 13m 09s\n- neutron-tempest-plugin-bgpvpn-bagpipe-zed https://zuul.opendev.org/t/openstack/build/e66e539ddb214789ba304a63cae81fdb : SUCCESS in 1h 26m 51s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/856b31757f854d6cb6ac9822a29ff383 : SUCCESS in 1h 26m 23s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/766486a6c11044c8ad1076601de69fd5 : SUCCESS in 47m 35s\n- neutron-tempest-plugin-dynamic-routing-yoga https://zuul.opendev.org/t/openstack/build/ed081b7df0184a9489ef807dc2b39256 : SUCCESS in 42m 21s\n- neutron-tempest-plugin-dynamic-routing-zed https://zuul.opendev.org/t/openstack/build/806fba21229245bdbcc08bd3bb479829 : SUCCESS in 54m 22s\n- neutron-tempest-plugin-dynamic-routing-2023-1 https://zuul.opendev.org/t/openstack/build/6c47d9755c4c4af0901c8029df015810 : SUCCESS in 46m 18s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/972933b11b604da6b300efc2387d0a26 : SUCCESS in 48m 36s\n- neutron-tempest-plugin-fwaas-zed https://zuul.opendev.org/t/openstack/build/10be2ceef152421099ee27478ac1e1fa : SUCCESS in 44m 44s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/16ac0856472d4fd5bb723b994922908f : SUCCESS in 45m 34s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/99df676230f84629a79c59a5c759519e : SUCCESS in 54m 34s\n- neutron-tempest-plugin-vpnaas-yoga https://zuul.opendev.org/t/openstack/build/115de5885dcf4cbabf01ecec40c0e5aa : SUCCESS in 53m 11s\n- neutron-tempest-plugin-vpnaas-zed https://zuul.opendev.org/t/openstack/build/71a7712dd8b44a9a814691d832584c8e : SUCCESS in 43m 42s\n- neutron-tempest-plugin-vpnaas-2023-1 https://zuul.opendev.org/t/openstack/build/b56f03856b94438983ed6b22230817fd : SUCCESS in 49m 15s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/afe03eeb0d784b5293de90f04c3f1e19 : SUCCESS in 1h 23m 55s\n- neutron-tempest-plugin-tap-as-a-service-yoga https://zuul.opendev.org/t/openstack/build/b723677dfa2c4fd8a9e14485a334c083 : SUCCESS in 1h 15m 18s\n- neutron-tempest-plugin-tap-as-a-service-zed https://zuul.opendev.org/t/openstack/build/9813827c435b46be98af90de10936412 : SUCCESS in 1h 20m 33s\n- neutron-tempest-plugin-tap-as-a-service-2023-1 https://zuul.opendev.org/t/openstack/build/1fe86d2e81d2403c956b5d56bb095b88 : SUCCESS in 1h 39m 42s","accounts_in_message":[],"_revision_number":4},{"id":"0d4741a68fc90dd6977ddaf5550d6b7b6719ce3a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-25 03:39:41.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"daa66beea469b7e45180f7831cb99a2dfea0b26d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-25 07:43:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a5398d0234979a754df3d0f53cb6c4597f404b03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 07:44:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9cfe26bd3b9d9e28fe8d48559e697e585646fcdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 09:47:22.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdaae3d878884b0b82fa9ec5a5793a92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb9a876e696c4982aee90768bf4da588 : SUCCESS in 3m 36s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/35fc3eefa4a141179b623d659af093e4 : SUCCESS in 1h 58m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9262266c08714a749e41dda648a0cc7e : SUCCESS in 1h 21m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/83932ab4fefe4afba76ba5951ece6eee : SUCCESS in 1h 50m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/069b362e3db14186a6866797a77ae21e : SUCCESS in 2h 01m 49s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3bc30a696a95462591eb0e9a99b89feb : SUCCESS in 1h 17m 22s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/c1e2c16f26e54b069701a512af264fb9 : SUCCESS in 1h 11m 17s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/bbb5497f5eb24175881542be276c3018 : SUCCESS in 1h 14m 09s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/fb47a6e66ea34690ac372d81f15f7c7c : SUCCESS in 1h 31m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6508024919649b6a8022bb10ae87b46 : SUCCESS in 3m 24s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/3caff01059bf487fbc2d25dee17823c3 : SUCCESS in 1h 24m 10s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/9f72c2a3d6594986abb9a824b2747029 : SUCCESS in 52m 59s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/f03f3032ff3e40fdb8cc068da69b6c61 : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":4},{"id":"073656c07d470df15ffd006e3810c42b387a004b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 09:47:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ee8a02870894a8ede910f2bf1cb5cc776fe911c4","tag":"autogenerated:gerrit:revert","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-06-08 16:53:01.000000000","message":"Created a revert of this change as I8dbc6e812c1a6fbe57a86827b3b9755bbd2eeee8","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"318d629308287c199ff937c1b4923a09a56f2984","revisions":{"76c88a4fdb7f5617ce892142166d04e834fe541c":{"kind":"REWORK","_number":1,"created":"2023-05-24 18:23:08.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/884270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/70/884270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/884270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/884270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/884270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/884270/1"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-05-24 18:21:13.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-05-24 18:21:13.000000000","tz":-240},"subject":"Make neutron-tempest-plugin-bgpvpn-bagpipe non-voting","message":"Make neutron-tempest-plugin-bgpvpn-bagpipe non-voting\n\nThis job has been unstable for a while on the master\nbranch, until changes merge lets make it non-voting\nin the check queue.\n\nChange-Id: I6426a60e01f757f1e97ac698c5cc7abf2b7924d4\nRelated-bug: #2020698\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/76c88a4fdb7f5617ce892142166d04e834fe541c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/76c88a4fdb7f5617ce892142166d04e834fe541c"}]},"branch":"refs/heads/master"},"8ef3e3332a91d916b41b615d58fbc8b8f689ae39":{"kind":"REWORK","_number":2,"created":"2023-05-24 18:30:52.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/884270/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/70/884270/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/884270/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/884270/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/884270/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/884270/2"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-05-24 18:21:13.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-05-24 18:30:35.000000000","tz":-240},"subject":"Make neutron-tempest-plugin-bgpvpn-bagpipe non-voting","message":"Make neutron-tempest-plugin-bgpvpn-bagpipe non-voting\n\nThis job has been unstable for a while on the master\nbranch, until changes merge lets make it non-voting\nin the check queue.\n\nRelated-bug: #2020698\nChange-Id: I6426a60e01f757f1e97ac698c5cc7abf2b7924d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8ef3e3332a91d916b41b615d58fbc8b8f689ae39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8ef3e3332a91d916b41b615d58fbc8b8f689ae39"}]},"branch":"refs/heads/master"},"4f8dd3ae1651fe91e96bed06618b3cbe36ed0f53":{"kind":"REWORK","_number":3,"created":"2023-05-24 18:33:00.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/884270/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/70/884270/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/884270/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/884270/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/884270/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/884270/3"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-05-24 18:21:13.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-05-24 18:32:54.000000000","tz":-240},"subject":"Make neutron-tempest-plugin-bgpvpn-bagpipe non-voting","message":"Make neutron-tempest-plugin-bgpvpn-bagpipe non-voting\n\nThis job has been unstable for a while on the master\nbranch, until changes merge lets make it non-voting\nin the check queue.\n\nRelated-bug: #2020698\nChange-Id: I6426a60e01f757f1e97ac698c5cc7abf2b7924d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4f8dd3ae1651fe91e96bed06618b3cbe36ed0f53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4f8dd3ae1651fe91e96bed06618b3cbe36ed0f53"}]},"branch":"refs/heads/master"},"318d629308287c199ff937c1b4923a09a56f2984":{"kind":"REWORK","_number":4,"created":"2023-05-24 18:37:44.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/884270/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/70/884270/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/884270/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/884270/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/884270/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/70/884270/4"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-05-24 18:21:13.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-05-24 18:37:35.000000000","tz":-240},"subject":"Make neutron-tempest-plugin-bgpvpn-bagpipe non-voting","message":"Make neutron-tempest-plugin-bgpvpn-bagpipe non-voting\n\nThis job has been unstable for a while on the master\nbranch, until changes merge lets make it non-voting\nin the check queue.\n\nRelated-bug: #2020698\nChange-Id: I6426a60e01f757f1e97ac698c5cc7abf2b7924d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/318d629308287c199ff937c1b4923a09a56f2984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/318d629308287c199ff937c1b4923a09a56f2984"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
