)]}'
{"id":"openstack%2Fneutron-tempest-plugin~884476","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I292fd053c7568853f56f7ab39d397889edfe65a4","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/1983053","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-05-31 20:29:32.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-05-31 17:11:12.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I292fd053c7568853f56f7ab39d397889edfe65a4","subject":"Define proper common API extensions for 2023.1 and Zed jobs","status":"MERGED","created":"2023-05-26 10:10:28.000000000","updated":"2023-05-31 20:29:32.000000000","submitted":"2023-05-31 20:29:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"884476-bug/1983053","meta_rev_id":"0b30fc9f639963ec89d4d84f5a13d4d59dec91e0","_number":884476,"virtual_id_number":884476,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-31 20:29:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2023-05-31 17:16:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-05-31 17:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2023-05-31 17:16:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-26 13:01:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-26 19:46:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-05-30 13:52:17.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-05-31 17:11:12.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-05-31 17:16:06.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"203058b5ad6b151da0d5c8044458c813bea0fc83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-26 10:10:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45db55fba5364bccb2275d23f0ceb80a83f71176","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 13:01:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/639d4c2c885d473696c73383f9a89c6a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d1b9060e58c4e8181903724f05148d6 : SUCCESS in 3m 13s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/b0dd7ed38ee0419d8e27f5793c8a6f01 : FAILURE in 1h 52m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/8cb33c5e3e414cf5ab864471e5c9ac57 : SUCCESS in 2h 03m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14a3f9528dfd414ea310ff1985b26f74 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"830e105133f1a89d4d4bd5e63e298702c6293dc9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-05-26 19:46:57.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8c62295651dafe8f983a81390ba1810c5f125a08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-29 08:09:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"22216607045442b05a92226a3496b04fdf89cbab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 10:00:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c573fb1444fc40cda886b68cdb8779c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40ae82644adc4f8f8849a85b5d538c49 : SUCCESS in 3m 17s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/ccbc591b593b44a1ab6748fe6fb146c9 : FAILURE in 1h 49m 43s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/680fc76cc7ce47e1acd38b9d7aa74fd6 : SUCCESS in 1h 17m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/f11de1a497a94156a79b53bc8c08949e : SUCCESS in 1h 44m 59s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/690c67daa9a1426ebfdaa59076da0c53 : SUCCESS in 1h 21m 22s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/f987aecbf1484295b170e8f6a450b812 : SUCCESS in 31m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/319b25ed5d6f4b91ba0201e21c895024 : SUCCESS in 3m 52s\n- neutron-tempest-plugin-sfc-zed https://zuul.opendev.org/t/openstack/build/a0993742b79a400eb4185de38d7b9d91 : SUCCESS in 1h 30m 54s\n- neutron-tempest-plugin-bgpvpn-bagpipe-zed https://zuul.opendev.org/t/openstack/build/374234d61694441eb66fe62f2d17c06f : SUCCESS in 1h 28m 05s\n- neutron-tempest-plugin-dynamic-routing-zed https://zuul.opendev.org/t/openstack/build/c782c5d1798248e0b60f371494256732 : SUCCESS in 47m 11s\n- neutron-tempest-plugin-fwaas-zed https://zuul.opendev.org/t/openstack/build/a713412f6d0e47d3bf0be6a9fb63be82 : POST_FAILURE in 38m 59s\n- neutron-tempest-plugin-vpnaas-zed https://zuul.opendev.org/t/openstack/build/980e3595ed424b1283685588bf6e56d7 : SUCCESS in 38m 07s\n- neutron-tempest-plugin-tap-as-a-service-zed https://zuul.opendev.org/t/openstack/build/ff9abaf96da7441f9b4a760cfaab52a8 : SUCCESS in 1h 08m 51s","accounts_in_message":[],"_revision_number":2},{"id":"045be867bbd9ded6cc9cc00a73db73a41db24190","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-30 13:52:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"36a59177f4ece5a3def6e20caacc923bd39fb0c5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-30 13:58:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1bd46def762030ffe65079343a2abf7b06a1b6d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-31 09:34:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8f1880f8f2d49137adf3e42d3b2601694149f6c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 11:41:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdc0e4fc877548cabefc8a8465395529\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93edfa6c9d124767961c89986a099009 : SUCCESS in 3m 23s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/5da690e7bf3b4f2798e5d4d7b4f4054b : SUCCESS in 2h 05m 13s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/7ae0eb7d264d4dabb3c1b90e322e4a07 : SUCCESS in 2h 03m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/fec74264caa543ac84e6fdccb26af709 : SUCCESS in 1h 17m 47s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/0fc87096af5241869c6c52ce1bf83019 : SUCCESS in 1h 45m 33s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/4de5219cc83340a1aa01d190f44324eb : SUCCESS in 41m 37s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/af17b645aef44945b6dc80876ea4592e : SUCCESS in 1h 28m 28s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/623e4b8a78ce4a55b9d59dd680255047 : SUCCESS in 1h 17m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/93eded3cf24b49cf94d9a6849f16e58f : SUCCESS in 1h 37m 58s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/0bfebb66502f40b2bd02a4e820c86389 : SUCCESS in 41m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08eb9dbc47b44c9287cb6bf6ffa38765 : SUCCESS in 3m 36s\n- neutron-tempest-plugin-sfc-zed https://zuul.opendev.org/t/openstack/build/9e7ac0b347b1407ca86422a9495b33f3 : SUCCESS in 1h 47m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe-zed https://zuul.opendev.org/t/openstack/build/80fccfa5025a4a839e3c2e5d1469acd2 : SUCCESS in 1h 40m 30s\n- neutron-tempest-plugin-dynamic-routing-zed https://zuul.opendev.org/t/openstack/build/6ce0b38c453f45a1ad2188675738b142 : SUCCESS in 1h 08m 30s\n- neutron-tempest-plugin-fwaas-zed https://zuul.opendev.org/t/openstack/build/0cdcfb673a964eee8c4ac779db62ff36 : SUCCESS in 50m 41s\n- neutron-tempest-plugin-vpnaas-zed https://zuul.opendev.org/t/openstack/build/61ec93429235453285f101a2c009d123 : SUCCESS in 50m 08s\n- neutron-tempest-plugin-tap-as-a-service-zed https://zuul.opendev.org/t/openstack/build/c1e156e7add248d98228d98e3dbfda40 : SUCCESS in 1h 18m 29s","accounts_in_message":[],"_revision_number":3},{"id":"6cd3144e69db3dce4cd35b020bffd956963d582f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-05-31 17:11:12.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9763c0260715c1f00a6fb715f1dc11f6b6e6fe2f","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-05-31 17:16:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a9e8c7ac5072f924ad60c584b748f71d4d833f6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 17:16:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d58bf86e7b1f498a61ec3990a43576f5011f2f5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 20:29:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc856b343675427e97ec2ae697f8d179\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/443c7434ac2947deb36d0d73bdaa0f50 : SUCCESS in 4m 04s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/ff432aab49274f4eb02f88ebcf208986 : SUCCESS in 1h 19m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/929349816a67464f88b8722b404ef98a : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":3},{"id":"0b30fc9f639963ec89d4d84f5a13d4d59dec91e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 20:29:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"764544f8bb42b93e41fd88da1a49eb535235d04c","revisions":{"be0ac20398c5dca2cce589ff08b587d9b484cad6":{"kind":"REWORK","_number":1,"created":"2023-05-26 10:10:28.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/76/884476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/76/884476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/76/884476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/76/884476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/76/884476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/76/884476/1"}}},"commit":{"parents":[{"commit":"bd7d0e8402af758a5de13ebc98dcb96ef558e61a","subject":"Merge \"[S-RBAC] Update DSCP marking rule create API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/bd7d0e8402af758a5de13ebc98dcb96ef558e61a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-26 10:08:36.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-26 10:08:36.000000000","tz":120},"subject":"Define proper common API extensions for 2023.1 and Zed jobs","message":"Define proper common API extensions for 2023.1 and Zed jobs\n\nIn Linuxbridge and openvswitch_iptables-hybrid jobs for stable/2023.1\nand stable/zed branches there was missing definition of the common API\nextensions thus list was inherited from master jobs.\n\nChange-Id: I292fd053c7568853f56f7ab39d397889edfe65a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/be0ac20398c5dca2cce589ff08b587d9b484cad6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/be0ac20398c5dca2cce589ff08b587d9b484cad6"}]},"branch":"refs/heads/master"},"a429743fd7856c80b2380cbdbd7827cf2dc2d602":{"kind":"REWORK","_number":2,"created":"2023-05-29 08:09:17.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/76/884476/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/76/884476/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/76/884476/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/76/884476/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/76/884476/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/76/884476/2"}}},"commit":{"parents":[{"commit":"318d629308287c199ff937c1b4923a09a56f2984","subject":"Make neutron-tempest-plugin-bgpvpn-bagpipe non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/318d629308287c199ff937c1b4923a09a56f2984"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-26 10:08:36.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-29 08:09:05.000000000","tz":120},"subject":"Define proper common API extensions for 2023.1 and Zed jobs","message":"Define proper common API extensions for 2023.1 and Zed jobs\n\nIn Linuxbridge and openvswitch_iptables-hybrid jobs for stable/2023.1\nand stable/zed branches there was missing definition of the common API\nextensions thus list was inherited from master jobs.\n\nChange-Id: I292fd053c7568853f56f7ab39d397889edfe65a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a429743fd7856c80b2380cbdbd7827cf2dc2d602"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a429743fd7856c80b2380cbdbd7827cf2dc2d602"}]},"branch":"refs/heads/master"},"764544f8bb42b93e41fd88da1a49eb535235d04c":{"kind":"REWORK","_number":3,"created":"2023-05-31 09:34:16.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/76/884476/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/76/884476/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/76/884476/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/76/884476/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/76/884476/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/76/884476/3"}}},"commit":{"parents":[{"commit":"318d629308287c199ff937c1b4923a09a56f2984","subject":"Make neutron-tempest-plugin-bgpvpn-bagpipe non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/318d629308287c199ff937c1b4923a09a56f2984"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-26 10:08:36.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-31 09:34:05.000000000","tz":120},"subject":"Define proper common API extensions for 2023.1 and Zed jobs","message":"Define proper common API extensions for 2023.1 and Zed jobs\n\nIn Linuxbridge and openvswitch_iptables-hybrid jobs for stable/2023.1\nand stable/zed branches there was missing definition of the common API\nextensions thus list was inherited from master jobs.\n\nChange-Id: I292fd053c7568853f56f7ab39d397889edfe65a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/764544f8bb42b93e41fd88da1a49eb535235d04c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/764544f8bb42b93e41fd88da1a49eb535235d04c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
