)]}'
{"id":"openstack%2Fneutron-tempest-plugin~887159","triplet_id":"openstack%2Fneutron-tempest-plugin~master~If4233a01575dfc9ffc84295f032ed19e32c38a7b","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"remove-lb-scenario-jobs","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-06-29 11:24:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4233a01575dfc9ffc84295f032ed19e32c38a7b","subject":"Add neutron-tempest-plugin-linuxbridge job to the experimental queue","status":"MERGED","created":"2023-06-28 11:03:08.000000000","updated":"2023-06-29 11:24:33.000000000","submitted":"2023-06-29 11:24:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887159-remove-lb-scenario-jobs","meta_rev_id":"b4e256e4944c7837d0332f0eada78bac456d631f","_number":887159,"virtual_id_number":887159,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-29 11:24:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2023-06-29 07:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-06-28 15:33:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2023-06-29 07:57:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-28 13:30:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-28 15:33:40.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-06-29 07:57:52.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"70b3fa1a7d41e00bb3a735edcda9036992917d06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-06-28 11:03:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"300593c537cd7f085fb14c5dab6082060f773b5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 13:30:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eee60d454512424fa4ea159ea9ab98b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2088edc2205c45f1a62587befdc6981d : SUCCESS in 4m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/50975c391fb24c67952d725066341f2d : SUCCESS in 2h 05m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/91b1a42d698241489afb2d54ec4f0868 : SUCCESS in 1h 32m 03s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/59befcc341ee4be984ee29266fb4873e : SUCCESS in 1h 17m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6ee113c09dc04ab2947d92f415656af9 : SUCCESS in 1h 25m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9f3acb2039f24f02b0a471afaa9160be : SUCCESS in 33m 49s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/494d3d4f4b1b4fea864de171b31de7d8 : SUCCESS in 1h 03m 29s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/4f2f260681844771bf7535e06c267477 : SUCCESS in 1h 42m 59s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/e1d1f6e516154c7d9c6ece3ac7870b12 : SUCCESS in 1h 02m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/7c40276d640c4f2684bbb76cd280510d : SUCCESS in 1h 12m 24s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/4a483b2f5f724c8b94fe2b4ad4a2d8fd : SUCCESS in 1h 30m 30s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/1e0ffb653de345f9a561da5184cdd796 : SUCCESS in 44m 38s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/ed75dfe1ca8c458282d262f8eef2bdf7 : SUCCESS in 1h 29m 57s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/eb4ffbca90a949a1b70b172bf1c9a384 : SUCCESS in 1h 36m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/2dc75480a8d446a891f0d8915df522c3 : SUCCESS in 1h 21m 20s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/7e2f6d9e202a4856a1fc6ce3d8dde4f5 : SUCCESS in 2h 12m 31s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/5a351ce6f6184f7ba5ba12bd50d30e39 : SUCCESS in 39m 54s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/a44c61d4c62d49ad97e213be9012ba69 : SUCCESS in 2h 26m 00s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/c81c3cbfa7f6419ab4a2744ee86e2c71 : SUCCESS in 1h 47m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/ff15c71763c047ebb2f10067af712ae8 : SUCCESS in 2h 13m 04s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/2a26852e53e149928307976c7fb1dfd2 : SUCCESS in 1h 31m 01s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/40cbc30a8be043db8228c5014b44e9a5 : SUCCESS in 34m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe39ad81032c4d958cd677f53ab5c741 : SUCCESS in 3m 10s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/dbfff05a55044d9eb222571ff4452d23 : SUCCESS in 1h 36m 35s\n- neutron-tempest-plugin-sfc-yoga https://zuul.opendev.org/t/openstack/build/e4b3783ef0d04b388ab6e9fb8f264e7d : SUCCESS in 1h 07m 03s\n- neutron-tempest-plugin-sfc-zed https://zuul.opendev.org/t/openstack/build/8f571aeb22fc4cc1ae026ef8ee1cbb36 : SUCCESS in 1h 07m 38s\n- neutron-tempest-plugin-sfc-2023-1 https://zuul.opendev.org/t/openstack/build/898c1ce12b744cdfbf9c2474802f5263 : SUCCESS in 1h 45m 15s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/b548b5db88904dca81e278e7b287bd7e : SUCCESS in 1h 36m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe-yoga https://zuul.opendev.org/t/openstack/build/1e145db08bf54aa4999ee0026b580d79 : SUCCESS in 1h 38m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe-zed https://zuul.opendev.org/t/openstack/build/a33a1d89e81949c3b10b5bfdd81dba37 : SUCCESS in 1h 30m 44s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/f73fcac04bd24874a1d95c7c6ca3ca03 : SUCCESS in 1h 40m 54s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/9537df1192a2485789bc20d7baded965 : SUCCESS in 1h 13m 06s\n- neutron-tempest-plugin-dynamic-routing-yoga https://zuul.opendev.org/t/openstack/build/cc8ffc57dbb64dd5b0970650a89f4fb8 : SUCCESS in 58m 42s\n- neutron-tempest-plugin-dynamic-routing-zed https://zuul.opendev.org/t/openstack/build/6b7c13fa93c94ef8866d66772f68ca56 : SUCCESS in 58m 55s\n- neutron-tempest-plugin-dynamic-routing-2023-1 https://zuul.opendev.org/t/openstack/build/0fe70ad09ec343aeab0181c615535026 : SUCCESS in 1h 11m 52s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/d2b88c90be734227b12d6dc38238096f : SUCCESS in 53m 13s\n- neutron-tempest-plugin-fwaas-zed https://zuul.opendev.org/t/openstack/build/5fca73154e00435e91acda01ceac412f : SUCCESS in 51m 34s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/8c3ab783cdac4d9b9fc5a96783de4f1a : SUCCESS in 52m 39s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/771090ae12684febbdadeaf502a00dc3 : SUCCESS in 55m 02s\n- neutron-tempest-plugin-vpnaas-yoga https://zuul.opendev.org/t/openstack/build/ed500e516f5140c799424f9de2b7ed27 : SUCCESS in 52m 40s\n- neutron-tempest-plugin-vpnaas-zed https://zuul.opendev.org/t/openstack/build/495e5add47784c2da180541f146c4d36 : SUCCESS in 53m 42s\n- neutron-tempest-plugin-vpnaas-2023-1 https://zuul.opendev.org/t/openstack/build/67f64676c49e42e98b18cbb25bedc547 : SUCCESS in 1h 02m 27s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/ec683a031582461d8f69c10d66ff3620 : SUCCESS in 1h 38m 34s\n- neutron-tempest-plugin-tap-as-a-service-yoga https://zuul.opendev.org/t/openstack/build/f1dc699ac6dd48c0a73360784785933b : SUCCESS in 1h 38m 31s\n- neutron-tempest-plugin-tap-as-a-service-zed https://zuul.opendev.org/t/openstack/build/daf07f1cee6f4df2acad0ad2a606d224 : SUCCESS in 1h 32m 43s\n- neutron-tempest-plugin-tap-as-a-service-2023-1 https://zuul.opendev.org/t/openstack/build/1be28c9c99a549ec8a923b97715b4735 : SUCCESS in 1h 36m 56s","accounts_in_message":[],"_revision_number":1},{"id":"085dd5d4422969585d5348e0f862a35636a7c4ff","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-06-28 15:33:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"60c85f1a6ebfcf28e71d4360cfcaf83b98e124af","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-06-29 07:57:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"024da93e2196b40cb9d85cdd94d683411f11a891","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 07:58:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"84769a9220edb74ef515ee164ead7cede45465c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 11:24:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c534c1abdc6844cf97e93d9ca3e77d78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e3b3f1db59447dd8f817af492706f00 : SUCCESS in 3m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7d6021166e884184a8cb1cc8029f79f2 : SUCCESS in 1h 26m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2d81fead7fd44d44abe49aeda20eb0e0 : SUCCESS in 2h 13m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/67ca0e0ebb8b4b008c692cb5d5054a4e : SUCCESS in 1h 47m 07s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/047bbc03064d4f7d912ba867dfcd40ae : SUCCESS in 1h 22m 42s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/33192b95746145008f471f1c16d8d7d3 : SUCCESS in 55m 12s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/c3022267c4a143239c6804a7588b5892 : SUCCESS in 1h 38m 06s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/17441d81002b479cb73594fcb204d2da : SUCCESS in 2h 07m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1974f38cadd543a9b4eaf0a3624a4d73 : SUCCESS in 3m 23s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/28433d2abc4c4aff9668ecd7f00a3fc4 : SUCCESS in 1h 35m 06s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/e48db71b25464a4a95275789890b5013 : SUCCESS in 1h 29m 01s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/e837ad9cb3c042108119d8f914cd5805 : SUCCESS in 54m 27s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/f44b7eff764d4dce8090818404c6bc52 : SUCCESS in 43m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b4e256e4944c7837d0332f0eada78bac456d631f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 11:24:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d6c02992943706ce8611f5a530372fa0b43a54b","revisions":{"2d6c02992943706ce8611f5a530372fa0b43a54b":{"kind":"REWORK","_number":1,"created":"2023-06-28 11:03:08.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/59/887159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/59/887159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/59/887159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/59/887159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/59/887159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/59/887159/1"}}},"commit":{"parents":[{"commit":"da45177a2f02c4daf2663872154576e51621c97f","subject":"Replace deprecated tenant_id property","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/da45177a2f02c4daf2663872154576e51621c97f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-06-28 11:01:52.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-06-28 11:01:52.000000000","tz":120},"subject":"Add neutron-tempest-plugin-linuxbridge job to the experimental queue","message":"Add neutron-tempest-plugin-linuxbridge job to the experimental queue\n\nIt was agreed on the Neutron CI meeting [1] that we will remove\nthis job from check and gate queues and add it to the experimental\nqueue. It is also added to the periodic queue in Neutron with [2].\n\nLinuxbridge agent is marked as experimental since at least one cycle\nalready and this scenario job is getting less and less stable now so\nit\u0027s probably time to move it out from our gate.\n\n[1] https://meetings.opendev.org/meetings/neutron_ci/2023/neutron_ci.2023-06-27-15.00.log.html#l-127\n[2] https://review.opendev.org/c/openstack/neutron/+/887158\n\nChange-Id: If4233a01575dfc9ffc84295f032ed19e32c38a7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2d6c02992943706ce8611f5a530372fa0b43a54b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/2d6c02992943706ce8611f5a530372fa0b43a54b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
