)]}'
{"id":"openstack%2Fneutron-tempest-plugin~888907","triplet_id":"openstack%2Fneutron-tempest-plugin~master~If58db5527f245b400708f00c9389e2120ef9adaa","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2028123","hashtags":[],"change_id":"If58db5527f245b400708f00c9389e2120ef9adaa","subject":"[DNM] Test devstack fix for multiple images","status":"ABANDONED","created":"2023-07-19 07:02:16.000000000","updated":"2023-07-19 07:57:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"39289971ee8cc602fcb94f7a1bb6ab03cf499666","_number":888907,"virtual_id_number":888907,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-07-19 07:52:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-19 07:52:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78b868301d2a52119ea378bd5c2817c709435749","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-07-19 07:02:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"302957dd58b47a4bd176ca2ac286c50596b3a2f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-07-19 07:04:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"686b99de2bfcfa18ba88dc4e8a071835e9156f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 07:52:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7078da30a1848fc9d604110dc187e6a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5639a76ea9014a018b51e220ecda53f0 : SUCCESS in 3m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7523e5df14324957bd301d65a0c31725 : SUCCESS in 40m 39s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/aae5f34dd9a347b7a1205f07cb51c781 : SUCCESS in 46m 28s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/684f7568e9ce422983c2bd8b693d293d : SUCCESS in 41m 36s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/b4916a95ffec4c519dc132b380dd3d34 : SUCCESS in 41m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dd94b94d7eb4465805cd3ce833b374b : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":2},{"id":"39289971ee8cc602fcb94f7a1bb6ab03cf499666","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-07-19 07:57:33.000000000","message":"Abandoned\n\njust for test, devstack fix approved","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"59d28d804cbb49f52f6197c8a11dd738253fea9a","revisions":{"9634ff9271f36e4e93d5b5317b5d1d1d593c2520":{"kind":"REWORK","_number":1,"created":"2023-07-19 07:02:16.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/07/888907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/07/888907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/888907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/888907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/888907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/888907/1"}}},"commit":{"parents":[{"commit":"d5023e1b9914fe93d909ea92b38baba76f2d7869","subject":"Use OVN v21.06 in Wallaby and Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d5023e1b9914fe93d909ea92b38baba76f2d7869"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-07-19 07:01:38.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-07-19 07:01:41.000000000","tz":330},"subject":"[DNM] Test devstack fix for multiple images","message":"[DNM] Test devstack fix for multiple images\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/888906\nChange-Id: If58db5527f245b400708f00c9389e2120ef9adaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9634ff9271f36e4e93d5b5317b5d1d1d593c2520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9634ff9271f36e4e93d5b5317b5d1d1d593c2520"}]},"branch":"refs/heads/master"},"59d28d804cbb49f52f6197c8a11dd738253fea9a":{"kind":"REWORK","_number":2,"created":"2023-07-19 07:04:34.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/07/888907/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/07/888907/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/888907/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/888907/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/888907/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/07/888907/2"}}},"commit":{"parents":[{"commit":"d5023e1b9914fe93d909ea92b38baba76f2d7869","subject":"Use OVN v21.06 in Wallaby and Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d5023e1b9914fe93d909ea92b38baba76f2d7869"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-07-19 07:01:38.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-07-19 07:04:22.000000000","tz":330},"subject":"[DNM] Test devstack fix for multiple images","message":"[DNM] Test devstack fix for multiple images\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/888906\nChange-Id: If58db5527f245b400708f00c9389e2120ef9adaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/59d28d804cbb49f52f6197c8a11dd738253fea9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/59d28d804cbb49f52f6197c8a11dd738253fea9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
