)]}'
{"id":"openstack%2Fneutron-tempest-plugin~890210","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ice0f3573dd51851752c745c0b1d35c82bd7808fd","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"fix-statless-sg-tests-skip","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-08-02 15:00:15.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-08-03 13:32:22.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-08-03 11:33:36.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-08-03 13:32:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ice0f3573dd51851752c745c0b1d35c82bd7808fd","subject":"Skip stateless SG dualstack tests when openvswitch fw driver is used","status":"MERGED","created":"2023-08-01 13:54:36.000000000","updated":"2023-08-03 13:32:22.000000000","submitted":"2023-08-03 13:32:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"890210-fix-statless-sg-tests-skip","meta_rev_id":"e7b8bc7b8ea1c5268484a59343a4612bff57e1b4","_number":890210,"virtual_id_number":890210,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-03 13:32:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2023-08-02 15:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2023-08-03 11:33:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2023-08-03 11:33:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-01 14:04:08.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2023-08-01 16:01:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-01 19:19:11.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-08-02 07:25:51.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-08-02 15:00:15.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"a9d52532ccab9358e7d98040f9aac290a2928d32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-01 13:54:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58447f083822261149eb5c383e43c6107556760a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-08-01 14:04:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a444b1a3ce8011b9e89dbaefaa4e60e5b1e9c214","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 16:01:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43b8d79fa8e744b485c5eb5bef24df70\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52696d0f5b144a579f8967902cfc808b : SUCCESS in 3m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ce80a1426b3c4926ae2ceb151d165969 : SUCCESS in 1h 43m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1bc57cc1e343468cbf1cbecaef379217 : SUCCESS in 1h 51m 27s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/adc43f9485b84f2585268a6a2ee86d3c : SUCCESS in 1h 44m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/949eda8261b84d0ba607c8d1da97e04c : SUCCESS in 1h 43m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6aac12cc00b74952b3378804c8303897 : SUCCESS in 41m 19s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/d0973e134da24b0b95ebdcf62369f564 : SUCCESS in 1h 19m 43s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/fdaeafd6dcd1492fb7e3097e6ef0eb71 : SUCCESS in 1h 36m 32s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/96c3916d719b45909d3f4a8e34e03f1f : SUCCESS in 1h 35m 11s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/ef305dd2199c4cf782ada08ca81241b2 : SUCCESS in 1h 13m 53s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/f96e73f62467443e9a60aea6125742de : SUCCESS in 33m 46s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/b79e6a025f634586a6a58aeb1ea32ece : SUCCESS in 1h 41m 47s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/a0a52eafafee4aafa4f97ec24b4e6395 : SUCCESS in 1h 28m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/916e1611e3c14268b9f8eebfe1621711 : SUCCESS in 1h 47m 44s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/4b7e1f991b9447e58f61bc2f83a7d24b : SUCCESS in 1h 47m 09s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/cbb52b7c34c948c49402e87ac83946e7 : SUCCESS in 38m 25s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/42a78b71e0e347c2ab42645ff4fe0ba0 : SUCCESS in 2h 04m 19s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/14b9558c7cf240779cac9a074858935f : SUCCESS in 1h 33m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/fafa85a03f8c4dd4b9183c05b4ef43ee : SUCCESS in 2h 05m 10s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/c61884e05dc2401498a6ac8598d15efd : SUCCESS in 1h 37m 14s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/d28f8bd9558b4a1699715bfe1d325f89 : SUCCESS in 33m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d08e7527b8994c93a3bc782bf5237c56 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5ec345966b20cd4c9f9a279cdbec6ab4705185de","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-01 19:19:11.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"075900a3e039c8b120266ad6135911fd842f3131","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-02 07:25:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"98d929b7e68d515804eaac9b30691bc994387f1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-02 14:56:12.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9af06af6127d9d4d526d4a261d30dc4ca966be55","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-02 14:58:30.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0f42680ef3c507dbad4218fff2fff7489db54523","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-08-02 15:00:15.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fc7533f779b4cfa44109c5e9050c9f18a1f620f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 18:44:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3110e9a5337f42288c77d46d5151f3ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2620798a3034ab8852de20d6393ab58 : SUCCESS in 3m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/66bfe22a5611418797d58bec3b8af061 : SUCCESS in 1h 17m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/18bd28864d214277bce722f308c4d4c9 : SUCCESS in 1h 44m 35s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4b2082ffb8c04ea184d24ac7dcd72434 : SUCCESS in 1h 31m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3cb256365dce4d778f166ed5112ce4a4 : SUCCESS in 1h 23m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f90639d7891d4dea9066825edf0e201d : SUCCESS in 35m 52s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/9c890545544643188bebbf1838909875 : SUCCESS in 1h 17m 50s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/3bdbb5185c3847e1805b45749c866015 : SUCCESS in 1h 01m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/b09cd78ab854421d9f712b0450027b0c : SUCCESS in 1h 17m 07s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/e9291bd9be414146b32a745652ade3b3 : SUCCESS in 1h 11m 54s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/d242aac970774daea3833b1bf5bc48f8 : SUCCESS in 32m 47s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/6692c0d3b6cf42ee8d48a03c138e6c5c : SUCCESS in 1h 44m 08s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/769e69c78d6943fcb63a02c0ee96ad20 : SUCCESS in 1h 11m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/5f1c6fa60641439387025a689249834c : SUCCESS in 1h 17m 17s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/0809e22c2a984cc9af7b2cf914fbd164 : SUCCESS in 1h 15m 17s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/42fd1904795442d9adf494b00c6ad6de : SUCCESS in 29m 57s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/5dccdc25d3864656957be3fbca5f75d5 : SUCCESS in 1h 27m 15s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/bb8a95a9a602427ab687c079b2f7413b : SUCCESS in 1h 28m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/fe1c4c3f71604d75966cd3ef5b0607d8 : SUCCESS in 1h 32m 25s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/962776e0f61a4605b6cefa6aa60b24ca : SUCCESS in 1h 31m 15s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/66a2c4ee563d4d50acd6912c6a49e526 : SUCCESS in 34m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/373163ea0ca14812ba6b31cf5c655dd7 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":2},{"id":"4856f37be6461bcb26b6b28a6a03942823d599b8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-03 11:33:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9dda1af5a3f90c1915d55d200c9840beb7b1421b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 11:34:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"72e828ceeaf2c5ce959d5188607bd08ef55ebfbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 13:32:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9776497ce3e24f9f87850c6d2dff54ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f558dce1b34044ac2d6f087a9aa7be : SUCCESS in 3m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/970b848b8aac40c6adde73fa00f18221 : SUCCESS in 1h 24m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/383117a1805f4acd9456403ff18eed6b : SUCCESS in 1h 45m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c7d1a751539047e693f61aae68460f29 : SUCCESS in 1h 29m 13s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/da97fb5451b7421eaacca9aa0768a522 : SUCCESS in 1h 41m 15s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/f891f87a04b8418c853ff11ff96c3af4 : SUCCESS in 1h 31m 36s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/a2082902a145427da67f7e11f1fb3da5 : SUCCESS in 1h 57m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/545d22f2e2f7488cbd0030015a3c15fa : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"e7b8bc7b8ea1c5268484a59343a4612bff57e1b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 13:32:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c1793ca2e84aa336c7a0eaa6a63861120d93758b","revisions":{"0ac45f3c5ecf2ebbcaeafe027fa487bcf77afef9":{"kind":"REWORK","_number":1,"created":"2023-08-01 13:54:36.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/10/890210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/10/890210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/890210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/890210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/890210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/890210/1"}}},"commit":{"parents":[{"commit":"9cf35d26c8a03610af7119c2c15ed17d8514ed88","subject":"Merge \"Using \"Designate API\" in addition to DNS query client check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9cf35d26c8a03610af7119c2c15ed17d8514ed88"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-01 13:51:50.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-01 13:54:33.000000000","tz":120},"subject":"Skip stateless SG dualstack tests when openvswitch fw driver is used","message":"Skip stateless SG dualstack tests when openvswitch fw driver is used\n\nOpenvswitch firewall driver does not support stateless SG so tests from\nclasses:\nStatelessSecGroupDualStackDHCPv6StatelessTest\nStatelessSecGroupDualStackSlaacTest\n\nfrom the module neutron_tempest_plugin.scenario.test_security_groups\nshould be skipped in such case.\n\nCloses-bz: #2227138\nChange-Id: Ice0f3573dd51851752c745c0b1d35c82bd7808fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0ac45f3c5ecf2ebbcaeafe027fa487bcf77afef9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0ac45f3c5ecf2ebbcaeafe027fa487bcf77afef9"}]},"branch":"refs/heads/master"},"c1793ca2e84aa336c7a0eaa6a63861120d93758b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-08-02 14:56:12.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/10/890210/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/10/890210/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/890210/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/890210/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/890210/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/890210/2"}}},"commit":{"parents":[{"commit":"9cf35d26c8a03610af7119c2c15ed17d8514ed88","subject":"Merge \"Using \"Designate API\" in addition to DNS query client check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9cf35d26c8a03610af7119c2c15ed17d8514ed88"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-01 13:51:50.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-02 14:55:10.000000000","tz":120},"subject":"Skip stateless SG dualstack tests when openvswitch fw driver is used","message":"Skip stateless SG dualstack tests when openvswitch fw driver is used\n\nOpenvswitch firewall driver does not support stateless SG so tests from\nclasses:\nStatelessSecGroupDualStackDHCPv6StatelessTest\nStatelessSecGroupDualStackSlaacTest\n\nfrom the module neutron_tempest_plugin.scenario.test_security_groups\nshould be skipped in such case.\n\nCloses-bz: https://bugzilla.redhat.com/show_bug.cgi?id\u003d2227138\nChange-Id: Ice0f3573dd51851752c745c0b1d35c82bd7808fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c1793ca2e84aa336c7a0eaa6a63861120d93758b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c1793ca2e84aa336c7a0eaa6a63861120d93758b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
