)]}'
{"id":"openstack%2Fneutron-tempest-plugin~892375","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Iacea90f6c571e25efd8767defe41be804a4a23c2","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"ndr_tox_role","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-08-23 17:04:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iacea90f6c571e25efd8767defe41be804a4a23c2","subject":"[DNM] Add \"tox\" role to n-d-r CI job","status":"ABANDONED","created":"2023-08-22 16:30:17.000000000","updated":"2023-08-23 17:04:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fbfdaa0532de2d492d98881764b45a27a3be2a92","_number":892375,"virtual_id_number":892375,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-22 19:42:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-22 19:42:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b5479df94410af8874c30d615c2f6dd75ca5cf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-22 16:30:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a4d4c5eb02fe42ce9ed5563457d0863204ad8a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-22 17:25:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"35f1d821e6b037bd7853cbbff299f2deb96cd565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 19:42:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf9df2ccf3014ecebb5f9f03743cd6f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be273db3a48d4897ba5108cebd789419 : SUCCESS in 3m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8e77130ac94a4700a49c0fd8169ffb2d : SUCCESS in 1h 28m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a7d5f9b22125416b8e32b86a03744dfd : SUCCESS in 1h 45m 51s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cc06197abe13438f899fda07c29b7291 : SUCCESS in 1h 26m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e2119758feff44dab55dd86afc716d8c : SUCCESS in 1h 55m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/da708c0ee5794deb99685a261fd4040d : SUCCESS in 41m 43s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/ac22e7b87b8d49f78a2a5de744de4f6e : SUCCESS in 51m 01s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/dc8f912fcc13461c913739d1fa564620 : SUCCESS in 1h 41m 59s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/aa59b7480e9c41a59f3056d12ba49ee2 : SUCCESS in 1h 07m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/baeb05db4d2f436e804ba2e09b25d328 : SUCCESS in 1h 23m 59s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/6e09eec49dc74b628e696d842f447aae : SUCCESS in 1h 26m 47s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/12eac1a3deae47d6b82da3fbb8a64acc : SUCCESS in 38m 16s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/fb6d97c006914d27b6b7603d7ea9ce8e : SUCCESS in 1h 59m 44s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/28d367f90eca48a18834a440d0f502ae : SUCCESS in 1h 38m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/1d8d0c4bb8ae462487b58d38b86185c6 : SUCCESS in 1h 47m 05s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/94e6853f52f74de3a03606dce53763c8 : SUCCESS in 1h 59m 41s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/3f2a8716e1fe414aa41d2ca130b928a5 : SUCCESS in 31m 54s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/000382e134724a2196df05099257c72f : SUCCESS in 1h 49m 48s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/547819f72a00481fa7fec373ac82f323 : SUCCESS in 1h 46m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/3bb780864cbe4a738f210f0d36acb76b : SUCCESS in 1h 38m 32s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/53f3e152a9fe4d079fe889f32ed21bfd : SUCCESS in 2h 12m 12s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/a09d83b0736842d4a5e0fc309afd7902 : SUCCESS in 31m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f84a87d615204d5daa0ccadb3d901b1c : SUCCESS in 3m 06s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/0c7700776818417a9876e476add5c965 : SUCCESS in 1h 47m 33s\n- neutron-tempest-plugin-sfc-yoga https://zuul.opendev.org/t/openstack/build/725c6184082d4409b25bbec09b28d148 : SUCCESS in 2h 00m 11s\n- neutron-tempest-plugin-sfc-zed https://zuul.opendev.org/t/openstack/build/56d3aa18f723415cab5771b4f7e1541f : SUCCESS in 1h 47m 53s\n- neutron-tempest-plugin-sfc-2023-1 https://zuul.opendev.org/t/openstack/build/e7c12783a5864d48835c225d59b008f3 : SUCCESS in 1h 46m 12s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/eaaa5c6070c74424837ed4a0b8d7d6b0 : SUCCESS in 1h 33m 22s\n- neutron-tempest-plugin-bgpvpn-bagpipe-yoga https://zuul.opendev.org/t/openstack/build/6e1c139eb0564367a9c34c92c5ed1706 : SUCCESS in 1h 35m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe-zed https://zuul.opendev.org/t/openstack/build/a79ebd9ba8a342698bce1640f17c2139 : SUCCESS in 1h 30m 53s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/2db83c26ce9c49458923dd3ec9d47208 : SUCCESS in 1h 27m 00s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/0cc1b20e1a234b8c85cc27dcdc8951ee : RETRY_LIMIT in 18m 37s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-yoga https://zuul.opendev.org/t/openstack/build/a310ab74271643cbacad1ab1613e9aca : RETRY_LIMIT in 18m 11s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-zed https://zuul.opendev.org/t/openstack/build/04932c6638004d7c8c9de510456e93b6 : RETRY_LIMIT in 5m 46s (non-voting)\n- neutron-tempest-plugin-dynamic-routing-2023-1 https://zuul.opendev.org/t/openstack/build/f62077fb9340434cb4738f2ad0138d53 : RETRY_LIMIT in 6m 19s (non-voting)\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/dcf72e0acffa4172b1122adabe8e03f1 : SUCCESS in 53m 24s\n- neutron-tempest-plugin-fwaas-zed https://zuul.opendev.org/t/openstack/build/5e74564fe7704f1bbdfbb225858aa0d0 : FAILURE in 32m 50s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/cbb4c63fcd464283aee7a28983f5203b : SUCCESS in 55m 48s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/d589cad9000e452aa93f962db0417f77 : SUCCESS in 1h 02m 27s\n- neutron-tempest-plugin-vpnaas-yoga https://zuul.opendev.org/t/openstack/build/01941c95342a4aceb5a756db1f0a93e0 : SUCCESS in 53m 58s\n- neutron-tempest-plugin-vpnaas-zed https://zuul.opendev.org/t/openstack/build/404dfbd38b9c46df858de9e03a97b2e2 : SUCCESS in 48m 40s\n- neutron-tempest-plugin-vpnaas-2023-1 https://zuul.opendev.org/t/openstack/build/1b159a01cc6c4b8cb2210b68a58c1b18 : SUCCESS in 57m 38s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/6ea63ac3b0b94818a1fb6d75838456ce : SUCCESS in 1h 29m 04s\n- neutron-tempest-plugin-tap-as-a-service-yoga https://zuul.opendev.org/t/openstack/build/8b30775ff43b48de8745cf135592ba55 : SUCCESS in 1h 26m 34s\n- neutron-tempest-plugin-tap-as-a-service-zed https://zuul.opendev.org/t/openstack/build/4de263aa64b14c4399de65ffdb160bdb : SUCCESS in 1h 29m 07s\n- neutron-tempest-plugin-tap-as-a-service-2023-1 https://zuul.opendev.org/t/openstack/build/dcaa0bbe5d644cd8b344f2cc871ce4dc : SUCCESS in 1h 25m 26s","accounts_in_message":[],"_revision_number":2},{"id":"fbfdaa0532de2d492d98881764b45a27a3be2a92","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-23 17:04:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ea3b139b17bfa79866b323e9a708c22e631479cf","revisions":{"74388457aaa121c0619f1dc242edcb982a144cf1":{"kind":"REWORK","_number":1,"created":"2023-08-22 16:30:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/75/892375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/75/892375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/892375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/892375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/892375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/892375/1"}}},"commit":{"parents":[{"commit":"010ed337eb52ed33975ecb23e5c6595c88bab3a6","subject":"Merge \"Update Cirros image to 0.6.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/010ed337eb52ed33975ecb23e5c6595c88bab3a6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-22 00:49:52.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-22 00:49:52.000000000","tz":0},"subject":"[DNM] Add \"tox\" role to n-d-r CI job","message":"[DNM] Add \"tox\" role to n-d-r CI job\n\nChange-Id: Iacea90f6c571e25efd8767defe41be804a4a23c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/74388457aaa121c0619f1dc242edcb982a144cf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/74388457aaa121c0619f1dc242edcb982a144cf1"}]},"branch":"refs/heads/master"},"ea3b139b17bfa79866b323e9a708c22e631479cf":{"kind":"REWORK","_number":2,"created":"2023-08-22 17:25:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/75/892375/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/75/892375/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/892375/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/892375/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/892375/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/75/892375/2"}}},"commit":{"parents":[{"commit":"010ed337eb52ed33975ecb23e5c6595c88bab3a6","subject":"Merge \"Update Cirros image to 0.6.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/010ed337eb52ed33975ecb23e5c6595c88bab3a6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-22 00:49:52.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-22 01:45:33.000000000","tz":0},"subject":"[DNM] Add \"tox\" role to n-d-r CI job","message":"[DNM] Add \"tox\" role to n-d-r CI job\n\nChange-Id: Iacea90f6c571e25efd8767defe41be804a4a23c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ea3b139b17bfa79866b323e9a708c22e631479cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ea3b139b17bfa79866b323e9a708c22e631479cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
