)]}'
{"id":"openstack%2Fneutron-tempest-plugin~896122","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ie7d37def794a3cbdb2444ebbeba237fd58ddfb8a","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"neutron_t_plugin_remove_dup_tests","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-11-22 13:46:40.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-11-22 13:46:40.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-11-22 10:13:46.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"35229":{"account":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"last_update":"2023-11-22 13:46:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie7d37def794a3cbdb2444ebbeba237fd58ddfb8a","subject":"Remove test duplication between tempest and n-t-p RoutersDVRTest","status":"MERGED","created":"2023-09-21 14:05:28.000000000","updated":"2023-11-22 13:46:40.000000000","submitted":"2023-11-22 13:46:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896122-neutron_t_plugin_remove_dup_tests","meta_rev_id":"a53fd4a3377abb3aecc7633e58ed4e9a82e9570c","_number":896122,"virtual_id_number":896122,"owner":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-22 13:46:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"recommended":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2023-11-08 13:25:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2023-11-22 10:13:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-11-21 21:07:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2023-11-22 10:13:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-21 16:11:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-25 11:53:38.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-10-04 08:30:06.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"CC"},{"updated":"2023-10-04 08:30:06.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-10-04 08:30:06.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-11-08 13:25:34.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2023-11-21 21:07:05.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"9e638daf4a693e3504b3878f6696af788479c833","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-09-21 14:05:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8403a2d3c8589e2c465e531bcf36a8e9279b45ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 16:11:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e1dd00864a3481a853585849d3e0c23\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e0ddc65a5314597ad2db684295b24e9 : SUCCESS in 3m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/79357759983446ef92b576dfb512ab06 : SUCCESS in 1h 32m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c6000a3480f54d85a9353648ff62d8ed : SUCCESS in 1h 40m 44s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/657c94017f8844968df40705cce171ad : SUCCESS in 1h 41m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f67214140409427088e5a67a5952985a : SUCCESS in 1h 38m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bfbebd5deb4b47108aadc6226b753425 : SUCCESS in 45m 23s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/17ad0b6bf0f24fb5bb3c4a0203984536 : SUCCESS in 1h 06m 04s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/b2c9b103d391403dbcc92bcb30194ffe : SUCCESS in 1h 32m 21s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/51cc3fa44d61491d8cea8adfd4cb854b : SUCCESS in 1h 12m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/fc0100645bab4beb878d67f251137863 : SUCCESS in 1h 21m 44s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/f29b93864c4c490aa119a811995d8d6e : SUCCESS in 1h 18m 47s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/37390d26ed5d456d91b70c69affd266b : SUCCESS in 40m 20s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/ae06a924c9be4fd68ffe09ed6f2af72a : SUCCESS in 1h 50m 10s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/dcda66e62cd1423faafdd2b9f4554dd3 : SUCCESS in 1h 37m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/c07753f32a184ac686544a9d8d3df637 : SUCCESS in 1h 35m 17s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/9bdeab774b054a8e8c1fd9d612140c9c : SUCCESS in 1h 47m 43s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/388036aedff948899e01bc943e9dd8f2 : SUCCESS in 37m 23s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/55c8fb583d354327bd4b663a27ffe4fb : SUCCESS in 1h 54m 02s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/c774f246cff541bcad3196cf66826cdd : SUCCESS in 1h 51m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/071c3d984e1544d09f988ce04ab9c2fc : SUCCESS in 1h 57m 51s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/0d627def5c7b4be48d2857b005a6eaa7 : SUCCESS in 2h 04m 01s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/62e0091f2643485fbe7d985d42886fb0 : SUCCESS in 38m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9dc02841867486db27e5dd5f2b3754a : SUCCESS in 3m 05s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/627f280dd25749cb98aa9a9d81faee8f : SUCCESS in 1h 20m 33s\n- neutron-tempest-plugin-sfc-yoga https://zuul.opendev.org/t/openstack/build/72cdcf1e3e754ba09361c46c2ace54e6 : SUCCESS in 1h 54m 44s\n- neutron-tempest-plugin-sfc-zed https://zuul.opendev.org/t/openstack/build/612945896d594202bc3ca1ce807a845e : SUCCESS in 1h 40m 27s\n- neutron-tempest-plugin-sfc-2023-1 https://zuul.opendev.org/t/openstack/build/d944ad6b8007401485dd4c42f8234a7b : SUCCESS in 1h 23m 06s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/2a6a300f0ade4f5780baf3a681a69ac1 : SUCCESS in 1h 37m 16s\n- neutron-tempest-plugin-bgpvpn-bagpipe-yoga https://zuul.opendev.org/t/openstack/build/d4819700a2dc4afea2a1a804a96f2cd0 : SUCCESS in 1h 29m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe-zed https://zuul.opendev.org/t/openstack/build/cd8bab09e65840a38af062e79782c7e3 : SUCCESS in 1h 31m 10s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/88d9e86128494b438f6249587863e496 : SUCCESS in 1h 33m 35s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/aae2c5d8ab7044a7afdc386d3f9ee718 : SUCCESS in 1h 09m 48s\n- neutron-tempest-plugin-dynamic-routing-yoga https://zuul.opendev.org/t/openstack/build/61780fdf63284bdda5ce94ae1179e604 : SUCCESS in 43m 14s\n- neutron-tempest-plugin-dynamic-routing-zed https://zuul.opendev.org/t/openstack/build/dfe1b85f5e94453d8d97b16ca131271b : SUCCESS in 54m 45s\n- neutron-tempest-plugin-dynamic-routing-2023-1 https://zuul.opendev.org/t/openstack/build/eccb73ae54be4a5094079e23b9583d07 : SUCCESS in 1h 09m 51s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/bd13d947fb084c2093e4f2897e59516c : SUCCESS in 52m 34s\n- neutron-tempest-plugin-fwaas-zed https://zuul.opendev.org/t/openstack/build/488e374da8144e9b9c8eb34e756ca94c : SUCCESS in 47m 49s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/4fbb445a2497418985098f842b3e98fe : SUCCESS in 36m 38s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/d10776932a26483495be1f6a8710fadc : SUCCESS in 55m 33s\n- neutron-tempest-plugin-vpnaas-yoga https://zuul.opendev.org/t/openstack/build/269a2b3795384f9baf2a48005dde17a8 : SUCCESS in 51m 07s\n- neutron-tempest-plugin-vpnaas-zed https://zuul.opendev.org/t/openstack/build/4ed6f73822c9443f8c0ef9f5217c2a90 : SUCCESS in 51m 22s\n- neutron-tempest-plugin-vpnaas-2023-1 https://zuul.opendev.org/t/openstack/build/e75264709b9f4104afdac0b7831ce590 : SUCCESS in 45m 19s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/ad578b726447424fb9e0347fc1f27c67 : SUCCESS in 1h 18m 31s\n- neutron-tempest-plugin-tap-as-a-service-yoga https://zuul.opendev.org/t/openstack/build/b2df422ad5fd451dbdc6b56301a742bd : SUCCESS in 1h 34m 54s\n- neutron-tempest-plugin-tap-as-a-service-zed https://zuul.opendev.org/t/openstack/build/13ddf5ce8aee47f3b7fd080e892ed077 : SUCCESS in 1h 27m 07s\n- neutron-tempest-plugin-tap-as-a-service-2023-1 https://zuul.opendev.org/t/openstack/build/ec4adb3538bf46b39d96599779ee6f6a : SUCCESS in 1h 13m 29s","accounts_in_message":[],"_revision_number":1},{"id":"490878a67955262b32ea413eaa9e39e11a5fe46b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-25 11:53:38.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"027c8a4273afc35562288cd9d1c86eaa47c3aa88","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-09-25 12:31:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"902355dd1c3eb45874b2dac80207672d45615d1e","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-11-08 13:25:34.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bfba9acd23385b1cbef8ed9bd78dcc07ea220077","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-21 21:07:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df1a126f54cf97fd3cca9570ffe99dd5b1c6f25e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-22 10:13:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14eb06518525ee60d84b909abfc2ecff6751084a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 10:14:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"84fd9b5f870da1fdb3941645504b2bcb2f7c2a00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 13:46:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83dde424e78549608e9eedc3f223a90f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02a7112432c045b490c1d623eb2b8248 : SUCCESS in 3m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/79bd27fb70ef4820b192b1fb6df768b0 : SUCCESS in 1h 31m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a65eeff7d468492d811abd3503d1d77e : SUCCESS in 1h 44m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7b91dd52789e490b8d0688a6314db370 : SUCCESS in 1h 44m 40s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c692b1429ee1462ebc8007ff15cc81c8 : SUCCESS in 1h 43m 55s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/83c11105e2654b15920475c071dad28e : SUCCESS in 49m 44s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/b403f6659db14150b8217ea7475fe8c3 : SUCCESS in 2h 00m 57s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/4536c3f36c2345b78e52e3925fed46b4 : SUCCESS in 1h 40m 21s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/888221ef929247fcb7c586dbe7b40604 : SUCCESS in 1h 37m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f68dbff949b4ed9ac8b99c4ab635d05 : SUCCESS in 3m 26s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/5d267b73d8cd4a5eb781ba2a1edd0009 : SUCCESS in 1h 29m 46s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/083af7c7a11c402da0bd2dd3b193730d : SUCCESS in 1h 43m 37s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/34f7b7508827406cbb3820754fa2c548 : SUCCESS in 1h 01m 54s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/b062367c63d24108943d165b1125b43f : SUCCESS in 36m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a53fd4a3377abb3aecc7633e58ed4e9a82e9570c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 13:46:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d8ed3fb42ad90558aa2c478c9bcc96c4d70439ce","revisions":{"d8ed3fb42ad90558aa2c478c9bcc96c4d70439ce":{"kind":"REWORK","_number":1,"created":"2023-09-21 14:05:28.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/22/896122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/22/896122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/22/896122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/22/896122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/22/896122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/22/896122/1"}}},"commit":{"parents":[{"commit":"370f71e9877fad13883413225bf25051c0c9c391","subject":"Mark \"test_subport_connectivity_soft_reboot\" as unstable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/370f71e9877fad13883413225bf25051c0c9c391"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-09-21 14:02:06.000000000","tz":120},"committer":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-09-21 14:05:27.000000000","tz":120},"subject":"Remove test duplication between tempest and n-t-p RoutersDVRTest","message":"Remove test duplication between tempest and n-t-p RoutersDVRTest\n\nRoutersDVRTest -https://opendev.org/openstack/tempest/src/branch\\\n/master/tempest/api/network/admin/test_routers_dvr.py has same\nfunctionality in tempest and neutron-tempest-plugin, so with\nremoving these tests from tempest we can get rid of whole module,\nreducing the test duplications.\n\nChange-Id: Ie7d37def794a3cbdb2444ebbeba237fd58ddfb8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d8ed3fb42ad90558aa2c478c9bcc96c4d70439ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d8ed3fb42ad90558aa2c478c9bcc96c4d70439ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
