)]}'
{"id":"openstack%2Fneutron-tempest-plugin~896188","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I66dcb56ab5ba9cfdefbeb25f1d01871ea0d0d362","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2037034","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-10-04 15:26:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66dcb56ab5ba9cfdefbeb25f1d01871ea0d0d362","subject":"Use RE2 compatible regex for branches","status":"MERGED","created":"2023-09-22 06:15:19.000000000","updated":"2023-10-04 15:26:00.000000000","submitted":"2023-10-04 15:26:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896188-bug/2037034","meta_rev_id":"d2966aca0634f6b864f136470b785946773d5646","_number":896188,"virtual_id_number":896188,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-04 15:26:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-10-04 12:56:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-10-04 04:48:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-10-04 04:48:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-04 04:48:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-10-04 12:56:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-22 06:16:31.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-09-22 08:17:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-22 08:39:12.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-09-22 09:44:41.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-09-22 14:11:01.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-10-04 12:56:21.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"045ead1fbfa4992f25620855edf8822379dc4fa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-09-22 06:15:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0c703944432bb2e9367e39d553c819df0b3a9bc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-09-22 06:16:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d86a8bf694d135636e6195c24d24523a502b2770","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 08:17:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74b440be0064441faf020fe24576baeb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d0752de2889407fa8c24bfa769eb16f : SUCCESS in 3m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c08eaceed6e8455b8f1d5eb651b2ced6 : SUCCESS in 1h 32m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b86acfaaf51744f5a8ad11aace7c3832 : SUCCESS in 1h 29m 36s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/579ba0ff0ec243d4905d601f087d09fd : SUCCESS in 1h 19m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d942358b03714e84a5119d465b289352 : SUCCESS in 1h 56m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9cb7539fac9248cd848e8a59dfb2f726 : SUCCESS in 39m 48s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/a9d395f64c2945b98a1cd60ebb2661c1 : SUCCESS in 1h 54m 24s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/57888329b04848d5977e83f7ddf0493f : SUCCESS in 1h 10m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/43cdf29890e744bf9c4b813be669002a : SUCCESS in 1h 23m 50s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/2f2a418f709e44c38f7c2485da3484a2 : SUCCESS in 1h 37m 56s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/c82c8ff36496498f8e0c3674006b1682 : SUCCESS in 37m 34s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/dca57bda51eb4b9dbd97cbcdd67f302c : SUCCESS in 1h 45m 03s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/62de5dbe78314d1ca9e135afa0cce08c : SUCCESS in 1h 38m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/6d08ace551e14fcebf18d2a2cc700ada : SUCCESS in 1h 32m 20s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/ed76679094884c7f82207eef0e4c3ba2 : FAILURE in 1h 32m 02s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/369de7daca6743f1bcf5d712bfdf21b8 : SUCCESS in 30m 05s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/079db43964a14934a82432a7feb78e66 : SUCCESS in 1h 56m 35s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/b6c0090e43bd4d2db7d065743017afe2 : SUCCESS in 1h 15m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/79d107e30068403fb3ef03a0c9523a91 : SUCCESS in 1h 59m 26s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/fd6a4e2653ff439ba6621740df09af75 : SUCCESS in 1h 30m 29s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/566fb7da41c8474d8efe97b9bab881c6 : SUCCESS in 34m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4af7f667da6745bea4ead88705d8cbb7 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3a93c252199f280f987c64282062ecb6150e5a49","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-22 08:39:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41bf4acf7488a6d928c91211767571311e5a8342","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-09-22 09:44:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6b292ffc9d502c173e0eb730cdd42d9a61f09f5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 10:34:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8c12f2c0ea748d49684a5174860d0a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c5fb2d476dc4da19488391b9b193b8a : SUCCESS in 4m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3333b8814bcb40f38499e7987ba44055 : SUCCESS in 1h 51m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/183be0b12b864ffa933f5401be04f5c9 : SUCCESS in 1h 47m 08s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d2d210152c2e49c995c795d9c4b6e2a1 : SUCCESS in 1h 35m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/be8fb2177d5f490f9d2c96e85bc455ca : SUCCESS in 1h 43m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2524271cdb73419fa1e61fa6624f24c9 : FAILURE in 21m 50s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/c00283dfa56948bb9bb4c454674f0b0f : FAILURE in 1h 43m 22s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/77b24b99c1044702b89f5bcc2508cca3 : SUCCESS in 1h 25m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/13d9866f5f844347a41101fb62fbb4d8 : SUCCESS in 1h 31m 57s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/520630828cc740a2b17639f99fdf3255 : SUCCESS in 1h 23m 36s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/b55d1cbebdf4449087b39fe5aa76aea6 : SUCCESS in 38m 48s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/27163d394561405e92e1011d9ceca433 : SUCCESS in 1h 21m 47s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/6ff77d37674e4b28a210f7a0fbdad11e : SUCCESS in 1h 22m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/77f273efa58945fc99f27bfb17ed8e18 : SUCCESS in 1h 39m 00s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/6f9ce97c728b40068de5260a43df011a : SUCCESS in 1h 53m 19s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/d4244c049e864012bc3cb57d3c3b228f : SUCCESS in 28m 40s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/8c9febd2db6a41a2a63c3d83f43b5de8 : SUCCESS in 1h 31m 12s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/73d0fe8f4e4b4f54bec60b8deadc6d16 : SUCCESS in 1h 39m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/4e6ba1fed57141aaa3692a4777115316 : SUCCESS in 1h 43m 34s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/bb82eda3a3b544d7bef280616037447e : SUCCESS in 1h 29m 11s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/5ed6ece094e74f70b89e4a7a4d43b290 : SUCCESS in 34m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9799e4c7a9cb40cf8fa3002d98e97bbd : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"1ae0dc907448be101ddbf8573613c0afe2d0e3a5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-22 14:11:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84f3f0b915b408eaa1a1e69b23db92dfe5231f37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 16:15:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb47385d2ae146c5885335eb183af1fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b3796ea3fb746b38b97d488f24508c5 : SUCCESS in 3m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b5c45b466a304687837f64f334c3e707 : SUCCESS in 1h 20m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/28eddd54ca7f4c49aed57d5fdfb302b5 : SUCCESS in 1h 36m 12s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/638f96f4d887408180024c45c979dd34 : SUCCESS in 1h 31m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0dc60802dfbd4eff9e3d92d04d3eb5c2 : FAILURE in 2h 01m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/01ef964c47e64274877dddf2f351e4e8 : SUCCESS in 43m 54s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/567c358290104c19b2ac5636d228b5f6 : SUCCESS in 1h 17m 39s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/61e54d68b2bc481989487cab7d4fcd08 : FAILURE in 1h 16m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/d6d3e21116664504a39155adde960397 : SUCCESS in 1h 22m 57s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/f254835e1cf24af6bc3d6c10b36b329b : SUCCESS in 1h 10m 51s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/6e16f46ddbe3495982a0ec3fe812c5d0 : SUCCESS in 33m 02s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/3242923e7e934f6483d2683a3d46df01 : SUCCESS in 1h 24m 53s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/9efe21dbc34844e18af798622299829c : SUCCESS in 1h 20m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/ca0fb7071375437b9ef2d8e0a1a510ea : SUCCESS in 1h 28m 59s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/60a35dcb5ce04be5ae0cfd46749481ec : SUCCESS in 1h 30m 16s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/db36582e7923444a90e6bc52ab9c8b6f : SUCCESS in 37m 08s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/d6ab11e596ff4113b17cdd6616457d25 : SUCCESS in 1h 54m 04s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/cbbc36379eab479abe9392287c0cb538 : SUCCESS in 1h 15m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/6daff2737d5f4d7eb9ce9bc4dcf709e9 : SUCCESS in 1h 22m 29s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/875175c81ff1490d94a02621222b66e7 : SUCCESS in 1h 20m 39s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/6ae715f11c4345d2b23a07bf55b18f98 : SUCCESS in 29m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f601955990244ddd892591fdb9147385 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0dea72fd7759d18d06794c1452cfb1ac8eebb3d8","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-25 11:54:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e10825411383468fae3ef2bf62d195f1f8f02f40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 14:46:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cabc277a2df743ddb34c5fe7eda34154\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f67839967094421b95d8f1f6c0c559d : SUCCESS in 4m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/880b87973bbb47fcb048d2f76c3cf425 : FAILURE in 1h 33m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4b00e0481b7a4b35991b489432a921a3 : TIMED_OUT in 2h 49m 15s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/735344167b2e4dca989269a479fa9d78 : FAILURE in 1h 23m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2c56978dab68499ba4d74c5f2483246a : FAILURE in 1h 58m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/73347c2269954311abe10f6a25f29f40 : SUCCESS in 39m 15s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/a5bb5836339645469cc6f02234ab4970 : SUCCESS in 1h 20m 51s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/2dcf87d4e81e4b5f863e1c35cb5a13c1 : SUCCESS in 1h 01m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/fd5d5688f5354a2197983b6ad17b51a1 : SUCCESS in 1h 21m 36s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/b00b2a0b11594b9493db62768119fa7c : SUCCESS in 1h 35m 52s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/73da27e9905b4c5cbd2a9583431e36e2 : SUCCESS in 33m 03s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/b793d8486aa048029f8c6dfee4111a11 : SUCCESS in 1h 26m 54s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/478e04293cd546308d299038279fa59d : SUCCESS in 1h 06m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/3c8f07d0b500408598c79be9dd8acaec : SUCCESS in 1h 25m 22s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/a7c3a1b19fc84bfeac26f8f46b5e2a45 : FAILURE in 1h 45m 05s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/8a0e063c1b0c40d0b4d976994a9caffd : SUCCESS in 32m 27s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/ca8808f58ca04ea899a2b781af0a0f3c : SUCCESS in 1h 48m 57s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/97d61df69ba94d1580cae96682f482b1 : SUCCESS in 1h 29m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/9813342b3fe0454f95b00454fc5480a4 : SUCCESS in 1h 30m 33s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/7632f88214474ebea8e6172cf836b70c : SUCCESS in 1h 48m 41s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/7702fdbd937b496a82e337468be70272 : SUCCESS in 38m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d282c2bf9a7e47118072001e0a9fb139 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3dcc80aca277a1b2eac18fb1c130a1d22d286d69","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-26 13:44:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd00234c61b2f2fab19b7c55927cdde25bced956","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 16:36:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e5cd97cc7374a6a8aa7fe6e6437e85e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7e0aaaf62f64e51a07c0ac457cefb66 : SUCCESS in 3m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/93c7124456164080ada87c0258b7a515 : FAILURE in 1h 23m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5be06f6335084c1c810dff00a76731a0 : TIMED_OUT in 2h 50m 33s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bcef6d6dbaf0462a92f3186b25839786 : SUCCESS in 1h 28m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b22be4f3bf4d4966aaef13e937cb969a : SUCCESS in 1h 55m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bbaf8a6bd02b48d8a6a3a5568c053c3c : SUCCESS in 35m 11s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/b8e1e54022dc4904984bbe355bedf6e2 : SUCCESS in 1h 18m 56s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/721107149d674ce5add4e711f8a876b7 : SUCCESS in 1h 20m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/c1382363abaa459d8777420b8846aaa4 : SUCCESS in 1h 21m 44s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/31f00daf1a3644c19b842bcfdd70381f : SUCCESS in 1h 41m 24s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/c2ebee9b6d06479c9c2e3dfc0514d737 : SUCCESS in 29m 38s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/4055782f860f43f8b3b1d94b7b802807 : SUCCESS in 1h 51m 29s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/daa0a30a40924aaf8f6fad9326560f4e : SUCCESS in 1h 23m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/4cc7a3d512b14e24b003cac2e5538d82 : SUCCESS in 1h 47m 40s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/8d0d8bb74beb4f73b6286652cf25819f : SUCCESS in 1h 29m 52s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/e5434cfe445b4e66ad7fa7bc813c05c3 : SUCCESS in 36m 31s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/83445a4f37754cef82cd10e048333a1a : SUCCESS in 2h 21m 51s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/61451b4756c54f11ae1cfae3409dbc4d : SUCCESS in 1h 48m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/77b7d3f76cbe4e6f9988a00112ebfd09 : SUCCESS in 1h 44m 19s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/4fe93c7d66d7433f9dc4e8f28138e305 : SUCCESS in 1h 39m 12s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/bc76e541f68b4fbb9d051ce2a0f88eaf : SUCCESS in 40m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94c559a7354841549d12e45a487fdbab : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"4505933f80223da149468967f1ccb50897acc17c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-04 04:48:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a0cb448e6750fc06ca5aba8e58182a983030d0c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 07:07:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68857a675db849d18dfb2f3139daedb4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c185b6302cb418a9d4fe43fb5bd02a0 : SUCCESS in 3m 32s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/280ec6dcc6a14c00bf852c266ba35507 : SUCCESS in 1h 40m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/98f7455b72f74384a4ceae28806d8e0a : SUCCESS in 1h 27m 44s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8874c6f2eae547f4bc35cb39188bef5e : SUCCESS in 1h 20m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7581b56eaef34cb2b56bc45214db753f : SUCCESS in 1h 59m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/83f6f6efea61433db32ec52eba7ea706 : SUCCESS in 36m 39s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/f4b43d319a1140228dd21fdbb5a73920 : SUCCESS in 1h 43m 55s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/20ebb1e3f31b4e9292604d0abf7bf84b : SUCCESS in 1h 13m 03s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/4b819c5c8d4d464f8a1fd383bff20673 : SUCCESS in 1h 24m 53s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/d01c5dcc9a63426a96db829659b21737 : SUCCESS in 1h 41m 01s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/7084c6b95fbb45e4b40d38d536a131af : SUCCESS in 37m 17s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/818ea65611f541fd89eef572f83dd45f : SUCCESS in 1h 17m 08s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/4fb16ec8d8764247895a8b9ae119d5ec : SUCCESS in 1h 18m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/752bf2d0675c48e398127965632a2f0c : SUCCESS in 1h 21m 54s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/79f52a7d016e4d52960963d1e83a1ea7 : SUCCESS in 1h 33m 00s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/e0135b13807d474ea605490b8a2d9a6d : SUCCESS in 35m 10s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/6db3fdee00634683af78d524e97a940f : SUCCESS in 2h 02m 12s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/3b0d128d3fd04a34b4b4d2a4fbce3311 : SUCCESS in 1h 20m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/5f10f874d30348f398a2d74b72045c98 : SUCCESS in 1h 37m 36s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/c5182c9ccd424aa18eec79a26dbaac43 : SUCCESS in 1h 35m 06s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/600b632e9ecd4340815c46f039819787 : SUCCESS in 34m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b904538a9f23448da62bb41f87d530d7 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":2},{"id":"48452517b8c9440dcdc2efe9ef0407f5316906bf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-04 12:56:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f0b3273397b73ce885f301915fdd0fc19e732d5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 12:56:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a1d6832ca31fd734f083293f198278ef24c0c020","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 15:26:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de26387b224c4b9a91e66843467faa3c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2a15e21014f4d96aa26566b3cee72c7 : SUCCESS in 3m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/663111e8f2374f84a11a3d99d84e172b : SUCCESS in 1h 14m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d7eef1f3a6dd4da8a119e6f2fff3a596 : SUCCESS in 1h 55m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/18c33dee7389447fb473a2a9f2d88bff : SUCCESS in 1h 31m 32s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/45172dbd99a14e878289650dac95add9 : SUCCESS in 1h 49m 53s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/82c191d4310444348124e3b8adb67544 : SUCCESS in 1h 19m 40s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/2c1c2414e7614f2aac1860d9dc93c41f : SUCCESS in 2h 20m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db97cc8cf0a74f00a88bc9fb1652e885 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":2},{"id":"d2966aca0634f6b864f136470b785946773d5646","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 15:26:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"97409c6992f87f323e135933945ecf8bc03eece3","revisions":{"3b671ebb8a8026448c857b4414f245624ab63f12":{"kind":"REWORK","_number":1,"created":"2023-09-22 06:15:19.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/88/896188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/88/896188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/88/896188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/88/896188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/88/896188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/88/896188/1"}}},"commit":{"parents":[{"commit":"370f71e9877fad13883413225bf25051c0c9c391","subject":"Mark \"test_subport_connectivity_soft_reboot\" as unstable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/370f71e9877fad13883413225bf25051c0c9c391"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-09-22 06:11:17.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-09-22 06:11:20.000000000","tz":330},"subject":"Use RE2 compatible regex for branches","message":"Use RE2 compatible regex for branches\n\nZuul has deprecated the usage of negative regex[1],\nthis patch switches the branches regex to the supported\none.\n\n[1] https://zuul-ci.org/docs/zuul/latest/project-config.html#regex\n\nCloses-Bug: #2037034\nChange-Id: I66dcb56ab5ba9cfdefbeb25f1d01871ea0d0d362\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3b671ebb8a8026448c857b4414f245624ab63f12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3b671ebb8a8026448c857b4414f245624ab63f12"}]},"branch":"refs/heads/master"},"97409c6992f87f323e135933945ecf8bc03eece3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-04 04:48:05.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/88/896188/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/88/896188/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/88/896188/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/88/896188/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/88/896188/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/88/896188/2"}}},"commit":{"parents":[{"commit":"209d363cce20d48d21455a196086a19f1d0d97e1","subject":"Temporary turn off l3_ha in ovs/lb master/bobcat jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/209d363cce20d48d21455a196086a19f1d0d97e1"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-09-22 06:11:17.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2023-10-04 04:48:05.000000000","tz":0},"subject":"Use RE2 compatible regex for branches","message":"Use RE2 compatible regex for branches\n\nZuul has deprecated the usage of negative regex[1],\nthis patch switches the branches regex to the supported\none.\n\n[1] https://zuul-ci.org/docs/zuul/latest/project-config.html#regex\n\nCloses-Bug: #2037034\nChange-Id: I66dcb56ab5ba9cfdefbeb25f1d01871ea0d0d362\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/97409c6992f87f323e135933945ecf8bc03eece3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/97409c6992f87f323e135933945ecf8bc03eece3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
