)]}'
{"id":"openstack%2Fneutron-tempest-plugin~896403","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I111eeba5c29cd7d703e6663be12e8e7e7659919e","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"neutron_t_plugin_remove_dup_tests","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-11-22 12:19:32.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-11-22 12:19:32.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-11-22 10:10:57.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"35229":{"account":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"last_update":"2023-11-22 12:19:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I111eeba5c29cd7d703e6663be12e8e7e7659919e","subject":"Remove test duplications between tempest and n-t-p plugin","status":"MERGED","created":"2023-09-25 13:41:28.000000000","updated":"2023-11-22 12:19:32.000000000","submitted":"2023-11-22 12:19:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"896403-neutron_t_plugin_remove_dup_tests","meta_rev_id":"78024ed7d6a5e6faa2a480ee3055376c389bf650","_number":896403,"virtual_id_number":896403,"owner":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-22 12:19:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"recommended":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2023-10-23 08:01:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":1,"date":"2023-10-03 08:30:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2023-11-22 10:10:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-11-21 21:10:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2023-11-22 10:10:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-25 13:44:05.000000000","updated_by":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-09-25 15:40:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-26 07:56:00.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"CC"},{"updated":"2023-10-03 08:30:00.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2023-10-04 08:29:42.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-10-04 08:29:42.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-10-23 08:01:59.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"state":"REVIEWER"},{"updated":"2023-11-21 21:10:08.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"ee16e5230d364d912c2b6f0ea8eae12ddcf7d5d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-09-25 13:41:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9ce51173e32332decf79026984fbd4fd816c6f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 15:40:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9f448aa45b244c1b6bdf357747eeec3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22d92c91d7eb41f5b5fdc5045fd15471 : SUCCESS in 4m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5d2c706c7624434685fa4fdcc664083c : FAILURE in 1h 46m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c16c59cd7bf244cb81e4dde7f39a16b1 : SUCCESS in 1h 46m 23s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/725d6e320b7b4f70853a99eb94daece5 : SUCCESS in 1h 23m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/43f2ebaabddd4fbe9b034b2ade677bd4 : SUCCESS in 1h 47m 14s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/00f9a78417cb45f3a8bba5a0a61d8606 : SUCCESS in 57m 49s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/00a4838162c143aca0caed08fc6ce5ec : SUCCESS in 1h 20m 19s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/e7caf0059e1f4071b08c8b06d4bdd05c : SUCCESS in 1h 08m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/92d5941b16844c81ac7e847fd08e877a : SUCCESS in 1h 12m 06s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/4ec121f1f12448af9faba6e23d9b3d41 : SUCCESS in 1h 23m 28s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/729bb0391062424e931e3e601e345470 : SUCCESS in 1h 51m 04s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/bdc78ad577754495aad12c8a3d1f2d5d : SUCCESS in 1h 05m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/56401ec48cb64bfb983bd34a9fd0763a : SUCCESS in 1h 17m 56s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/870b597906084a64af2cf8b10f4de948 : SUCCESS in 1h 31m 22s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/8a491cbdb1344e3b911e4a0ef41c1dfd : SUCCESS in 1h 27m 59s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/fe44f835bcb2421ebf8abe903911bfdc : SUCCESS in 1h 12m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/067f2cecae46457c9942b2236440b352 : SUCCESS in 1h 34m 20s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/16ff5edf0dae414286dabf68fbc26275 : SUCCESS in 1h 19m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e432d98395a4e4fa041d35e829724af : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0eca3768cde64fa530f4bb7b1cefa17251f51366","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-09-26 07:56:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c5c6fc168a07a4b794b19d4563666e0afff1fba2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 10:55:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61a7aaee2f8d4aa8ad4ec35b3ef26f3c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea4f0158304a4edaa3c9d8d2445ef769 : SUCCESS in 3m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8a714fcfc371431597af905e75521186 : TIMED_OUT in 2h 49m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6b63507ece954f96b8ab850692de1cf5 : SUCCESS in 1h 46m 43s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3e5bb74cd97946f4894d7d0401e2aa1b : SUCCESS in 1h 34m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1234b7a2e9f7477c90aeb02d6f99e341 : SUCCESS in 1h 53m 43s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/cd4f186f17b34dd088939526c12b6fff : SUCCESS in 41m 53s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/bab3a02b156d4b25821e31b768896f18 : SUCCESS in 1h 21m 32s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/80ac6bb40a8e4fcf88f2dd07b875ef43 : SUCCESS in 1h 07m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/c673eab0a95f4e88bd3bf2d7050aedb8 : SUCCESS in 1h 41m 58s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/9457ce65924c4864bfcd8702062a4bd7 : SUCCESS in 1h 10m 20s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/239a910abe454546b8ac7da75a0406c1 : SUCCESS in 1h 38m 43s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/56646047834f4a05a5e6575aac562733 : SUCCESS in 1h 51m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/1594d0eae36240279f0be4beaae2e7a8 : SUCCESS in 1h 26m 46s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/776e096c6e66442da23e72963b1a2807 : SUCCESS in 1h 33m 59s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/ad0ba03c54d84efca6c9d3f93250f5af : SUCCESS in 1h 36m 50s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/4c97822e01524c1e9b012a606dd5f005 : SUCCESS in 1h 29m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/b3fb7914544a480fa014bfac6d1bd177 : SUCCESS in 1h 26m 11s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/259ce23b3d4145e49b774c2850fdaecf : SUCCESS in 1h 44m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa693db8bd3b47ecb436332570f7e7c4 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"8436ff4b89fdf9ba9a3a87614b36e829d6597fb4","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-10-02 13:28:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1eef62f4ae212bf8cc1599edd4e13bebf94e5f1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 15:33:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86eaf0dba7e54d9ebdd868143ef5f833\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48a611ddcd9c46f8a24d2c7ffd807049 : SUCCESS in 4m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7b1b5b80abc24161a5248c43d787f067 : SUCCESS in 1h 29m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/251a51a24110427ab1ceba5bb7c04e30 : SUCCESS in 1h 53m 39s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5a249df738f0412ea26b62e91d98a0ca : SUCCESS in 1h 26m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/279c22fee74b447ca1eba64e7db5d56b : SUCCESS in 1h 56m 39s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/b6de46978cd34248aa618466e1b2dfe1 : SUCCESS in 40m 52s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/0c2cc9c6fe084b8f999c36248f77431b : SUCCESS in 1h 38m 41s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/2aa9d5f5446a4860abd4001b725f93a4 : SUCCESS in 1h 03m 48s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/091321a10afa4683a5ace3695d166f2f : SUCCESS in 1h 26m 36s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/94c9fe4677c341c09d22218f0934db19 : SUCCESS in 1h 13m 38s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/c8dba795f76c43d9a44257c6277147c7 : SUCCESS in 1h 26m 48s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/0bc950273fd342c0a59480fee309d09e : SUCCESS in 1h 23m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/a6ebd6faafdb499986927179842c7678 : SUCCESS in 1h 43m 34s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/18fd575522664da490f3245acf3f5b3b : SUCCESS in 1h 28m 18s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/6b96f8e785554c729e7983b28d781be1 : SUCCESS in 1h 28m 19s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/6525a152a1f946e3834148a5d2f7d9c9 : SUCCESS in 1h 14m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/1ab640fe259a4f0caebb516a1de03851 : SUCCESS in 1h 48m 55s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/c1edad8a90f8489dbb0d559bf433ca67 : SUCCESS in 1h 34m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a130a3e605a6460ab984ff872489f826 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"22efe78b5ba8c4c731f81780ea979f15bcf04b11","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-10-03 08:30:00.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4ea2c9b2b53816dd43a65b66a8fdaa32e9256ef4","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-10-23 08:01:59.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ede19b8572b6acae23093153d560d88893f4d285","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-11-08 13:26:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec2408094222acad638b3d4ea8949950a0120507","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-21 21:10:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2905f9d6b12bf1dccc7ec454e06c028877c9a217","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-22 10:10:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"75badb2a1058deb2ea5816bd7f3d270a2c307cfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 10:11:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01d3d2c0a66f8a2b2f9b7cfbc6286a2639160fdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 12:19:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da76f1e790a94c3f8fff9b98236cc84b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eb30cd4fbc44160922fd024bc61626d : SUCCESS in 4m 10s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f3fc281ecaef44ea9ff33bd39d6aa595 : SUCCESS in 1h 30m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/191fe83230974c6e8e2c4d27714d7ce8 : SUCCESS in 2h 07m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4a36ca3ac4d84222be6affab1068e752 : SUCCESS in 1h 57m 26s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4da471ca2fe646b397a1feb59b6a34cb : SUCCESS in 1h 29m 51s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/8f052e2bbe9a460798e59523476ba34a : SUCCESS in 51m 13s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/58ada2c5df6349b2b0070d37ab9dca86 : SUCCESS in 1h 56m 07s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/76feeba0d4104bcbaf0bf011464498c4 : SUCCESS in 1h 55m 33s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/0b2058a6377e414dad10f1985c890221 : SUCCESS in 2h 03m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99b6bd8cac4b43399d3b6a68d1c059c6 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"78024ed7d6a5e6faa2a480ee3055376c389bf650","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 12:19:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99c85da78729ed8f03b83abd3f5cceb1555c3ffd","revisions":{"99c85da78729ed8f03b83abd3f5cceb1555c3ffd":{"kind":"REWORK","_number":1,"created":"2023-09-25 13:41:28.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/03/896403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/03/896403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/896403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/896403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/896403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/896403/1"}}},"commit":{"parents":[{"commit":"370f71e9877fad13883413225bf25051c0c9c391","subject":"Mark \"test_subport_connectivity_soft_reboot\" as unstable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/370f71e9877fad13883413225bf25051c0c9c391"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-09-25 13:38:42.000000000","tz":120},"committer":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-09-25 13:41:26.000000000","tz":120},"subject":"Remove test duplications between tempest and n-t-p plugin","message":"Remove test duplications between tempest and n-t-p plugin\n\nExtraDHCPOptionsIpV6TestJSON and ExtraDHCPOptionsTestJSON -\nhttps://opendev.org/openstack/tempest/src/branch/master/tempest/\\\napi/network/test_dhcp_ipv6.py has same functionality in tempest\nand neutron-tempest-plugin, so with removing these tests from\nneutron-tempest-plugin we can get rid of whole module, reducing\nthe test duplications.\n\nChange-Id: I111eeba5c29cd7d703e6663be12e8e7e7659919e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/99c85da78729ed8f03b83abd3f5cceb1555c3ffd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/99c85da78729ed8f03b83abd3f5cceb1555c3ffd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
