)]}'
{"id":"openstack%2Fneutron-tempest-plugin~896410","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I4df71d0d249c522fea17081c183bfa24a3c51743","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"neutron_t_plugin_remove_dup_tests","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2023-12-04 08:57:14.000000000","reason":"\u003cGERRIT_ACCOUNT_30674\u003e replied on the change","reason_account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2023-10-19 08:19:06.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-12-04 12:44:31.000000000","reason":"Change was abandoned"},"35229":{"account":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"last_update":"2023-12-04 12:44:31.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4df71d0d249c522fea17081c183bfa24a3c51743","subject":"Remove test duplications between tempest and n-t-p plugin","status":"ABANDONED","created":"2023-09-25 15:30:10.000000000","updated":"2023-12-04 12:44:31.000000000","total_comment_count":10,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"b405df78a4b20d249accdbe817739fd3345a348d","_number":896410,"virtual_id_number":896410,"owner":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-10-19 10:57:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":-1,"date":"2023-11-21 21:18:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"CC":[{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-25 18:22:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-26 07:57:36.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2023-10-23 08:19:11.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"state":"CC"},{"updated":"2023-11-21 21:18:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"fe66e779db83731b31b626083478d8a4ee3a6407","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-09-25 15:30:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d3a8d9e8931bcbbf9ff43d8fe37edb8e2128cf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 18:22:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ffb975796514cafa86e2603b2e1fb60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9015d3512de84c71b35519a0cadea033 : SUCCESS in 3m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b44a1436bd7441fb9f972e7aaa4e8a13 : FAILURE in 1h 51m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d57355d8c87644589ed02fdb9329b87d : FAILURE in 1h 59m 20s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b53a1f2e20a14afbac3da5f52d87cf68 : TIMED_OUT in 2h 50m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/370911f6679243b6b2625191a2245346 : FAILURE in 1h 32m 29s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/7f9baee9620b432dba0762e3bfdda6ac : FAILURE in 59m 41s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/cc3601ea5a3145c3bd55fb1a7f5a8d81 : SUCCESS in 1h 26m 47s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/5c2537df6e71422084fd89fdd2753b2c : SUCCESS in 1h 14m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/1981133077fe4ff2b017935747e19bfa : SUCCESS in 1h 33m 51s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/956243eb37714b09ac72fe4edf195b23 : SUCCESS in 1h 22m 34s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/f612e9e1551e45cd84dd1a0ba61964b7 : FAILURE in 1h 35m 43s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/6dd9bd169a21489894f37e0d994c9abd : FAILURE in 1h 11m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/ae8513f61f92498d990ec70663127e7f : FAILURE in 1h 52m 11s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/4675ea62ff434c7ab9f240a8939f7880 : FAILURE in 1h 34m 20s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/006a513064944d698be6764416b54a62 : FAILURE in 1h 51m 45s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/b181d0925cde4c07b3537a10c5bffcfe : FAILURE in 1h 35m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/3720d8cb9fad4bd9aa3899d9d84d37c1 : FAILURE in 1h 37m 10s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/114e5de5131044c8a8de3ad7bfe2d99e : FAILURE in 1h 21m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eba03ab887c48e1b425f48b161c8526 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"424789fd8319dbb462c327ba6f2825b9889aa9ec","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-09-26 07:57:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c232ff3462d47b8fd015217497fc2c133057bab3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-10-04 08:54:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d1439de7e2f743d091f81ed8620f545fa69f00e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 11:06:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae2b148469da4737b1e65eda0c10bf51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1bd3f8b59a64d83a7f39c8e97a59a1a : SUCCESS in 3m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5d08945926b54f849ac84225b15b3edd : FAILURE in 2h 09m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/42762510f98e4872aebd216034f05891 : SUCCESS in 1h 45m 38s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c997ef35adcf4929877bf7c6eb293112 : FAILURE in 1h 44m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bc094cd20c244d70bdd8e7f10cea4eca : SUCCESS in 2h 02m 13s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/5a7a1548d53f4e8886067798a2ff0257 : SUCCESS in 58m 23s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/8e9de2ced75b496cb26a41e00a8136a9 : SUCCESS in 1h 35m 02s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/c71b055cc10647c3acbf1e21d8f95dde : SUCCESS in 1h 35m 02s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/75d3fd13a8224f82906bb1168e222537 : SUCCESS in 1h 39m 52s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/5e529e6d62034168a4c9f8a3563268f6 : SUCCESS in 1h 26m 07s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/5e00c9c1288c436caa3a909c69f758b3 : SUCCESS in 1h 51m 29s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/4016e81ec9124b5d91c99b8da27da04c : SUCCESS in 1h 38m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/683294a6a5f84c5faa8bb2aa81e4f3b8 : SUCCESS in 1h 47m 23s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/36e5d2d5d80644789cb26ae50c5ebbf6 : SUCCESS in 1h 58m 02s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/0124ba343b354be5bfe2af9dee35e909 : SUCCESS in 1h 53m 43s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/79ad4693944d4bb09c79267b55ceb842 : SUCCESS in 1h 11m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/14f6fec82d7b42b996d8b26c1fdeb7e0 : SUCCESS in 1h 22m 22s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/d6322c19dcd646ac8bf4a2a8acaf3166 : SUCCESS in 1h 30m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78d312abb4734c3a9dea299b8be36ea7 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":2},{"id":"e836d3843ba152fd848283471bb84d3835a5e44b","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-10-19 08:19:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba90cdd1cd5897632ffa1e07f430376648ca8daa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 10:57:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d0c889af9014fff9f578ca7b8399415\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2da4d190c3248189bc7b1418efb3ae0 : SUCCESS in 4m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/eaef5183b39e43b1aeeeb4985c646ad7 : SUCCESS in 1h 57m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5a3ab563b2074cb69c688d2e894f7dce : SUCCESS in 1h 47m 55s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/43dbd24b65d44ae09e4d83b600a9017c : SUCCESS in 1h 53m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2c8f3c627b08458dac6dfcb3d1066e91 : SUCCESS in 2h 10m 55s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/8b7aba8061ad4d55b157413cb5ae9433 : SUCCESS in 59m 34s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/3ea48f82466a4765be1b20ad185aae73 : SUCCESS in 1h 31m 18s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/4a8ead0f2424457d8832382c0964af8f : SUCCESS in 1h 10m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/557b81ba9c0441a5b9d648569178f0db : SUCCESS in 1h 35m 50s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/666006f5d1bd4c3aad4b057e7e1ddec1 : SUCCESS in 1h 28m 41s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/075ba8e6ef234372982bb242f9a89b25 : SUCCESS in 1h 34m 38s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/3a3b5d2cf3144914a22a0cfae0096777 : SUCCESS in 1h 21m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/853b2ae8f3a34e87ac1842f0cb058920 : SUCCESS in 1h 49m 51s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/859979f7a0dd4fb089271b57c9e5252e : SUCCESS in 1h 53m 31s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/d25d728858714a2f9ca3b8dd00e6c265 : SUCCESS in 1h 46m 30s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/adc5751cbd5745e9832c6703b00dd4c6 : SUCCESS in 1h 32m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/9e6b9cabf3194fecb07229d877cb4a21 : SUCCESS in 1h 28m 36s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/7fd8a27165b84239a8126e6a8fc14e6c : SUCCESS in 1h 19m 32s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/b2780bd47e91401e9d48626381496de5 : SUCCESS in 1h 23m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/4d1b505b6c414646917513dae00437a7 : SUCCESS in 1h 41m 03s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/1a3c7efe21c746ebabfb298df44bafcd : SUCCESS in 1h 45m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fd9bc013c064b46b431200997ba3ca7 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"d5cf927fe49b9c8077b0d314db014f20d8198c51","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-10-23 08:19:11.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"bbaa2d9e72bf9063a5447225bc822ced6b5ec977","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-11-15 16:42:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9093f9d9e3a8b148fc3a6d1568cf24c8335cf680","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-11-16 08:21:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"16048ba6f93754a41e7e23c2dccf6a43ce6547f6","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-11-20 13:21:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bc5552326bcfe134838224104616d9c24851d49c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-21 21:18:56.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"810264bcd4e32339652883aa446fcc669a98fadf","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-12-04 08:57:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b405df78a4b20d249accdbe817739fd3345a348d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-12-04 12:44:31.000000000","message":"Abandoned\n\naccidentally created another review: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/901051","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3d9a1148b0f25e11043780694b144d4e0bf4874a","revisions":{"4f7c665bcfcfeb7cba5cc30abec5408ba6278cdb":{"kind":"REWORK","_number":1,"created":"2023-09-25 15:30:10.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/10/896410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/10/896410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/896410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/896410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/896410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/896410/1"}}},"commit":{"parents":[{"commit":"370f71e9877fad13883413225bf25051c0c9c391","subject":"Mark \"test_subport_connectivity_soft_reboot\" as unstable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/370f71e9877fad13883413225bf25051c0c9c391"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-09-25 15:24:29.000000000","tz":120},"committer":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-09-25 15:30:08.000000000","tz":120},"subject":"Remove test duplications between tempest and n-t-p plugin","message":"Remove test duplications between tempest and n-t-p plugin\n\nDvrRoutersNegativeTest - https://opendev.org/openstack/tempest\\/\nsrc/branch/master/tempest/api/network/test_routers_negative.py has\ntest_router_create_tenant_distributed_returns_forbidden and\nSecGroupTest - https://opendev.org/openstack/tempest/src/branch\\/\nmaster/tempest/api/network/test_security_groups.py has\ntest_create_list_update_show_delete_security_group implemented in\nboth tempest and neutron-tempest-plugin, so with removing these\ntests from neutron-tempest-plugin we can reduce the test\nduplications.\n\nChange-Id: I4df71d0d249c522fea17081c183bfa24a3c51743\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4f7c665bcfcfeb7cba5cc30abec5408ba6278cdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4f7c665bcfcfeb7cba5cc30abec5408ba6278cdb"}]},"branch":"refs/heads/master"},"3d9a1148b0f25e11043780694b144d4e0bf4874a":{"kind":"REWORK","_number":2,"created":"2023-10-04 08:54:24.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/10/896410/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/10/896410/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/896410/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/896410/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/896410/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/896410/2"}}},"commit":{"parents":[{"commit":"370f71e9877fad13883413225bf25051c0c9c391","subject":"Mark \"test_subport_connectivity_soft_reboot\" as unstable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/370f71e9877fad13883413225bf25051c0c9c391"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-09-25 15:24:29.000000000","tz":120},"committer":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-10-04 08:53:45.000000000","tz":120},"subject":"Remove test duplications between tempest and n-t-p plugin","message":"Remove test duplications between tempest and n-t-p plugin\n\nDvrRoutersNegativeTest - https://opendev.org/openstack/tempest\\/\nsrc/branch/master/tempest/api/network/test_routers_negative.py has\ntest_router_create_tenant_distributed_returns_forbidden and\nSecGroupTest - https://opendev.org/openstack/tempest/src/branch\\/\nmaster/tempest/api/network/test_security_groups.py has\ntest_create_list_update_show_delete_security_group implemented in\nboth tempest and neutron-tempest-plugin, so with removing these\ntests from neutron-tempest-plugin we can reduce the test\nduplications.\n\nChange-Id: I4df71d0d249c522fea17081c183bfa24a3c51743\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3d9a1148b0f25e11043780694b144d4e0bf4874a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3d9a1148b0f25e11043780694b144d4e0bf4874a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
