)]}'
{"id":"openstack%2Fneutron-tempest-plugin~900772","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I90b9cf495f35aca7a9171dc685696ae4cdd4d0f2","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"fix_importing_sec_grp_tests","attention_set":{},"removed_from_attention_set":{"29350":{"account":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"last_update":"2023-11-16 17:23:31.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-11-16 17:23:31.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-11-16 08:42:51.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-11-16 15:11:04.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I90b9cf495f35aca7a9171dc685696ae4cdd4d0f2","subject":"Fix importing security group tests","status":"MERGED","created":"2023-11-13 10:11:33.000000000","updated":"2023-11-16 17:23:31.000000000","submitted":"2023-11-16 17:23:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900772-fix_importing_sec_grp_tests","meta_rev_id":"ec5ad195d9c4adc11119e3d98f1aab3b785b2e52","_number":900772,"virtual_id_number":900772,"owner":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-16 17:23:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-11-16 08:42:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-11-16 15:11:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-11-16 15:11:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-13 12:38:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-13 16:18:47.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-11-14 13:30:41.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-11-15 13:21:54.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-11-16 15:11:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"b32edbd032a8f690ae72bc24b70a59c225fe7cba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2023-11-13 10:11:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3f67392c3137da479bc635bf6a4df2fb995bfaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 12:38:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b626a28f5114ea7bc05b7a2e29a13bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d9b8e34e76841f8bcebb0cdbbc9d77b : SUCCESS in 3m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/85f142303b6c4f3e9d26f19038505d86 : SUCCESS in 1h 56m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/011fd351f9ea47c8862330f77cb162fc : SUCCESS in 1h 49m 50s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/7271d5dcd5304c53afd400bded0757bb : SUCCESS in 1h 44m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3a577afb750448f5877377d01dbc3bed : SUCCESS in 1h 54m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e423c6a9edd04c1498b653535384ca77 : SUCCESS in 47m 42s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/bece106ca19946d5ac3c34e3ebacef57 : SUCCESS in 1h 31m 04s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/b3ed95dc66044b638a6b38f0849efbc8 : SUCCESS in 1h 31m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/2ec7ddb48299430cbba441fdc588a1ba : SUCCESS in 1h 24m 09s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/823385653e584e3cb00f19cda5c1c769 : SUCCESS in 1h 32m 09s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/8a8778aa592d477e8aed4db4cdd73faf : SUCCESS in 33m 08s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/9016d0a46ddb41fc82d65deccdf42cca : SUCCESS in 1h 52m 54s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/b5f6b62738614cc0b9f3f245a6db4ddf : SUCCESS in 1h 24m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/35d4c690f16b4c1cbef6d5dc388805e2 : SUCCESS in 1h 42m 46s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/119b0921f325483d8bc7029cd0926159 : SUCCESS in 1h 40m 53s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/1586c6090c594eadb380500716dbbb02 : SUCCESS in 39m 42s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/72d0243b1b7a4ca2a5f64c7c7d2c5ca5 : SUCCESS in 1h 36m 16s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/05f833e1b91b4a58b5f50ec626c3b57b : SUCCESS in 1h 38m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/45cd8d4e09af4c78bf1e8d36fb61c910 : SUCCESS in 1h 23m 35s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/08e40b972c784372914bd535924d5502 : SUCCESS in 2h 24m 11s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/416f9be23cfa493e9bf5ee67abef464b : SUCCESS in 30m 31s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/f707ac78468c4d08aa03435abb7f16bf : SUCCESS in 2h 04m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/3b0f441099714eb685af94abd2188cff : SUCCESS in 1h 48m 13s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/9675dcfe362244cba3fe6afefbd1eac2 : SUCCESS in 2h 06m 46s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/fe808a642205484abb90e37850c98e6a : SUCCESS in 41m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf4d5000edd248348f67aa182243065a : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c5b4deed8b7a779634c03b52b4e6728cada5874d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-13 16:18:47.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7c0441b22b3fb932f6b325094569135168bc90a1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-11-14 13:30:41.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"117c913330800a1f8a280d91a91102a6be9592da","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-15 13:21:54.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"04431c316a640fedeef78a84e71ef26d25673c3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2023-11-15 13:53:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1623fbcabcfa40be0c13eed0ad27701e60600857","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2023-11-15 13:56:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cf69167386280f7a95da988c8de535919f62a162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 16:32:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5255060456fb447db3d8f8be5ff76b6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b9463e61b4e4f4ca54500a8d52c7613 : SUCCESS in 3m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c70ecc144193409eb93eef4cea58c47b : SUCCESS in 1h 32m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ba603f10418a49fbaae17c371c6cdf11 : SUCCESS in 2h 37m 20s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/928b9ef50b9243e19986bb1f5b051de1 : SUCCESS in 1h 41m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e549636edde4470794cbfd854366efb1 : SUCCESS in 1h 40m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f6eba669935e47d88edac19c7662c1c2 : SUCCESS in 39m 58s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/dcb4dbd6aaea4bd3a19de7f146079cbd : SUCCESS in 1h 28m 04s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/2fe7d23850c246ceaddf75e998602022 : SUCCESS in 1h 36m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/4866713b053b4f7493838149a88ae1b6 : SUCCESS in 1h 44m 47s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/d54c9fb34fee400382f409ff22c616c0 : SUCCESS in 1h 55m 37s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/255055e6ddff4e70a672afa971f58bc3 : SUCCESS in 33m 46s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/62ed34722e3445d08b51862446567a92 : SUCCESS in 2h 07m 06s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/a118ce108eda4fbc9e977e543a28e7a3 : SUCCESS in 1h 35m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/b48f9c093b714d319264f548756dcf00 : SUCCESS in 2h 11m 34s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/69d114910fcf4200a3553752cc9111bd : SUCCESS in 1h 41m 59s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/fa91e3fe887b49c4a63d53c163f45925 : SUCCESS in 35m 38s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/54ce2e2dc6294a51bf0627bf501470af : SUCCESS in 1h 49m 03s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/763ad3eac1894101ba9afcc87deb98d8 : SUCCESS in 1h 51m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/4a9e4b07c1954fcdb76ac3fbb635c708 : SUCCESS in 1h 39m 17s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/57555a5ae374461ab47f9550629d1df9 : SUCCESS in 2h 15m 46s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/31f6f47056d349a9821eb37165ac4ed6 : SUCCESS in 40m 49s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/957c5f102e304e5bbbaf41c54984c29a : SUCCESS in 1h 45m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/4361e8ce05b144788a071ab69afb29f1 : SUCCESS in 2h 19m 57s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/20723360b6974c769f14e543d8e5f4d2 : SUCCESS in 1h 45m 48s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/d70626d0f4bc4c72be072813d3d276ba : SUCCESS in 40m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f96fbd312d94abc927945d3867beece : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"c393d62af8044e9b62aa9e2078873f771c15835d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-16 08:42:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c64205d4e753ac351c91f5d468605521c04e0cee","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-16 15:11:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f59dd0db8858f9f75e61869857b1b8477094163e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 15:11:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5656a47175c1a4454e409f83304084d627a1eb3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 17:23:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9ea51caad074ababbd1804dd6c48005\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1263f06bcb3c4502aa1dd5547101546b : SUCCESS in 4m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3b4df5ddc9d64ea6a6a0cee95bb8bc94 : SUCCESS in 1h 33m 30s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4a25f301d7cf4d14a188358b853f945c : SUCCESS in 2h 02m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ae14c858d8c742c393018dd73b12c186 : SUCCESS in 1h 53m 01s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/48cc42bc5b2b448f88d46820ca69e65b : SUCCESS in 1h 43m 00s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/bc2e2731734e4189a76099fef40f2d47 : SUCCESS in 1h 37m 46s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/8b31cff4e94f4b54bca517d91a1fe4b9 : SUCCESS in 1h 39m 14s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/09e26d740a944bc589db0a3c369b6a9b : SUCCESS in 2h 09m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9a5b300928b46b19d6f94e00e283f1d : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":2},{"id":"ec5ad195d9c4adc11119e3d98f1aab3b785b2e52","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 17:23:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"052808c2e62489fbe5e560131f3f7209b364a764","revisions":{"b3faa840ff1954a72e1c48bc6b90f8e8bab31fb1":{"kind":"REWORK","_number":1,"created":"2023-11-13 10:11:33.000000000","uploader":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"ref":"refs/changes/72/900772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/72/900772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/72/900772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/72/900772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/72/900772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/72/900772/1"}}},"commit":{"parents":[{"commit":"7944b08b1f40c3948f8c77af7189526d7bb710e0","subject":"Use default RBAC policies in other projects together with Neutron policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7944b08b1f40c3948f8c77af7189526d7bb710e0"}]}],"author":{"name":"Roman Safronov","email":"rsafrono@redhat.com","date":"2023-11-13 10:03:06.000000000","tz":120},"committer":{"name":"Roman Safronov","email":"rsafrono@redhat.com","date":"2023-11-13 10:03:06.000000000","tz":120},"subject":"Fix importing security group tests","message":"Fix importing security group tests\n\nImporting StatelessSecGroupDualStackDHCPv6StatelessTest was causing\na type error in case CONF.neutron_plugin_options.firewall_driver\nwas set to \u0027openvswitch\u0027 [1]. This patch fixes the issue by\nseparation code to a class with base functions, that is never skipped\nand classes with tests that optionally can be skipped.\n\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d2249407\n\nChange-Id: I90b9cf495f35aca7a9171dc685696ae4cdd4d0f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b3faa840ff1954a72e1c48bc6b90f8e8bab31fb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b3faa840ff1954a72e1c48bc6b90f8e8bab31fb1"}]},"branch":"refs/heads/master"},"052808c2e62489fbe5e560131f3f7209b364a764":{"kind":"REWORK","_number":2,"created":"2023-11-15 13:53:48.000000000","uploader":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"ref":"refs/changes/72/900772/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/72/900772/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/72/900772/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/72/900772/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/72/900772/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/72/900772/2"}}},"commit":{"parents":[{"commit":"7944b08b1f40c3948f8c77af7189526d7bb710e0","subject":"Use default RBAC policies in other projects together with Neutron policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7944b08b1f40c3948f8c77af7189526d7bb710e0"}]}],"author":{"name":"Roman Safronov","email":"rsafrono@redhat.com","date":"2023-11-13 10:03:06.000000000","tz":120},"committer":{"name":"Roman Safronov","email":"rsafrono@redhat.com","date":"2023-11-15 13:53:04.000000000","tz":120},"subject":"Fix importing security group tests","message":"Fix importing security group tests\n\nImporting StatelessSecGroupDualStackDHCPv6StatelessTest was causing\na type error in case CONF.neutron_plugin_options.firewall_driver\nwas set to \u0027openvswitch\u0027 [1]. This patch fixes the issue by\nseparation code to a class with base functions, that is never skipped\nand classes with tests that optionally can be skipped.\n\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d2249407\n\nChange-Id: I90b9cf495f35aca7a9171dc685696ae4cdd4d0f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/052808c2e62489fbe5e560131f3f7209b364a764"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/052808c2e62489fbe5e560131f3f7209b364a764"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
