)]}'
{"id":"openstack%2Fneutron-tempest-plugin~901051","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I01e15af5e1ecb34bff2da0ad7c9dc25930189a0e","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-12-06 14:29:01.000000000","reason":"\u003cGERRIT_ACCOUNT_35229\u003e replied on the change","reason_account":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"}}},"removed_from_attention_set":{"35229":{"account":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"last_update":"2023-12-06 14:29:01.000000000","reason":"\u003cGERRIT_ACCOUNT_35229\u003e replied on the change","reason_account":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"}}},"hashtags":[],"change_id":"I01e15af5e1ecb34bff2da0ad7c9dc25930189a0e","subject":"Remove test duplications between tempest and n-t-p plugin","status":"NEW","created":"2023-11-15 16:37:32.000000000","updated":"2023-12-06 14:29:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f921144274986a7fdd5c31e8b0b38a594933d4d4","_number":901051,"virtual_id_number":901051,"owner":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-12-04 15:04:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":-1,"date":"2023-12-04 15:12:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-15 18:49:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-21 21:18:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"0e3bd8f7166cf6f7a8e12e9a6c825b51f39ae342","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-11-15 16:37:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81e813fbee7b8840e6311436cd1f6f76861b867c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-11-15 16:39:33.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"205fc2fc125cf1053cb952627ac7d98deb73cde7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 18:49:09.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b59d8a1e4bfe49189d1f75b4defa0dc9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b738a4056e23438d86258d5345ac3364 : SUCCESS in 4m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ba42b3c142e741e4917cca6d474b27a9 : SUCCESS in 1h 56m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3327e53a8420429a832e9f0a8fb7423d : SUCCESS in 2h 02m 30s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/d3f9a7dcbbcf4c6f86edac0d4a750164 : SUCCESS in 1h 30m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/eb0727f05a4d4ff0a3ee0655206981a9 : SUCCESS in 1h 45m 37s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/294ce683a41047dc8d8ffaf4d9a11e11 : SUCCESS in 48m 37s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/a6880f99a6af43888ada480181143ab1 : SUCCESS in 1h 22m 36s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/5c6372e14a5e44f5a63d0e60e9d6245c : SUCCESS in 1h 30m 19s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/d0060965c1d54a5080562b900eefcd68 : SUCCESS in 1h 54m 33s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/2c68ee36b6c54934838bad1f636de922 : SUCCESS in 1h 17m 55s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/83374b10a3ac43938eb4746159e885ac : SUCCESS in 1h 48m 43s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/a797921289c74048bbc78aa4be1cc03b : SUCCESS in 1h 28m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/ba97f6ef825c4085b53abb508f8a5efa : SUCCESS in 1h 27m 56s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/b91c94a23c3e48269f55b9e68400c32a : SUCCESS in 1h 33m 52s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/d44b9a5f133544e19f4a627de1e1515a : SUCCESS in 1h 40m 49s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/27032e8b3d3044af83e8e090fde92ad2 : SUCCESS in 1h 32m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/6f14bab67ee14722876d6f476dfd93ac : SUCCESS in 2h 08m 49s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/f8833d6b360e41d8b806275624996488 : SUCCESS in 1h 26m 34s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/8da72fcd863744e0ad288fd3423cade4 : SUCCESS in 1h 44m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/77291a1f03a4409aadb90a2814a0a153 : SUCCESS in 2h 00m 13s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/ca273f0f94ff4121be2b18a10ec85638 : SUCCESS in 1h 50m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/155fd7204ddc403fa97388be3b97b80b : FAILURE in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"f4e58d93f3e59403109a2135b168b66bdb139552","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-11-20 10:39:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"87273af916bdb7819284e7ab76e3be914cb0368d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 13:20:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbf91642511c4c66aed33199bfbae134\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf25c30c2b744662b5ee5af4676731a2 : SUCCESS in 3m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6a4d4ce664dc41e58d30162c1f95c1d6 : SUCCESS in 1h 21m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/54ecb5a3b31c4252a053ee0232d0904a : SUCCESS in 1h 38m 53s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/e55ec885622d4d2cbf54cd0d42b7b9f6 : SUCCESS in 1h 47m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ad6774d0b962448aa52bc1d2a6793e87 : SUCCESS in 1h 42m 25s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/c0d4067444f7445596fba512c2999576 : SUCCESS in 1h 01m 42s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/458ea5f0350a4a6694fefdabf1007113 : SUCCESS in 1h 35m 16s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/841f1c91c61645f994dac73b005641ed : SUCCESS in 1h 44m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/fd0345c3f78a4fc293b32b010879d2e2 : SUCCESS in 1h 50m 49s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/5f3d5cb45e024bac919593b0b5a242dd : SUCCESS in 1h 48m 32s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/6cac5987af614608a1b31478dc306eed : SUCCESS in 1h 34m 01s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/960004fb41aa4b6083aee6ef97eff9a2 : SUCCESS in 1h 30m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/f1a7eb5bf13e4815ba27b04e3f5e66d7 : SUCCESS in 1h 28m 24s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/c1f770b06ec64940872cb73760ccdf73 : SUCCESS in 1h 35m 27s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/baf0bfd44379470e8df9d7c23206de54 : SUCCESS in 1h 35m 47s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/43930f0ed9434a729477f0c0876b53f5 : SUCCESS in 1h 45m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/c64c22cdf436418f8d47c4e742a9caa2 : SUCCESS in 2h 08m 15s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/92289f0fde8c469cba4003df79a828fe : SUCCESS in 1h 47m 03s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/cc18d9fe550b4ed980ddaa4ff8716ed3 : SUCCESS in 2h 06m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/d80833ac3cf04a22be676027fc755ddf : SUCCESS in 2h 04m 58s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/2fda9e2579ee4c09ae81f335287ca371 : SUCCESS in 1h 32m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0018c0adf641494aaa4871ce53ee5de7 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":3},{"id":"9e64d551383eb26960a02ca4a50df6d498a05590","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-21 21:18:04.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9c7f79f9d10440367f35762ef1366347a58c8ba5","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-12-04 12:47:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ade2abbc4511c7f0df39beae2b90a8379fdc958e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-12-04 12:51:00.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c65aa7a33d196e26670956a940dd2f471ec3bac6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 15:04:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4148448ff1e5471798055f38aebd88b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24cd54d3ca5a4a6fabe07d3525ed3cbd : SUCCESS in 4m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4ba983a3ad414538b4bea07746730020 : SUCCESS in 1h 47m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/74f5277a222d4c9eafad884c8af1dec1 : SUCCESS in 2h 11m 54s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/6d940e028906498ca22b58aef52de162 : SUCCESS in 1h 34m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/28f1b1d43c754fb3b5108bf6510c7432 : SUCCESS in 1h 42m 18s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/3330228ec87843be85823cd479708169 : SUCCESS in 56m 54s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/159eb187114c43a79434e34e05fe8f32 : SUCCESS in 1h 40m 05s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/8d5114d4f9d640fa9d2654e15d469036 : SUCCESS in 1h 11m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/f54762856cf3467192e773eca014687a : SUCCESS in 1h 38m 46s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/3b7793ace5c84a1684d818932b8456f9 : SUCCESS in 1h 20m 24s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/95dd2d1dcc7b431abff3ec6130108c02 : SUCCESS in 1h 44m 02s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/59e08578ee354efa932b34ae76325fcd : SUCCESS in 1h 37m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/14466c5dd71640f6ac0a649382ad01cc : SUCCESS in 1h 42m 02s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/bc513c3ca8354ce99380722bfc36dc23 : SUCCESS in 1h 45m 59s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/27691b208d2b474b9529d6ed148a2ce8 : SUCCESS in 1h 39m 54s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/00fb3ab6adf34dc19b4a71841f6902a2 : SUCCESS in 1h 26m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/aa3d3faa62d34fa1b6006a19cbf4e34d : SUCCESS in 2h 04m 25s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/25fe29821c384aa48678efb58c8f4396 : SUCCESS in 1h 38m 06s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/c964f9364ca94b2b8f7155f29a0c15a3 : SUCCESS in 1h 15m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/44b296ae658445f4ab1b3d27090b9b6f : SUCCESS in 2h 09m 55s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/c5660ec60593493b9c695846882e5037 : SUCCESS in 1h 58m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a1ad9977ef148d1b699c986e10b60a9 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":4},{"id":"d789b3f363da106286924151bb09a3eefe363fa8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-04 15:12:14.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f921144274986a7fdd5c31e8b0b38a594933d4d4","author":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"date":"2023-12-06 14:29:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f561b302d7f656c8481da5597a59708ea2398ad3","revisions":{"87bf662a937bad0c78a8f749393108c4f5e61963":{"kind":"REWORK","_number":1,"created":"2023-11-15 16:37:32.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/51/901051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/51/901051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/51/901051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/51/901051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/51/901051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/51/901051/1"}}},"commit":{"parents":[{"commit":"370f71e9877fad13883413225bf25051c0c9c391","subject":"Mark \"test_subport_connectivity_soft_reboot\" as unstable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/370f71e9877fad13883413225bf25051c0c9c391"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-09-25 15:24:29.000000000","tz":120},"committer":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-11-15 16:37:21.000000000","tz":60},"subject":"Remove test duplications between tempest and n-t-p plugin","message":"Remove test duplications between tempest and n-t-p plugin\n\nDvrRoutersNegativeTest[1] - has\ntest_router_create_tenant_distributed_returns_forbidden and\nSecGroupTest[2] has\ntest_create_list_update_show_delete_security_group implemented in\nboth tempest and neutron-tempest-plugin, so with removing these\ntests from neutron-tempest-plugin we can reduce the test\nduplications.\n\n[1] - https://opendev.org/openstack/tempest/src/branch/master/tempest/api/network/test_routers_negative.py\n[2] - https://opendev.org/openstack/tempest/src/branch/master/tempest/api/network/test_security_groups.py\nChange-Id: I4df71d0d249c522fea17081c183bfa24a3c51743\n\nChange-Id: I01e15af5e1ecb34bff2da0ad7c9dc25930189a0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/87bf662a937bad0c78a8f749393108c4f5e61963"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/87bf662a937bad0c78a8f749393108c4f5e61963"}]},"branch":"refs/heads/master"},"552e74bffdb0bfe4c902d72ef46f98608aaae117":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-11-15 16:39:33.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/51/901051/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/51/901051/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/51/901051/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/51/901051/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/51/901051/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/51/901051/2"}}},"commit":{"parents":[{"commit":"370f71e9877fad13883413225bf25051c0c9c391","subject":"Mark \"test_subport_connectivity_soft_reboot\" as unstable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/370f71e9877fad13883413225bf25051c0c9c391"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-09-25 15:24:29.000000000","tz":120},"committer":{"name":"Jakub Skunda","email":"jskunda@redhat.com","date":"2023-11-15 16:39:33.000000000","tz":0},"subject":"Remove test duplications between tempest and n-t-p plugin","message":"Remove test duplications between tempest and n-t-p plugin\n\nDvrRoutersNegativeTest[1] - has\ntest_router_create_tenant_distributed_returns_forbidden and\nSecGroupTest[2] has\ntest_create_list_update_show_delete_security_group implemented in\nboth tempest and neutron-tempest-plugin, so with removing these\ntests from neutron-tempest-plugin we can reduce the test\nduplications.\n\n[1] - https://opendev.org/openstack/tempest/src/branch/master/tempest/api/network/test_routers_negative.py\n[2] - https://opendev.org/openstack/tempest/src/branch/master/tempest/api/network/test_security_groups.py\n\nChange-Id: I01e15af5e1ecb34bff2da0ad7c9dc25930189a0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/552e74bffdb0bfe4c902d72ef46f98608aaae117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/552e74bffdb0bfe4c902d72ef46f98608aaae117"}]},"branch":"refs/heads/master","description":"Edit commit message"},"957963f1e275b94cbfcc2c6ea64e2d4a76e2026b":{"kind":"REWORK","_number":3,"created":"2023-11-20 10:39:41.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/51/901051/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/51/901051/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/51/901051/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/51/901051/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/51/901051/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/51/901051/3"}}},"commit":{"parents":[{"commit":"370f71e9877fad13883413225bf25051c0c9c391","subject":"Mark \"test_subport_connectivity_soft_reboot\" as unstable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/370f71e9877fad13883413225bf25051c0c9c391"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-09-25 15:24:29.000000000","tz":120},"committer":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-11-20 10:39:21.000000000","tz":60},"subject":"Remove test duplications between tempest and n-t-p plugin","message":"Remove test duplications between tempest and n-t-p plugin\n\nDvrRoutersNegativeTest[1] - has\ntest_router_create_tenant_distributed_returns_forbidden and\nSecGroupTest[2] has\ntest_create_list_update_show_delete_security_group implemented in\nboth tempest and neutron-tempest-plugin, so with removing these\ntests from neutron-tempest-plugin we can reduce the test\nduplications.\n\n[1] - https://opendev.org/openstack/tempest/src/branch/master/tempest/api/network/test_routers_negative.py\n[2] - https://opendev.org/openstack/tempest/src/branch/master/tempest/api/network/test_security_groups.py\n\nChange-Id: I01e15af5e1ecb34bff2da0ad7c9dc25930189a0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/957963f1e275b94cbfcc2c6ea64e2d4a76e2026b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/957963f1e275b94cbfcc2c6ea64e2d4a76e2026b"}]},"branch":"refs/heads/master"},"f561b302d7f656c8481da5597a59708ea2398ad3":{"kind":"REWORK","_number":4,"created":"2023-12-04 12:51:00.000000000","uploader":{"_account_id":35229,"name":"Jakub Skunda","display_name":"jskunda","email":"jskunda@redhat.com","username":"jskunda"},"ref":"refs/changes/51/901051/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/51/901051/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/51/901051/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/51/901051/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/51/901051/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/51/901051/4"}}},"commit":{"parents":[{"commit":"370f71e9877fad13883413225bf25051c0c9c391","subject":"Mark \"test_subport_connectivity_soft_reboot\" as unstable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/370f71e9877fad13883413225bf25051c0c9c391"}]}],"author":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-09-25 15:24:29.000000000","tz":120},"committer":{"name":"jskunda","email":"jskunda@redhat.com","date":"2023-12-04 12:50:41.000000000","tz":60},"subject":"Remove test duplications between tempest and n-t-p plugin","message":"Remove test duplications between tempest and n-t-p plugin\n\nDvrRoutersNegativeTest[1] - has\ntest_router_create_tenant_distributed_returns_forbidden and\nSecGroupTest[2] has\ntest_create_list_update_show_delete_security_group implemented in\nboth tempest and neutron-tempest-plugin, so with removing these\ntests from neutron-tempest-plugin we can reduce the test\nduplications.\n\n[1] - https://opendev.org/openstack/tempest/src/branch/master/tempest/api/network/test_routers_negative.py\n[2] - https://opendev.org/openstack/tempest/src/branch/master/tempest/api/network/test_security_groups.py\n\nChange-Id: I01e15af5e1ecb34bff2da0ad7c9dc25930189a0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f561b302d7f656c8481da5597a59708ea2398ad3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f561b302d7f656c8481da5597a59708ea2398ad3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
