)]}'
{"id":"openstack%2Fneutron-tempest-plugin~908549","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ic7a8ac8e98f4b56b2ae2b76071c6dc64c0f34066","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-12 06:06:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic7a8ac8e98f4b56b2ae2b76071c6dc64c0f34066","subject":"Replace usage of deprecated path for test decorators","status":"MERGED","created":"2024-02-09 04:08:11.000000000","updated":"2024-02-12 06:06:32.000000000","submitted":"2024-02-12 06:06:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908549","meta_rev_id":"e188fd769ec277d48d4e0f2cb5979a7f12bdb369","_number":908549,"virtual_id_number":908549,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-12 06:06:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2024-02-12 05:28:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2024-02-09 23:30:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2024-02-12 05:28:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-09 05:26:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-09 23:30:13.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-02-12 05:28:24.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"28c97eac0bd5c3f6cce7a1d3746f576e79ccec43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-09 04:08:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86c87adcda73e5c3e15329ceaed564ba47dde44a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-09 04:17:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"da699579e287aca438b718eb71f949b1441e82f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-09 04:17:30.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"24dbacf6ba27baf1a182e2049882f95c3482dfc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 05:26:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20b3d09ad8eb4ec0b9c16381baed156f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b9f62d48f7c4a08ac7e7247e2471133 : SUCCESS in 4m 52s\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/61b61ba0725d445f93901cf131360854 : FAILURE in 16m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bc3c3974a4045b484f0575cb0f74134 : SUCCESS in 3m 20s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/3da92c9243584a0fbecb7f667a8936e2 : SUCCESS in 41m 55s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/de18f11838db414999e8f9eb5af483a2 : SUCCESS in 1h 02m 19s\n- neutron-tempest-plugin-vpnaas-yoga https://zuul.opendev.org/t/openstack/build/03f80820a067454c9d1d12532d604684 : FAILURE in 13m 54s\n- neutron-tempest-plugin-vpnaas-zed https://zuul.opendev.org/t/openstack/build/fc7d757d959a4dbca2b1831bdceb4567 : SUCCESS in 46m 42s\n- neutron-tempest-plugin-vpnaas-2023-1 https://zuul.opendev.org/t/openstack/build/3386334a1f5e4474a8e7330ae0500f8d : SUCCESS in 42m 50s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/ee4e97a233044846a171e1a9f2ccea0f : SUCCESS in 41m 28s","accounts_in_message":[],"_revision_number":3},{"id":"c7e633ce06d2c7079f2e57a776ac3291ef50f93a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-09 05:32:28.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"bdd4260106650967cc1eb124c2d1804a10aaca5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-09 06:19:24.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"232c121646f2f08f2e46e3a78cf68fe63010b1ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 07:32:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5bf2975663f4d0098777ecee6be25ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90c9d7cf6fc34e1ba82db9d3c768c268 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1f701395df044ac86c0bdc49d10476d : SUCCESS in 3m 29s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/21de897289f6495b988871706f5559cd : SUCCESS in 53m 12s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/e19ce204c9f44a33aaed771c8ff4ebc2 : SUCCESS in 52m 56s\n- neutron-tempest-plugin-vpnaas-zed https://zuul.opendev.org/t/openstack/build/a788710dffcb4f21a2ae8a87e433b3e2 : SUCCESS in 49m 16s\n- neutron-tempest-plugin-vpnaas-2023-1 https://zuul.opendev.org/t/openstack/build/6c672ef785184f64b598b0a269b4dedc : SUCCESS in 1h 02m 56s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/d3c14d2818c94701bbdd6c9d84cca2a5 : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":5},{"id":"bc99246715ae1292b83cf41926a4495bbab6865c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-02-09 23:30:13.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"33e8f87125215c919dae3f900a8f5b632ac50390","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-02-12 05:28:24.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c3b6925ef08d16f9d4bdf2bffa5f223e0377e0c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 05:28:52.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"527b32b4be4a77c8b07b16421e812d8d4b7eaf4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 06:06:32.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8b04ba3d1c74dadb248db1a7fe237c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db1ec26b50ab481ba59d0488b75b34df : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55cd169386ea4362b9b7a8c3f70c4158 : SUCCESS in 3m 14s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/3cb016dbd4934d989ca5727592f8c0b8 : SUCCESS in 37m 08s","accounts_in_message":[],"_revision_number":5},{"id":"e188fd769ec277d48d4e0f2cb5979a7f12bdb369","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 06:06:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"815b0d26b4d7a54df464cb368adcb74c3ee1c339","revisions":{"226b105e494289a73440a270833fc3bf9cf23e50":{"kind":"REWORK","_number":1,"created":"2024-02-09 04:08:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/908549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/49/908549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/1"}}},"commit":{"parents":[{"commit":"7d950d78a2898977325a737bad8cf890ea0c0075","subject":"Merge \"Add VNaaS OVN job to the project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7d950d78a2898977325a737bad8cf890ea0c0075"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-09 04:06:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-09 04:07:59.000000000","tz":540},"subject":"Replace usage of deprecated path for test decorator","message":"Replace usage of deprecated path for test decorator\n\nThese decorators were moved to the common module in tempest 17.0.0[1].\n\n[1] cd36841ca25b39b9c8ad1b83e0abd0a191d538a0\n\nChange-Id: Ic7a8ac8e98f4b56b2ae2b76071c6dc64c0f34066\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/226b105e494289a73440a270833fc3bf9cf23e50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/226b105e494289a73440a270833fc3bf9cf23e50"}]},"branch":"refs/heads/master"},"8f16a90f887bf14c10040e1dc0c75dceb5ad1c99":{"kind":"REWORK","_number":2,"created":"2024-02-09 04:17:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/908549/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/49/908549/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/2"}}},"commit":{"parents":[{"commit":"7d950d78a2898977325a737bad8cf890ea0c0075","subject":"Merge \"Add VNaaS OVN job to the project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7d950d78a2898977325a737bad8cf890ea0c0075"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-09 04:06:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-09 04:17:18.000000000","tz":0},"subject":"Replace usage of deprecated path for test decorator","message":"Replace usage of deprecated path for test decorator\n\nThese decorators were moved to the common module in tempest 17.0.0[1].\n\n[1] cd36841ca25b39b9c8ad1b83e0abd0a191d538a0\n\nChange-Id: Ic7a8ac8e98f4b56b2ae2b76071c6dc64c0f34066\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8f16a90f887bf14c10040e1dc0c75dceb5ad1c99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8f16a90f887bf14c10040e1dc0c75dceb5ad1c99"}]},"branch":"refs/heads/master"},"b1087cb51194e748a569d1afad0a2f8b7a06a031":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-02-09 04:17:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/908549/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/49/908549/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/3"}}},"commit":{"parents":[{"commit":"7d950d78a2898977325a737bad8cf890ea0c0075","subject":"Merge \"Add VNaaS OVN job to the project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7d950d78a2898977325a737bad8cf890ea0c0075"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-09 04:06:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-09 04:17:30.000000000","tz":0},"subject":"Replace usage of deprecated path for test decorators","message":"Replace usage of deprecated path for test decorators\n\nThese decorators were moved to the common module in tempest 17.0.0[1].\n\n[1] cd36841ca25b39b9c8ad1b83e0abd0a191d538a0\n\nChange-Id: Ic7a8ac8e98f4b56b2ae2b76071c6dc64c0f34066\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b1087cb51194e748a569d1afad0a2f8b7a06a031"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b1087cb51194e748a569d1afad0a2f8b7a06a031"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9c87ed8b8f3797afd43bf8f20828e82662a1e3bc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-02-09 05:32:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/908549/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/49/908549/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/4"}}},"commit":{"parents":[{"commit":"d776334a401053cfb6b45bd1d45dd1233a72b8f0","subject":"Remove yoga jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d776334a401053cfb6b45bd1d45dd1233a72b8f0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-09 04:06:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-09 05:32:28.000000000","tz":0},"subject":"Replace usage of deprecated path for test decorators","message":"Replace usage of deprecated path for test decorators\n\nThese decorators were moved to the common module in tempest 17.0.0[1].\n\n[1] cd36841ca25b39b9c8ad1b83e0abd0a191d538a0\n\nChange-Id: Ic7a8ac8e98f4b56b2ae2b76071c6dc64c0f34066\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9c87ed8b8f3797afd43bf8f20828e82662a1e3bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9c87ed8b8f3797afd43bf8f20828e82662a1e3bc"}]},"branch":"refs/heads/master","description":"Rebase"},"815b0d26b4d7a54df464cb368adcb74c3ee1c339":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-02-09 06:19:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/908549/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/49/908549/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/49/908549/5"}}},"commit":{"parents":[{"commit":"ff96cdecac4240f9bab35f17a5d88f5ed020e889","subject":"Stop running yoga jobs in check/gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ff96cdecac4240f9bab35f17a5d88f5ed020e889"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-09 04:06:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-09 06:19:24.000000000","tz":0},"subject":"Replace usage of deprecated path for test decorators","message":"Replace usage of deprecated path for test decorators\n\nThese decorators were moved to the common module in tempest 17.0.0[1].\n\n[1] cd36841ca25b39b9c8ad1b83e0abd0a191d538a0\n\nChange-Id: Ic7a8ac8e98f4b56b2ae2b76071c6dc64c0f34066\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/815b0d26b4d7a54df464cb368adcb74c3ee1c339"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/815b0d26b4d7a54df464cb368adcb74c3ee1c339"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
