)]}'
{"id":"openstack%2Fneutron-tempest-plugin~911445","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ieabab4ccaa0e2a365939425dff0c0776839251eb","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"remove-testscenarios","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-03-07 11:10:46.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-03-07 13:28:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieabab4ccaa0e2a365939425dff0c0776839251eb","subject":"Remove usage of testscenarios and replace it with ddt library","status":"MERGED","created":"2024-03-06 11:27:19.000000000","updated":"2024-03-07 13:28:30.000000000","submitted":"2024-03-07 13:28:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"911445-remove-testscenarios","meta_rev_id":"649efa487adb4c68a5a3769efd4a8151fd622e78","_number":911445,"virtual_id_number":911445,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-07 13:28:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-03-07 11:10:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-03-07 09:25:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-03-07 11:10:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-06 11:53:41.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-03-06 14:23:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-03-06 15:21:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-07 09:25:07.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"5ed7de0cf62f2843f47e6b9c03be0975f3785b49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-03-06 11:27:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"656aa84e09b2fe427c3849eb146db035425ddf73","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-03-06 11:53:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17410133f4415f8d192c19c15c0d117a42be7869","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-03-06 12:11:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b90b0fde5bd5ee6af81922a5d63dd6e1e1175391","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-03-06 12:15:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91f81d27210d3f03c697532653d9d99827f487f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-03-06 13:27:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"366843eae391c690297c50c8c5b49dd6981d8648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 15:21:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eb53cfc1b474fcea0dbd6ba8458edbf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed300632efd541679c493d3823629a67 : SUCCESS in 3m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d1dcfacd75044b418272540fef2143fa : SUCCESS in 1h 24m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6784b404b35e4217bb24c5a57209a963 : SUCCESS in 1h 28m 17s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/0948daa4fb44486a8cff7ea6b221e8cd : SUCCESS in 1h 18m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dd70c1667cae42b1b671e1c7edd60f25 : SUCCESS in 1h 50m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c620861b0f84472c82fdf85308604553 : SUCCESS in 35m 58s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/dffb8a349d7b4e3688367ba734242f4d : SUCCESS in 1h 34m 44s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/b32c5141b4fa48ec804d40a33975452f : SUCCESS in 1h 12m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/591b40d8d71e47f7803f557b394c57a3 : SUCCESS in 1h 27m 05s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/d08a684865434be9885a3a480fb94543 : SUCCESS in 1h 34m 40s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/f0127da3577e4b9f83ba740e442ba982 : SUCCESS in 30m 08s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/ce9c3a4ffe9a4b17b135475d893f98a1 : SUCCESS in 1h 42m 56s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/7215891a278b4562a83a9efa5eafde35 : SUCCESS in 1h 06m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/1a2f0aaa8295457b93d01abd4bcefd71 : SUCCESS in 1h 33m 57s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/a9525a4f8b7c4725a6b55400a97f9483 : SUCCESS in 1h 22m 14s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/5b670ef2c67543ba927bf796d835b260 : SUCCESS in 39m 29s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/ee26157d1e7344fe92fbe830c1fded58 : SUCCESS in 1h 09m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/b943c75abfa64421ae11cf7556130cc3 : SUCCESS in 1h 27m 03s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/dda15ff156324b1d914030cf63ea5956 : SUCCESS in 1h 17m 07s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/0806a1f48bf44ac4a91308bd0e33e48c : SUCCESS in 40m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bade8265efa040bfa20cc3d8941d07f0 : SUCCESS in 1m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5e8bed4e9984e90bf3e000b92b9f401 : SUCCESS in 2m 46s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/bcf934de73174416bad7b176b696ab7d : SUCCESS in 1h 44m 46s\n- neutron-tempest-plugin-sfc-zed https://zuul.opendev.org/t/openstack/build/c687fea10082493e872bbc2850bee608 : SUCCESS in 1h 16m 57s\n- neutron-tempest-plugin-sfc-2023-1 https://zuul.opendev.org/t/openstack/build/c9de7915b01f48c2900913378eaafaa0 : SUCCESS in 1h 21m 29s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/05fbdc26913d4f9094a6f2070d5cee40 : SUCCESS in 1h 19m 01s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d41caa1f07c14fc4afda698a12b37733 : SUCCESS in 1h 23m 20s\n- neutron-tempest-plugin-bgpvpn-bagpipe-zed https://zuul.opendev.org/t/openstack/build/f2b4dc4a265c42c2a511828f127fc809 : SUCCESS in 1h 17m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/a7ab24ddb25640dd81b9ab5c556aa778 : SUCCESS in 1h 26m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/3e6b3fd29df34e4680e2694feb085034 : SUCCESS in 1h 23m 37s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/b7cb133f58c24c6caf897b2da49cb651 : SUCCESS in 56m 38s\n- neutron-tempest-plugin-dynamic-routing-zed https://zuul.opendev.org/t/openstack/build/7f2b273b4d7d40aba5b4fc0916d1a065 : SUCCESS in 55m 03s\n- neutron-tempest-plugin-dynamic-routing-2023-1 https://zuul.opendev.org/t/openstack/build/554f0529acc54238853acfd142c5b4ac : SUCCESS in 49m 09s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/4cc21b0239a646b49534b5c28f14355a : SUCCESS in 1h 01m 22s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/fc517fccfb10459ebd1eeff54fd06d46 : SUCCESS in 48m 44s\n- neutron-tempest-plugin-fwaas-zed https://zuul.opendev.org/t/openstack/build/a26669a3cc0f40fa87f6304566e2aa3b : SUCCESS in 37m 33s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/ea24369cdc8645a1ac4f88afe7e48b62 : SUCCESS in 45m 00s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/0a49fe3a3e4446eea0cad10fbebe65b4 : SUCCESS in 49m 11s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/054abdf1919b4efba57830866571c01a : SUCCESS in 53m 04s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/6986ab77cdd24d0e91d3ffe98c361641 : SUCCESS in 52m 40s\n- neutron-tempest-plugin-vpnaas-zed https://zuul.opendev.org/t/openstack/build/6e248d6224b34be8a6927d4a4e6e39b5 : SUCCESS in 52m 41s\n- neutron-tempest-plugin-vpnaas-2023-1 https://zuul.opendev.org/t/openstack/build/55c4e51691fa4ebf8fc2089c53af5db0 : SUCCESS in 53m 01s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/8822940077104ec283df25fa784011ab : SUCCESS in 51m 56s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/cc4a9f6445834782aecc9b599ee7bdec : SUCCESS in 1h 33m 48s\n- neutron-tempest-plugin-tap-as-a-service-zed https://zuul.opendev.org/t/openstack/build/11dc09c37f014fc291bb75cbff76abc5 : SUCCESS in 1h 33m 45s\n- neutron-tempest-plugin-tap-as-a-service-2023-1 https://zuul.opendev.org/t/openstack/build/1cd76ca1ded544d6978d337ad64344bf : SUCCESS in 1h 27m 39s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/7d78089c3f114277be20988d54ed53d5 : SUCCESS in 1h 38m 21s","accounts_in_message":[],"_revision_number":2},{"id":"d7b76dc3ebbe33b1c7f3ae13457e0ce97c10b7dd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-03-06 15:25:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"96b337f8fd1dad0c50bc8977d95dd6d1e962d2e6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-03-07 09:25:07.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"872d6c45bc0beaed44c2f17f5ce52e1b0419b869","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-03-07 11:10:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"539ec7641165a0f6da6cc6b28d40e6c60c1aa75b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 11:11:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"038a68ef64b5316568942b6fbb3b4c826de297a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 13:28:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5778a19a0fdc411bb1f3338e425196a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/881f217098b647cfb24628a07253a996 : SUCCESS in 5m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2fa1f400e3bb46e98ee012fdea1f4f24 : SUCCESS in 1h 47m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9283a1cb20184bb99bd0e4fd05961dca : SUCCESS in 2h 15m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9d4e5d98624445c396c1fa1370c41acb : SUCCESS in 1h 17m 32s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/ab929de5edcf4c0fa0e23384b9c5eb20 : SUCCESS in 1h 17m 40s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/18b5359244304d9d8a5430b7d454a113 : SUCCESS in 1h 12m 45s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/5cdb9a2455294233b2fa0145599dd9b2 : SUCCESS in 2h 05m 59s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/66cbc727ea464cf58d2580a8e420f6dc : SUCCESS in 1h 38m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/436b01f8d84743d3b085c6c2fcb5eeff : SUCCESS in 1m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/857770ceb16f435aaf833a976db867b0 : SUCCESS in 3m 35s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/69847795096c465eaf92dd6a8abed3e2 : SUCCESS in 1h 32m 08s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f92042196e7641adbb8086715afb2d23 : SUCCESS in 1h 39m 03s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/b9a75ddfaa2647489252a870e38023f1 : SUCCESS in 1h 04m 10s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/a7800bb2c2d54acd96b31f188aca21ee : SUCCESS in 45m 18s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/7880dd61921c47adaed86c8244167574 : SUCCESS in 47m 34s","accounts_in_message":[],"_revision_number":2},{"id":"649efa487adb4c68a5a3769efd4a8151fd622e78","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 13:28:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ec162e0f56588419e3efb7e70eee8b1d826c307b","revisions":{"fcb2e5b30ba630798345012a62c7b4b845bc2a87":{"kind":"REWORK","_number":1,"created":"2024-03-06 11:27:19.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/45/911445/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/911445/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/911445/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/911445/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/911445/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/911445/1"}}},"commit":{"parents":[{"commit":"468deaf6fbc67fa3abf0b49ddbc19f37b7523c11","subject":"Merge \"[OVN] Do not check L3 agents state with ML2/OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/468deaf6fbc67fa3abf0b49ddbc19f37b7523c11"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-03-06 11:21:24.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-03-06 11:27:15.000000000","tz":60},"subject":"Remove usage of testscenarios and replace it with ddt library","message":"Remove usage of testscenarios and replace it with ddt library\n\nAs our internal QE reported to us, ddt is producing more consistent\nnames of the same tests with different variables then testscenarios.\nWith testscenarios it is something like:\n\nfoo_tempest_plugin.tests.something.SomeClass.test_foo[id-123456](c1,c2)\n\nand with ddt is more similar to tests without scenarios:\n\nfoo_tempest_plugin.tests.something.SomeClass.test_foo_1[id-123456]\n\nApparently due to this inconsistency results of some tests couldn\u0027t be\nproperly reported properly to some tools which we are using downstream.\n\nSo this patch proposes to remove usage of testscenarios in\nneutron_tempest_plugin and replace it with ddt where it is needed.\nActually ddt was already used in some of the tests here so this will\nmake it also more consistent across the tests in that repo.\n\nAlso as part of this effort I relalised that in some cases testscenarios\nwhere used, tests were running e.g. 2 times but actually variable set to\ndifferent value in different scenarios were not used at all. So in such\ncase I simply dropped usage of testscenarios and left only one variant\nof the test to be run.\n\nChange-Id: Ieabab4ccaa0e2a365939425dff0c0776839251eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fcb2e5b30ba630798345012a62c7b4b845bc2a87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fcb2e5b30ba630798345012a62c7b4b845bc2a87"}]},"branch":"refs/heads/master"},"ec162e0f56588419e3efb7e70eee8b1d826c307b":{"kind":"REWORK","_number":2,"created":"2024-03-06 13:27:51.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/45/911445/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/911445/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/911445/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/911445/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/911445/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/911445/2"}}},"commit":{"parents":[{"commit":"468deaf6fbc67fa3abf0b49ddbc19f37b7523c11","subject":"Merge \"[OVN] Do not check L3 agents state with ML2/OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/468deaf6fbc67fa3abf0b49ddbc19f37b7523c11"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-03-06 11:21:24.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-03-06 13:27:39.000000000","tz":60},"subject":"Remove usage of testscenarios and replace it with ddt library","message":"Remove usage of testscenarios and replace it with ddt library\n\nAs our internal QE reported to us, ddt is producing more consistent\nnames of the same tests with different variables then testscenarios.\nWith testscenarios it is something like:\n\nfoo_tempest_plugin.tests.something.SomeClass.test_foo[id-123456](c1,c2)\n\nand with ddt is more similar to tests without scenarios:\n\nfoo_tempest_plugin.tests.something.SomeClass.test_foo_1[id-123456]\n\nApparently due to this inconsistency results of some tests couldn\u0027t be\nproperly reported properly to some tools which we are using downstream.\n\nSo this patch proposes to remove usage of testscenarios in\nneutron_tempest_plugin and replace it with ddt where it is needed.\nActually ddt was already used in some of the tests here so this will\nmake it also more consistent across the tests in that repo.\n\nAlso as part of this effort I relalised that in some cases testscenarios\nwhere used, tests were running e.g. 2 times but actually variable set to\ndifferent value in different scenarios were not used at all. So in such\ncase I simply dropped usage of testscenarios and left only one variant\nof the test to be run.\n\nChange-Id: Ieabab4ccaa0e2a365939425dff0c0776839251eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ec162e0f56588419e3efb7e70eee8b1d826c307b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ec162e0f56588419e3efb7e70eee8b1d826c307b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
