)]}'
{"id":"openstack%2Fneutron-tempest-plugin~912910","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Idb49963618c45dbe5976e32d5db466f35ed534c8","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"OSPRH-2772","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-03-14 17:44:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idb49963618c45dbe5976e32d5db466f35ed534c8","subject":"[Stateless SG] Add test for port security and stateless SGs","status":"MERGED","created":"2024-03-14 10:18:53.000000000","updated":"2024-03-14 17:44:23.000000000","submitted":"2024-03-14 17:44:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"912910-OSPRH-2772","meta_rev_id":"e887dbc0fb0de7b277a6ac5a5c781ec8f9ef484d","_number":912910,"virtual_id_number":912910,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-14 17:44:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2024-03-14 15:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-03-14 15:43:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2024-03-14 15:57:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-14 12:12:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-14 15:43:09.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-03-14 15:57:52.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"d4009de49981c19ba699c108ed76f8fd4064bae6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-03-14 10:18:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de9091b7dc98db31ffac9de63a9e38668a3685e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 12:12:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e39e6f681124b0483f19de0b55b8c0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4457cbb4037940e6b62895e161af37f0 : SUCCESS in 3m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d994a9b25be9485ea7670801cc205500 : SUCCESS in 1h 11m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2173ab5c4e664dfaa3eb574fff46dec6 : SUCCESS in 1h 26m 48s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/5bcb4754398b406ebcb1bdbebd54fa3b : SUCCESS in 1h 08m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9f8e33347538440d86964cd078db4cbb : SUCCESS in 1h 27m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/22ad4252150a4693afbff4fe8f611cc6 : SUCCESS in 35m 21s\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/3c0f150496da4a358e1e071a8b3bdeb8 : SUCCESS in 1h 48m 09s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/35b1c5d1f8ba486ca6cd248db2f1129e : SUCCESS in 1h 10m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/8bea02da4b8e4c9080881198df4b3da0 : SUCCESS in 1h 23m 09s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/59381f4cdfe343bda42a3ce6cb7aa9f2 : SUCCESS in 1h 40m 30s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/c9e1b4f3e2304d9095a94c9d959f7226 : SUCCESS in 33m 13s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/59088669b26f4db0986011c0efa32477 : SUCCESS in 1h 51m 10s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/ba9e877119ed48a4a87727ea12f11886 : SUCCESS in 1h 27m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/f47ea9f6875a4e4a96c947437c90acba : SUCCESS in 1h 36m 59s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/0120dc147cc9481c86ba78e7760fb4cd : SUCCESS in 1h 32m 30s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/1af9bbfa00d74232a45b6f9dc42b6ed4 : SUCCESS in 39m 44s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/8d049b82a3204949a5a3ed7bdc01ef73 : SUCCESS in 1h 32m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/246b70f0e9384e5597b58f1d61d7e69d : SUCCESS in 1h 45m 39s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/578a315aa515479ab8ce7664ad270d97 : SUCCESS in 1h 24m 06s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/6f817cca2f244aabb16b4fc9f913396b : SUCCESS in 38m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3379ef44b0b84bd19df652ab7f9f97bc : SUCCESS in 3m 02s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/5b39b07c04ee4285b9d5ac453f4161ef : SUCCESS in 1h 44m 59s\n- neutron-tempest-plugin-sfc-zed https://zuul.opendev.org/t/openstack/build/1e2105b28b184e8cbb48a0b40cae2290 : SUCCESS in 1h 32m 52s\n- neutron-tempest-plugin-sfc-2023-1 https://zuul.opendev.org/t/openstack/build/efec8b798a6c4e34a80f475283a6e5da : SUCCESS in 1h 34m 18s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/6373b1b0f16146bfb10313b4f7dc53fa : SUCCESS in 1h 32m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/c990d7bd735644c7a11efec6c14945bf : SUCCESS in 1h 29m 05s\n- neutron-tempest-plugin-bgpvpn-bagpipe-zed https://zuul.opendev.org/t/openstack/build/f124e8c4d5a9464fbd903efde397f3c6 : SUCCESS in 1h 29m 06s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/9db3679ea65b45eb9c035d3ed28092f1 : SUCCESS in 1h 11m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/bf5e92b5165048c7a9673a27fcf6b47e : SUCCESS in 1h 31m 19s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/68cd193572124e239fc8735b9bfdc0ba : SUCCESS in 1h 01m 29s\n- neutron-tempest-plugin-dynamic-routing-zed https://zuul.opendev.org/t/openstack/build/b1ececce65b646c9b864bddde2d1514f : SUCCESS in 37m 24s\n- neutron-tempest-plugin-dynamic-routing-2023-1 https://zuul.opendev.org/t/openstack/build/90b4713f41cf4d9885d6af6e7cf36701 : SUCCESS in 43m 18s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/e63342a8095449c9bbc885d7166474cc : SUCCESS in 59m 56s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/6e32ad5f3f2e4aee915db2f35d8c26e2 : SUCCESS in 51m 34s\n- neutron-tempest-plugin-fwaas-zed https://zuul.opendev.org/t/openstack/build/79fd2365e8e84cce9b734d4bbab888fd : SUCCESS in 29m 52s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/8ca0595ba5c54e4c9a2015fe79885165 : SUCCESS in 30m 58s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/a13492fc813b4e2cb34653b47834a53e : SUCCESS in 48m 54s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/4d8d8c283fcf4621b45c42e56f17fa25 : SUCCESS in 54m 31s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/23163c65da684751972e3ed22bf0bf91 : SUCCESS in 47m 02s\n- neutron-tempest-plugin-vpnaas-zed https://zuul.opendev.org/t/openstack/build/3c041346ff5c49119be57876f2d2476f : SUCCESS in 42m 18s\n- neutron-tempest-plugin-vpnaas-2023-1 https://zuul.opendev.org/t/openstack/build/c5500d9344da49d7ac9d2a9c5ff9a4dc : SUCCESS in 37m 27s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/7fa2e8ead4424577a49da8f037ac9d1e : SUCCESS in 56m 03s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/934799cc278540a5bd1a07399ce38a26 : SUCCESS in 1h 28m 01s\n- neutron-tempest-plugin-tap-as-a-service-zed https://zuul.opendev.org/t/openstack/build/c333e755c8bb449d937126e37dc1a665 : SUCCESS in 1h 25m 30s\n- neutron-tempest-plugin-tap-as-a-service-2023-1 https://zuul.opendev.org/t/openstack/build/02b0c6d0f7964e6da5405d2074e78928 : SUCCESS in 1h 27m 11s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/23a6cc4d572e46768b4f52f8bf86e883 : SUCCESS in 1h 30m 07s","accounts_in_message":[],"_revision_number":1},{"id":"fdd7cb36984d091ae6dfdf3de3bbccfaedd32cd4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-03-14 15:43:09.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44d8d8119452aa965fe35d97df61ab7dbfad1636","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-03-14 15:57:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f161948e7fa0241b0d689d39eed9bae49843a1fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 15:58:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"840c61e2af9f1ddc84313149e047d04265e5b077","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 17:44:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1e3feb8a89d49faaa7dc2fbe4529557\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d03566e83227453d849c6cfca4d8e540 : SUCCESS in 4m 58s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9bbe572af148485396909721890c3b25 : SUCCESS in 1h 43m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dd68909d93cc40e798071a9b52a0c00f : SUCCESS in 1h 40m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6b9c2977a4d340c39752e35c203bbad5 : SUCCESS in 1h 26m 38s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/b15433083cad4b6ba0f451754e3811d9 : SUCCESS in 1h 03m 35s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/b7c57a1648bd4d07999d2aae8d04b95b : SUCCESS in 1h 37m 10s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/b6afd3adcd7b4f88b64d9900f02ec7a5 : SUCCESS in 1h 33m 23s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/6001f7768a874e86833a61e5665a1391 : SUCCESS in 1h 42m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa65599b9bb04febb5f08eeb2f117597 : SUCCESS in 4m 07s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/c3bb600cd6bd483aa1f572f0641880e4 : SUCCESS in 1h 39m 17s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/b835912d653b4e4db7a89f4674fe9d44 : SUCCESS in 1h 34m 18s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/9604067ab33143b09cc8290676960fa3 : SUCCESS in 57m 21s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/3d7036796c904f3888f5bb01e61c4d9e : SUCCESS in 40m 34s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/1f620d9c9cd84fb08df0e3d059b66cff : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e887dbc0fb0de7b277a6ac5a5c781ec8f9ef484d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 17:44:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ad4ddcbaa1690b40ea61a4ce575ec198408dd715","revisions":{"ad4ddcbaa1690b40ea61a4ce575ec198408dd715":{"kind":"REWORK","_number":1,"created":"2024-03-14 10:18:53.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/10/912910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/10/912910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/912910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/912910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/912910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/10/912910/1"}}},"commit":{"parents":[{"commit":"ec162e0f56588419e3efb7e70eee8b1d826c307b","subject":"Remove usage of testscenarios and replace it with ddt library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ec162e0f56588419e3efb7e70eee8b1d826c307b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-03-14 10:15:59.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-03-14 10:15:59.000000000","tz":60},"subject":"[Stateless SG] Add test for port security and stateless SGs","message":"[Stateless SG] Add test for port security and stateless SGs\n\nThis patch modifies exisitng port_security_removed_added to be run\nfor both stateful and stateless SGs.\n\nChange-Id: Idb49963618c45dbe5976e32d5db466f35ed534c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ad4ddcbaa1690b40ea61a4ce575ec198408dd715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/ad4ddcbaa1690b40ea61a4ce575ec198408dd715"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
