)]}'
{"id":"openstack%2Fneutron-tempest-plugin~914637","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I33e87b3abd1a7c6422b6b1bf21c5e8c0ad9d33bb","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"add_logging_ext","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-04-04 09:04:58.000000000","reason":"Change was abandoned"},"33341":{"account":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"last_update":"2024-04-04 09:04:13.000000000","reason":"\u003cGERRIT_ACCOUNT_33341\u003e replied on the change","reason_account":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}}},"hashtags":[],"change_id":"I33e87b3abd1a7c6422b6b1bf21c5e8c0ad9d33bb","subject":"Add logging as common extension","status":"ABANDONED","created":"2024-03-28 11:33:56.000000000","updated":"2024-04-04 09:04:58.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"02a92eeb2ac0d351ea60f97ae01df62b49579d24","_number":914637,"virtual_id_number":914637,"owner":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-28 13:20:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2024-03-29 07:59:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-28 13:20:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-29 07:59:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"c34f05b47f8fcf5820cbb864e669143ebc06220a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2024-03-28 11:33:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c864bb5dff25cdb55b114ae248899fd9668bdb0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 13:20:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6370c30dc7a460180b5a5b12edfa9a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95756b62021f4c049fbaf0d7ed6c52a6 : SUCCESS in 3m 47s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/85aebf1016a848bc9febc2d08ce849af : FAILURE in 1h 20m 34s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/0f23c6c00b9f43208ba6e3f1b5b038f4 : SUCCESS in 1h 09m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/e308efc424784f479e2c4178c17c94f5 : SUCCESS in 1h 40m 22s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/4ce4178908fd4380b94d6c48e6079281 : SUCCESS in 41m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed3cfcbd5b18416c98440123a99bbec4 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f154333520582afe159ee155c2ae30f5e0e72731","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2024-03-28 16:15:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"35fff91d404d05b42d3c337c859e8e63a520f873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 18:38:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c446c77d8ab4dfdbac41bd0be5963e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a770b0fc71474462aca5bb0ece1fbe74 : SUCCESS in 4m 54s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/9be02907c97a4bee92ec06b13843bf88 : FAILURE in 2h 20m 39s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/3011cbc6d4ba408d87ef598d92658d9b : SUCCESS in 1h 08m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/0ab3090d768b4cce93d6e03a0243d973 : SUCCESS in 1h 37m 27s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/db1210e784f8474a98ca288ad4ac2c3e : SUCCESS in 31m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e77b7a769c574695a93218d5badbad91 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"630791f919f5e5c848ba5ffb43a9ac863ea92157","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-03-29 07:59:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1731cb6677372a7aa2ca3354bd98654076ca3b36","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2024-04-04 09:04:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"02a92eeb2ac0d351ea60f97ae01df62b49579d24","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2024-04-04 09:04:58.000000000","message":"Abandoned\n\nFix handled locally in whitebox plugin instead, https://review.opendev.org/c/x/whitebox-neutron-tempest-plugin/+/913824/24/zuul.d/master_jobs.yaml#104","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cecc0cd85285afba68339bfd4a41dafa62eb38f0","revisions":{"cecc0cd85285afba68339bfd4a41dafa62eb38f0":{"kind":"REWORK","_number":1,"created":"2024-03-28 11:33:56.000000000","uploader":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"ref":"refs/changes/37/914637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/37/914637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/37/914637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/37/914637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/37/914637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/37/914637/1"}}},"commit":{"parents":[{"commit":"395e6fa2905cdf819937edfd4008d51e5826d246","subject":"Add Active Active L3 GW API test cases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/395e6fa2905cdf819937edfd4008d51e5826d246"}]}],"author":{"name":"Maor Blaustein","email":"blue@redhat.com","date":"2024-03-28 11:29:35.000000000","tz":120},"committer":{"name":"Maor Blaustein","email":"blue@redhat.com","date":"2024-03-28 11:29:35.000000000","tz":120},"subject":"Add logging as common extension","message":"Add logging as common extension\n\nLogging feature is supported in both ML2/OVN and ML2/OVS,\nadding extension to enable related testing in plugin,\nand also in dependent whitebox neutron tempest plugin.\n\nChange-Id: I33e87b3abd1a7c6422b6b1bf21c5e8c0ad9d33bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cecc0cd85285afba68339bfd4a41dafa62eb38f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cecc0cd85285afba68339bfd4a41dafa62eb38f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
