)]}'
{"id":"openstack%2Fneutron-tempest-plugin~916971","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ia7e740fd955a405ac0c5a7e38cc0849ce92433f3","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"30380":{"account":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"last_update":"2024-04-30 06:27:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia7e740fd955a405ac0c5a7e38cc0849ce92433f3","subject":"add concurrent list_firewall_groups","status":"ABANDONED","created":"2024-04-25 07:35:22.000000000","updated":"2024-04-30 06:27:57.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9790d44116dd7d358e19d1c6a3371c48e9b2bd80","_number":916971,"virtual_id_number":916971,"owner":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-04-25 08:35:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-25 08:35:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-25 09:37:38.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"}],"messages":[{"id":"712d4475de8c3b854ca140c651b784a8dbb4cefb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2024-04-25 07:35:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddda63e36c7584084d1ee3e6329031d2c072413b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 08:35:16.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a58305f659b0439996412b5ef9980764\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42afa55d2360492fa722360e69370365 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34c4dd7466e3410ab1812a1da15754cf : FAILURE in 3m 11s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/de91822101b14b12ab0dc67c82f4d09b : SUCCESS in 50m 06s\n- neutron-tempest-plugin-fwaas-zed https://zuul.opendev.org/t/openstack/build/60094a22122c404b9656976821de2c8e : SUCCESS in 34m 09s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/5cf2377f53b147b7879dff3abb864705 : SUCCESS in 31m 47s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/947a78d1c0574265b5fee63f0c6c046a : SUCCESS in 35m 03s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/4bcfcccc2113449592f49b97c4fdbd31 : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":1},{"id":"0b4a7872e3a2933c9a4fcbc2ff0f40d0bf7cef55","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-25 09:37:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9790d44116dd7d358e19d1c6a3371c48e9b2bd80","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2024-04-30 06:27:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20152ef9c3473ac98c20768a9abbce9b602f8c38","revisions":{"20152ef9c3473ac98c20768a9abbce9b602f8c38":{"kind":"REWORK","_number":1,"created":"2024-04-25 07:35:22.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/71/916971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/71/916971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/71/916971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/71/916971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/71/916971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/71/916971/1"}}},"commit":{"parents":[{"commit":"395e6fa2905cdf819937edfd4008d51e5826d246","subject":"Add Active Active L3 GW API test cases","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/395e6fa2905cdf819937edfd4008d51e5826d246"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2024-04-25 07:34:46.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2024-04-25 07:35:15.000000000","tz":480},"subject":"add concurrent list_firewall_groups","message":"add concurrent list_firewall_groups\n\nChange-Id: Ia7e740fd955a405ac0c5a7e38cc0849ce92433f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/20152ef9c3473ac98c20768a9abbce9b602f8c38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/20152ef9c3473ac98c20768a9abbce9b602f8c38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
