)]}'
{"id":"openstack%2Fneutron-tempest-plugin~920728","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I841dee939ebd1ba1adea13f7a216bcb1c945d44d","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-29 09:57:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I841dee939ebd1ba1adea13f7a216bcb1c945d44d","subject":"Remove devstack-gate","status":"ABANDONED","created":"2024-05-29 09:54:27.000000000","updated":"2024-05-29 09:57:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2c877e8fc06ccaeef5a884ddbc7eaad23287eb24","_number":920728,"virtual_id_number":920728,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-29 09:54:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-29 09:54:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1270b03f1b0790d4f75fc2661a2ac77e16addccf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-29 09:54:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e0250efe86efdf0f4d5df16181bf630ec5ea9cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 09:54:48.000000000","message":"Patch Set 1: Verified-1\n\n(15 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/neutron-fwaas on branch master.  The\nproblem was:\n\n  Job neutron-fwaas-v2-dsvm-tempest-multinode not defined\n\nThe problem appears in the the \"opendev.org/openstack/neutron-fwaas\" project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - openstack-cover-jobs-neutron\n        - openstack-python3-jobs-neutron\n        - periodic-stable-jobs-neutron\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n  ...\n\n  in \"openstack/neutron-fwaas/.zuul.yaml@master\", line 13\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/neutron-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the \"neutron-tempest-plugin-base\" job stanza:\n\n  job:\n      name: neutron-tempest-plugin-base\n      parent: devstack-tempest\n      abstract: true\n      description: |\n        Perform setup common to all Neutron tempest tests\n      roles:\n        - zuul: openstack/devstack\n      required-projects:\n  ...\n\n  in \"openstack/neutron-tempest-plugin/zuul.d/master_jobs.yaml@master\", line 2","accounts_in_message":[],"_revision_number":1},{"id":"2c877e8fc06ccaeef5a884ddbc7eaad23287eb24","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-29 09:57:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b25b07681bb97b5854cfc02723c244a25096eaef","revisions":{"b25b07681bb97b5854cfc02723c244a25096eaef":{"kind":"REWORK","_number":1,"created":"2024-05-29 09:54:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/920728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/28/920728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/28/920728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/28/920728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/28/920728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/28/920728/1"}}},"commit":{"parents":[{"commit":"17e95518ead3e0787e1bfc77764252c4307109bd","subject":"Remove executable from python files which don\u0027t really needs it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/17e95518ead3e0787e1bfc77764252c4307109bd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-29 09:54:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-29 09:54:13.000000000","tz":540},"subject":"Remove devstack-gate","message":"Remove devstack-gate\n\ndevstack-gate was deprecated in xena and was completely retired[1].\n\n[1] https://review.opendev.org/c/openstack/governance/+/919629\n\nChange-Id: I841dee939ebd1ba1adea13f7a216bcb1c945d44d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b25b07681bb97b5854cfc02723c244a25096eaef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/b25b07681bb97b5854cfc02723c244a25096eaef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
