)]}'
{"id":"openstack%2Fneutron-tempest-plugin~922706","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ib4baa25934c95838824629177f484fa1f4ad4444","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2051831","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-08-04 16:46:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib4baa25934c95838824629177f484fa1f4ad4444","subject":"WIP \u003d\u003d Add \"subnet-external-network\" extension to the master jobs","status":"ABANDONED","created":"2024-06-25 07:32:30.000000000","updated":"2024-08-04 16:46:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7dea11737e5feac73083c43ebaf1b302d38e3fa9","_number":922706,"virtual_id_number":922706,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-06-25 10:28:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-25 10:28:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-04 16:46:45.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"bebafcee809bd7ab727af13c2311dcfa8555058b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-25 07:32:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e964d7606ca618d22fa65c1383b0473c121a9605","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 10:28:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c59e1892c3ca4bfbab6d5afb92861086\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d54840194e64956a8a4f1206585b88a : SUCCESS in 3m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/821c2fae67bc493798c3434ec1a4474e : FAILURE in 1h 36m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8c1ce31492944b368f7d8b51ec7a38c6 : FAILURE in 1h 44m 35s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/9fbdbc6d478f4c08bde6914127ecbbea : FAILURE in 1h 37m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/68f432f4209c46c6bc2f5a336e0bf7d5 : SUCCESS in 2h 10m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/15d991cfbcec49b89ece9f5b397e24b5 : SUCCESS in 32m 52s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/102b58b6d3c74c72aaa039e8a966eb16 : SUCCESS in 1h 33m 42s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/5ffc6a51042b4b30aa07f14b8808ca3a : SUCCESS in 1h 35m 17s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/62aafddff3b048f69ac0aa672c3b7867 : SUCCESS in 1h 42m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49489899deb44bf1836e36caf405adab : SUCCESS in 2m 59s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/cd3ac1b8bd9f40bb81772bd0cff1c69a : FAILURE in 2h 50m 23s (non-voting)\n- neutron-tempest-plugin-sfc-2023-1 https://zuul.opendev.org/t/openstack/build/e31054c61a2a4ba7ba4279b3e494ac2f : SUCCESS in 1h 25m 30s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/d5dbb96f9a83407d99063e503e195851 : SUCCESS in 1h 22m 50s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/3bfa1dede56a426db37ee9426000fb20 : SUCCESS in 1h 32m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/a8921650bf254a08801d0961bc4287f9 : SUCCESS in 1h 26m 18s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/f8100acf976741c7a951713d26b00365 : SUCCESS in 1h 36m 59s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/a081752bd04540198b7702e08c99806f : SUCCESS in 1h 30m 39s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/deb11574232a4f89bac6c4d66594f89c : SUCCESS in 1h 11m 48s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/9b4cbc38825a4d61a9e6286a1db2e85f : SUCCESS in 46m 11s\n- neutron-tempest-plugin-dynamic-routing-2023-1 https://zuul.opendev.org/t/openstack/build/cb5424b144db4967b432a8378296d659 : SUCCESS in 40m 31s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/524af1001a5345bab8ae9001d142c717 : SUCCESS in 57m 17s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/56c1dd5c90b1428e8606c1034158bc6b : SUCCESS in 57m 39s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/75b0ace7957549eb9e99696d3033747c : SUCCESS in 34m 37s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/f3b1b5f80b14418998f384fc6b436398 : SUCCESS in 45m 15s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/fdf6256a1ead4d4190337e44fa1ab2ac : SUCCESS in 44m 24s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/2684d0e5c6074ce9b070b5862d34133b : SUCCESS in 47m 39s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/58a87e9ef6314e8aa73c6d623719eae4 : SUCCESS in 49m 07s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/8895c0ca68814737914961c4e6830bf6 : SUCCESS in 40m 52s\n- neutron-tempest-plugin-vpnaas-2023-1 https://zuul.opendev.org/t/openstack/build/81e3651ff5804df2b7b71ffd7b4a691f : SUCCESS in 39m 50s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/de1cafd7c20e44ae9072833cf758224b : SUCCESS in 45m 33s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/9dd54aa27a504f8ab5aee589263a98c2 : SUCCESS in 53m 50s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/782867fbd831493fbe375ce39143b27e : SUCCESS in 1h 12m 06s\n- neutron-tempest-plugin-tap-as-a-service-2023-1 https://zuul.opendev.org/t/openstack/build/da1a8d28563f41beae49b73ea2e7dc54 : SUCCESS in 1h 16m 34s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/ca72d51af5fc4decbceca21d404c159c : SUCCESS in 1h 21m 48s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/ee69b8699db6454a900db65c5e34348e : SUCCESS in 1h 23m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7dea11737e5feac73083c43ebaf1b302d38e3fa9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-04 16:46:45.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90194cdcdf7541e911493936ff890057f844dbff","revisions":{"90194cdcdf7541e911493936ff890057f844dbff":{"kind":"REWORK","_number":1,"created":"2024-06-25 07:32:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/922706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/06/922706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/06/922706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/06/922706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/06/922706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/06/922706/1"}}},"commit":{"parents":[{"commit":"9aec7b695ef165b7a427a0b25e378b2247a1342f","subject":"Add test for the validate auto_allocated_topology API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9aec7b695ef165b7a427a0b25e378b2247a1342f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-06-24 16:42:34.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-06-24 16:42:34.000000000","tz":0},"subject":"WIP \u003d\u003d Add \"subnet-external-network\" extension to the master jobs","message":"WIP \u003d\u003d Add \"subnet-external-network\" extension to the master jobs\n\nRelated-Bug: #2051831\nChange-Id: Ib4baa25934c95838824629177f484fa1f4ad4444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/90194cdcdf7541e911493936ff890057f844dbff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/90194cdcdf7541e911493936ff890057f844dbff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
