)]}'
{"id":"openstack%2Fneutron-tempest-plugin~924124","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I9f3a72da13212f735d51c7929b05706ba063f313","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2045549","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-07-17 06:12:05.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-07-17 01:41:08.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-07-16 15:34:21.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I9f3a72da13212f735d51c7929b05706ba063f313","subject":"Temporary switch lb/ovs/ovn scenario jobs to cirros-0.5.3","status":"MERGED","created":"2024-07-15 06:58:47.000000000","updated":"2024-07-17 06:12:05.000000000","submitted":"2024-07-17 06:12:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924124-bug/2045549","meta_rev_id":"868fba430aaa769d50bfc777cd0f8810740d8270","_number":924124,"virtual_id_number":924124,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-17 06:12:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-07-16 18:21:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2024-07-16 15:28:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2024-07-16 15:34:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-07-16 18:21:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-07-16 14:46:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-15 10:02:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-15 10:12:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-07-15 11:50:40.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-07-16 15:28:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-07-16 18:21:50.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"825cf5b5737d94ecb8eb6ab0c4f739ea0b788ccf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-07-15 06:58:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4eb297b8a53fa6c87ee4828f04d0bef97537c03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 10:02:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1131b9485654ae39a5ec97ffb8261ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffca397408154224bfab0803636491b6 : SUCCESS in 4m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8d16b15633104d31a324d7e113413795 : SUCCESS in 1h 58m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3f47627a14224389b48f950afb1df3b3 : SUCCESS in 1h 56m 00s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/d956422d51644079be0d77e0febec742 : SUCCESS in 1h 56m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b5e2e31015ae457b8dc6575e3f8a7b8c : SUCCESS in 2h 37m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3eade8baead047128771efd5362a8b0e : SUCCESS in 32m 33s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/0e3ac2122298474f916cd1f2b0105f90 : SUCCESS in 1h 22m 49s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/c4a1b333881f449cb312a072de231bff : SUCCESS in 1h 15m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/e23ae81465da4fe6878fd75136b1bb90 : SUCCESS in 1h 51m 48s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/1660c143464441caaa123e77fdb2278f : SUCCESS in 1h 32m 52s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/d22df2fab7d348759963b6fe530ffae7 : SUCCESS in 37m 06s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/74b8e64d156d4db48ae6f5efcb231741 : SUCCESS in 1h 11m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/c76b7e6201ed4038a0a25c4c42a5009c : SUCCESS in 1h 23m 07s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/10ba52412112435a9b1ad7d9c5dc3c9f : SUCCESS in 1h 47m 18s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/d9db5972d73f4a128a63a8380964945b : SUCCESS in 31m 21s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/10920c0bfca342c4816a200bf2588992 : SUCCESS in 1h 10m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/a171cfb8b4894a82999cd4840dae5aa1 : SUCCESS in 1h 39m 50s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/04b5c732839047d0b2ce0c5fc90c0e33 : SUCCESS in 1h 45m 48s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/049683b712404f4e9cd50872792457ee : SUCCESS in 31m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e342700cc1c24aef9fa3ad9a9492ad63 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"9faec2f33af86d8842a1cdc21add62d36000b2b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-07-15 10:12:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"305a0afb7c9d164da597f4a7bb468491d420c931","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-15 10:12:50.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"884878b036ed1d2b57e22b6cd87a339731ecd57b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-15 11:50:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d39f135cb67ef6ca3399a0ed5e2a69d6bf2d7c85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 13:05:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13dd1781640a418aa732319114be16bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efa366b47b0d4989b3993688314c7759 : SUCCESS in 3m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b2d737bb03ac462d9dc1ea6a6b82aa88 : SUCCESS in 1h 25m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a532a74d2efe4987970018cd42437bcf : SUCCESS in 1h 28m 25s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4be3649a05db477d8e8f58b54a0dd40f : SUCCESS in 1h 27m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c6e826cb53de49d491651763118bdf76 : FAILURE in 2h 09m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d55c5586b78c4707aa2296a2e55ddaaf : SUCCESS in 31m 36s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/9403c8a8ca964afd8ab1a30e4886009d : SUCCESS in 1h 13m 18s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/fa2d1243750e4c878873f9a2e75aeeff : SUCCESS in 1h 13m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/1463a75b4d3443bc9ab4fbf35cb59c1d : SUCCESS in 1h 49m 57s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/d60d97b0978c478fa5029e14a75b3ded : SUCCESS in 1h 53m 14s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/2efa7d92cc984588b5bab04df296f0b4 : SUCCESS in 47m 37s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/33dfce770ea54696ae5b4a038bc650f9 : SUCCESS in 1h 10m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/f229c7d4f3134e04be3eadc988d54719 : SUCCESS in 1h 23m 36s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/84f3640450374563b85499535478c32a : SUCCESS in 1h 57m 47s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/fb9813249ed8429e8ef38165ea59c06f : SUCCESS in 34m 32s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/fe21b7499087452ca113c94df88d1b98 : SUCCESS in 1h 37m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/6f6ac6bc2fbb407d9b2245396ddd1837 : SUCCESS in 1h 28m 44s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/00c759f4c2814b41981fee8e8fdfdda3 : SUCCESS in 1h 35m 36s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/22be6fb8ac71456da03c893ce9495f30 : SUCCESS in 30m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f4e040b40524986b3b75aead0fc72cc : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":2},{"id":"7d41d42fd00c27571ada8464edc273cb286dd672","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-07-15 13:34:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ecb741ec768bc026e245d91f622da90ad255bee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 17:21:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59f4c468142345bda9c3de36f4dd2cde\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e6ea3e292504df0803beeae684188a3 : SUCCESS in 5m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/94e864c3e3c44738be51378b5bd67379 : SUCCESS in 1h 49m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/21e00a38f13a4ad383540730d3aab159 : SUCCESS in 2h 17m 49s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/ab7791e9f9a84f869a831e5de6d37b95 : SUCCESS in 1h 36m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a34c8fd9a3c24d27bf651f84f8f2002f : SUCCESS in 1h 31m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6b968b5f27684bd8836a83d43a5ac358 : SUCCESS in 35m 11s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/e532c5e33043474fb8afb88e6188b2fc : SUCCESS in 1h 43m 11s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/8ed81f4f68d141c1aacd1ab1b40ca39b : SUCCESS in 1h 05m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/877da48d26da4e8dbd9fcdfb5a6942b7 : SUCCESS in 1h 51m 41s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/91953362855b4f9ab82b6ebb1fcbccdc : SUCCESS in 1h 38m 54s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/3edf158432654ff697993d308ef37d7c : SUCCESS in 30m 03s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/8f1d8132560646f9bd8ecbd3b4fe867b : SUCCESS in 1h 46m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/30a90206b43a4c86852d86475a7da4d3 : SUCCESS in 1h 29m 55s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/7f8d3ecd09dc4ef58153b81d46178e48 : SUCCESS in 1h 33m 05s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/e2de43f57c714c73bb765cd844b80160 : SUCCESS in 33m 22s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/e2245d1e2a8c4a809f406965ed4b8641 : SUCCESS in 1h 31m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/cbab13748a6545008847d02472f96cd8 : SUCCESS in 1h 21m 20s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/1d260ac35ff84d598c64d2c15a5a64ab : SUCCESS in 2h 09m 40s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/6ff0370f0799442a9453eec456e370b5 : SUCCESS in 31m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/296c04654a1c4329a66aa571f51af0cb : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":2},{"id":"669daecf4696c6a1b8d50276754286fa068ee9d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 17:21:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0b8cddbaf1d89ef3443e7e8c7edfc32f04750052","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 21:13:20.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3185dfb775974b28a9e2542574abc52b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64174e9e21f84ed58cb32c3214526b61 : SUCCESS in 3m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/21e2f4ffdb784287a57ae29ce9139889 : SUCCESS in 1h 38m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/09266fb0742c4b8ea84c8741ae68fd13 : SUCCESS in 1h 29m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8d455e3b428b4e92a909c43e2d065f53 : SUCCESS in 1h 31m 50s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/cc4b44f0b09746e6b43f159b9f24c686 : FAILURE in 1h 59m 38s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/05557dd71dea4d4aacee656b07de2421 : SUCCESS in 1h 34m 55s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/fd53b009c8654e89b759df86d19d626d : SUCCESS in 1h 49m 45s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/7a3cadb057eb42ec83c120d84b77704f : SUCCESS in 1h 37m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/371e816cc5a9410e8ae3c2a579852a20 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":2},{"id":"a068af56b9b440203d8fbaa2e023e0f9b6f547f1","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-07-16 04:58:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2e4e6e651f3f75926ad4960d5afab76bc81b51b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 07:21:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df7739ca394749af90ada333ade0eeb9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c4af8a680324fc2ab637a97acc5b5b2 : SUCCESS in 3m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ee9aa44d7cda418ca5c313dc76528c82 : SUCCESS in 1h 28m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e2ece0e1f32b4d3fa5dec196a8ae93f2 : SUCCESS in 1h 51m 46s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/791bb2725c9f4aa69f537f033b3c3ba2 : FAILURE in 1h 29m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dbff30bcaa4a473098642ab1b2b71079 : SUCCESS in 1h 45m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e0fa8664b3a94af2ad278edd586d9897 : SUCCESS in 31m 21s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/5fc1ed4ffebd46169594a495d91b89a1 : SUCCESS in 1h 44m 27s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/d10673188e5041ddb9284c7be4704546 : SUCCESS in 1h 13m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/ca16f847bc9d42f0b5f3b98aa530b250 : SUCCESS in 1h 25m 12s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/6685b9832fb04a1d9deb2f5580798fa4 : SUCCESS in 1h 28m 05s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/6284f1c6726b4719986ee422a2387759 : SUCCESS in 31m 10s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/b86d22aa60124cbcb368c7d72c66180d : SUCCESS in 1h 06m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/2f87f66bf12947459071721877227b68 : SUCCESS in 1h 17m 38s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/e9e659a0c176414c9b770ef2eac74e8e : SUCCESS in 1h 33m 22s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/62ba802e71b34fccb715e02c7d80b961 : SUCCESS in 31m 58s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/93ec85bef9704aefa4409987b65e653c : SUCCESS in 1h 42m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/856ab71c117f46fb950a9eb1e013c3c4 : SUCCESS in 2h 20m 37s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/776910d9247e4572900125b29d2cba57 : SUCCESS in 2h 04m 05s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/73909951de8f4adaaa2f7e4f39426ac0 : SUCCESS in 42m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05fc158ab01540649a7c212085cd9274 : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":2},{"id":"f81d5fd0cc1e2a99f5e92ce1903bee091303a947","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-07-16 10:01:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1994650aaec95aa531f22d0bd5beecd70449a30c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-07-16 11:25:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e234175516c50379472cea799828bb454549ce9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 12:17:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ffbc4de809a4b5990a678a680af6fc5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb5bd12ff236419db37382d28e0dd3ab : SUCCESS in 3m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/43fcfb2cf510496b8c3ec35994338ec6 : SUCCESS in 2h 06m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/95e7e440b0be44b291d9cec989305d77 : SUCCESS in 2h 04m 16s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/cbabcc3a60bb4f51a6562794d762c574 : SUCCESS in 1h 17m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4a4f904e8c9f4ef19fd33c7236fb3f50 : SUCCESS in 2h 10m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a5be2930edb1416a8e80c4da74af8877 : SUCCESS in 32m 05s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/e5230d84dbb64f0db667598882c2a53f : SUCCESS in 1h 44m 38s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/daedacda944e4368b1677c2c4bc244b4 : SUCCESS in 1h 32m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/4a16aa563be34606b01939e871c51189 : SUCCESS in 1h 50m 17s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/f24858f9e8fa4f49b9a35e6e69e89627 : SUCCESS in 2h 04m 17s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/fb1408fb31bc48faa13de4c432629a93 : SUCCESS in 34m 49s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/f648ad09b84d42209e41530cac41da0e : SUCCESS in 1h 26m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/3ce008eefbc04f1c8406bfcec6a7aefa : SUCCESS in 1h 44m 14s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/e96c8fccdaa143019cd377e4eb527988 : SUCCESS in 2h 13m 14s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/68814ddde7c54f5999cc037f4d24f235 : SUCCESS in 45m 18s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/1ed36f498159404090365f2b9629ea38 : SUCCESS in 1h 15m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/facf487e957a424897ba6d014c1e2b69 : SUCCESS in 1h 32m 04s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/c369faecb0a3456fb0424c9b2d94a37b : SUCCESS in 1h 31m 33s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/d7ed11792ea74d86b53f555277619348 : SUCCESS in 40m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c497cc211184240b43946f8eec26c49 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":2},{"id":"385e906e8a72242ce739bb94e1c876161f12eeac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 12:17:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"45b6fab112e33eb2b225f37775f356b51a7c8c8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 14:22:13.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f687f1b00d04f9b934eb24481339372\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be6b172112f8471996364421fec2bd47 : SUCCESS in 3m 26s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/208a5c83c2db4f18b93a2f1a4c0d91c8 : FAILURE in 1h 39m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/414151b74d614beaa4422aef74dfdc69 : SUCCESS in 1h 52m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3de049afb3ca41e2aafcda30910889db : SUCCESS in 1h 32m 20s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/bc6359a5b0c1446cb02db5383646db1c : SUCCESS in 1h 32m 16s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/b4557f352f1848ea8d24fed1df52abde : SUCCESS in 1h 49m 19s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/15a57f0118734796963a01924c6e98ca : SUCCESS in 2h 03m 00s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/2fc4e7971fd942fd9446a269506306de : SUCCESS in 1h 51m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eb9502213944a9e9ed6a46678e4d2c4 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":2},{"id":"cdc47b34be175b5d007fc23c52a5e3af8d4e05d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-07-16 14:46:11.000000000","message":"Patch Set 3: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"cf3fc11719db881b7ab562ea655a32cf3f03a91e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-07-16 15:28:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3a2720034827527d2ca5c48b8bfb7fba6fa2957f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-16 15:34:21.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"de9cf8f1238c7f971d9291d3be249238cf61b355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 17:47:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cab1f99b667349368ae2a17110b42cbd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fd09cc200014e17ac3200468ce610eb : SUCCESS in 3m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dd0bc3801171418ea33777401bdea442 : SUCCESS in 1h 51m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/755813f916394a3c9f49596a35b508bf : SUCCESS in 2h 05m 53s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4459267d47234731a6bb1b1506afcbd1 : SUCCESS in 1h 45m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f87364895e254776af75755c8db47fbd : SUCCESS in 1h 55m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d59eae10fa814ab2b3e0cd6b7d2494dc : SUCCESS in 39m 09s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/29d4bcac56aa4d82a162c83c617bfd5f : SUCCESS in 2h 13m 24s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/8b4e461eeb774cd4a54a09120e11f20e : SUCCESS in 1h 31m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/861deff97b354a72a760cd5c87921fec : SUCCESS in 1h 56m 44s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/eb1b954455b840e3b26d815a8662ce7e : SUCCESS in 1h 56m 05s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/b268a78969224720b0b91bd3a825a45b : SUCCESS in 47m 41s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/37818c2330e74cce94f0b40a2c67722f : SUCCESS in 1h 10m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/c5d16aba742c4a64aee8c822daff82da : SUCCESS in 1h 33m 32s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/c23834d23dc142f9bc433611aa03213a : SUCCESS in 1h 38m 28s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/213a85c5ccd0417ba54e1a6b0e9d0970 : SUCCESS in 50m 55s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/78402357dce149b4ab7d067aa7f5b5c7 : SUCCESS in 1h 20m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/c20ee5e5475f4c7d9c5276d8568cafae : SUCCESS in 1h 55m 02s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/f501d15fb7664cc9b64c1c72696f7f26 : SUCCESS in 2h 56m 43s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/4189a33c8a7c4c758213b51f40a425fb : SUCCESS in 34m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2681625adea4ba8b67d4e00e311a2dd : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":3},{"id":"b1a7a51557770d8d253c460ed22550a55c517443","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-07-16 18:21:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5556a4de073f402abf31cdc5f495d31de2051656","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 22:27:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3a07e69f484a3b438e6e7953cb7b5a48a8693977","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 00:17:30.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/924213 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"d2658a5b794830b5f200b088acaac8b3a3101403","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-07-17 01:41:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2d3a4e56ab2b4aad2e706fc608e889c6beb854d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 04:08:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73754cbaed764c719d342e359d1d5456\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93833b9c034e435ab600c1f3131b3505 : SUCCESS in 3m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8db79cff60c140b397099b2fb27a6038 : SUCCESS in 1h 27m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d65cbc0173f74711b11e15ddc6886b17 : SUCCESS in 1h 56m 46s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/5040e483170e453584a75774a085a4f9 : SUCCESS in 1h 24m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8e6f4efe611c416e891ea955b47b79fc : SUCCESS in 1h 52m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5bf90721c2564023a9e0d09e648df71e : SUCCESS in 37m 56s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/55f7305a2b67426b9f229718fce74213 : SUCCESS in 2h 16m 03s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/6df6c54ac5aa446eba73ea9c808f9342 : SUCCESS in 1h 29m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/8ea66dbd102140e19e0ed4c65a63047e : SUCCESS in 1h 59m 14s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/49c11be5892c4af8b175e5d37d865a85 : SUCCESS in 1h 36m 00s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/7d5e5a0c84884bdabfcc517b8e1aa439 : SUCCESS in 43m 58s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/126913ca12f0406793bdc53172d162b7 : SUCCESS in 1h 20m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/0364a477f02641d4b6a265418210600d : SUCCESS in 1h 34m 35s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/93f2c2b42b44411d941ba51a6b214822 : SUCCESS in 1h 54m 13s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/e89fddeada20489c9667a2163d484e35 : SUCCESS in 33m 51s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/6e117c411fad4cf99ffa3ca8276a0314 : SUCCESS in 1h 17m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/fe78ff1448dc465688471057ffb4793a : SUCCESS in 1h 23m 04s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/5a38076640e443feaf0bea9004e6a077 : SUCCESS in 1h 32m 06s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/ac177233bb7949008d4346e589ea2fb4 : SUCCESS in 33m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc0af6b6928d4debbc9236b2908e0ee6 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":3},{"id":"36216725dbc582cb02a334a693e5d0e58ad0ebf6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 04:08:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bfc11a95505117c6a3211db65ef25db3db58111d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 06:12:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8fb77a7c9d04547a7328826ba089369\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c681a29a3f044819745c96dd4fb69a6 : SUCCESS in 3m 42s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f3082be2c4c9420a97a2343eb7602c02 : SUCCESS in 1h 47m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/183b76001ae44a3997e5f1be3da0b69f : SUCCESS in 1h 56m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/78c605503e47497f9e5251f039e3b6ed : SUCCESS in 1h 40m 45s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/1e656a31e5af43b29ebc26bb2811c663 : SUCCESS in 1h 37m 04s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/89a25b89100d4916a1c9179b80d5eb69 : SUCCESS in 1h 42m 48s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/ec9f0ce88f4f42609cab4a73c70dc16b : SUCCESS in 2h 00m 13s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/e2f37d0721e64ab7be9fb094ce9c0706 : SUCCESS in 1h 38m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c37ec1afce349adb846a3acf8fd1b59 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":3},{"id":"868fba430aaa769d50bfc777cd0f8810740d8270","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 06:12:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a7a7ce8d342c88fc9dcd8d57607daa446a0b98e1","revisions":{"aa1b78ffe1839c24689c7e8ad061cfb92098f023":{"kind":"REWORK","_number":1,"created":"2024-07-15 06:58:47.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/24/924124/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/24/924124/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/24/924124/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/24/924124/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/24/924124/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/24/924124/1"}}},"commit":{"parents":[{"commit":"5c5dc014bf5fb31c2c7dbbf716d917630e55e48f","subject":"Make neutron-tempest-plugin-designate-scenario non-voting temporary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5c5dc014bf5fb31c2c7dbbf716d917630e55e48f"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-07-15 06:54:46.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-07-15 06:58:26.000000000","tz":330},"subject":"Temporary switch lb/ovs/ovn scenario jobs to cirros-0.5.3","message":"Temporary switch lb/ovs/ovn scenario jobs to cirros-0.5.3\n\nWith cirros-0.6+ we hitting random issues as mentioned\nin the related bug, until the cirros fix[1] merge and have\nnew cirros release let\u0027s switch to older cirros image.\n\n[1] https://github.com/cirros-dev/cirros/pull/116\nRelated-Bug: #2045549\n\nChange-Id: I9f3a72da13212f735d51c7929b05706ba063f313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/aa1b78ffe1839c24689c7e8ad061cfb92098f023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/aa1b78ffe1839c24689c7e8ad061cfb92098f023"}]},"branch":"refs/heads/master"},"8d6b19e623e067c516d44735bebf5e71a4bd1cc6":{"kind":"REWORK","_number":2,"created":"2024-07-15 10:12:29.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/24/924124/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/24/924124/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/24/924124/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/24/924124/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/24/924124/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/24/924124/2"}}},"commit":{"parents":[{"commit":"5c5dc014bf5fb31c2c7dbbf716d917630e55e48f","subject":"Make neutron-tempest-plugin-designate-scenario non-voting temporary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5c5dc014bf5fb31c2c7dbbf716d917630e55e48f"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-07-15 06:54:46.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-07-15 10:12:09.000000000","tz":330},"subject":"Temporary switch lb/ovs/ovn scenario jobs to cirros-0.5.3","message":"Temporary switch lb/ovs/ovn scenario jobs to cirros-0.5.3\n\nWith cirros-0.6+ we hitting random issues as mentioned\nin the related bug, until the cirros fix[1] merge and have\nnew cirros release let\u0027s switch to older cirros image.\n\n[1] https://github.com/cirros-dev/cirros/pull/116\nRelated-Bug: #2045549\n\nChange-Id: I9f3a72da13212f735d51c7929b05706ba063f313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8d6b19e623e067c516d44735bebf5e71a4bd1cc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8d6b19e623e067c516d44735bebf5e71a4bd1cc6"}]},"branch":"refs/heads/master"},"a7a7ce8d342c88fc9dcd8d57607daa446a0b98e1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-07-16 14:46:11.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/24/924124/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/24/924124/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/24/924124/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/24/924124/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/24/924124/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/24/924124/3"}}},"commit":{"parents":[{"commit":"5c5dc014bf5fb31c2c7dbbf716d917630e55e48f","subject":"Make neutron-tempest-plugin-designate-scenario non-voting temporary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5c5dc014bf5fb31c2c7dbbf716d917630e55e48f"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-07-15 06:54:46.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2024-07-16 14:46:11.000000000","tz":0},"subject":"Temporary switch lb/ovs/ovn scenario jobs to cirros-0.5.3","message":"Temporary switch lb/ovs/ovn scenario jobs to cirros-0.5.3\n\nWith cirros-0.6+ we hitting random issues as mentioned\nin the related bug, until the cirros fix[1] merge and have\nnew cirros release let\u0027s switch to older cirros image.\n\n[1] https://github.com/cirros-dev/cirros/pull/116\nRelated-Bug: #2045549\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/924213\nChange-Id: I9f3a72da13212f735d51c7929b05706ba063f313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a7a7ce8d342c88fc9dcd8d57607daa446a0b98e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a7a7ce8d342c88fc9dcd8d57607daa446a0b98e1"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
