)]}'
{"id":"openstack%2Fneutron-tempest-plugin~926469","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I40d9f26a3d76d872de93d86ded5d40ccbd696f2a","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2067239","attention_set":{},"removed_from_attention_set":{"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2025-09-23 15:32:14.000000000","reason":"removed on reply"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-12-04 13:56:04.000000000","reason":"Change was abandoned"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-10-04 08:02:37.000000000","reason":"Change was abandoned"},"36716":{"account":{"_account_id":36716,"name":"Kyuyeong Lee","display_name":"Kyuyeong Lee","email":"kyu0.lee@samsung.com","username":"kyu0"},"last_update":"2025-12-04 13:56:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I40d9f26a3d76d872de93d86ded5d40ccbd696f2a","subject":"Add test \"test_sg_rule_quota_with_default_sg_rules\"","status":"ABANDONED","created":"2024-08-16 12:21:34.000000000","updated":"2025-12-04 13:56:04.000000000","total_comment_count":10,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"8153bbd540298f79ded3974cc7c9813dbf929b78","_number":926469,"virtual_id_number":926469,"owner":{"_account_id":36716,"name":"Kyuyeong Lee","display_name":"Kyuyeong Lee","email":"kyu0.lee@samsung.com","username":"kyu0"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-23 13:36:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2025-09-23 15:32:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":-1,"date":"2025-09-23 21:41:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-16 13:18:05.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-08-16 14:15:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-04 08:02:37.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2024-10-31 16:10:49.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"c7e8e16edb4890dc4751ba171421fdd046dea733","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36716,"name":"Kyuyeong Lee","display_name":"Kyuyeong Lee","email":"kyu0.lee@samsung.com","username":"kyu0"},"date":"2024-08-16 12:21:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62617f26741881fd794c19b8e3cf8ee8f8dabfbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 14:15:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd16909426994d37bf240c9dd9fc1f2c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a72dc120a5e54583b4b8b02d473a99c9 : SUCCESS in 3m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0d7a162f3f494ec69b1f369c300d81d5 : SUCCESS in 1h 26m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/26463579096840e98abfdb9dbe06dd81 : SUCCESS in 1h 44m 22s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/0388532a5ec9498d9c1a408172e7cd5a : SUCCESS in 1h 19m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f2126dcb0c0844baafc71d949782f224 : SUCCESS in 1h 43m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5e5182a643194219bd70d096958676ae : SUCCESS in 41m 00s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/ea056b852e544fc1ab352fe95be42b2a : FAILURE in 1h 20m 09s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/4fffe4a450bf4edfaa7241eef5b0b7aa : FAILURE in 1h 10m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/8bee0a85b97f4e03872c15529479bb79 : FAILURE in 1h 46m 32s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/640d6c7349124ebd8756d88fa8f9cd19 : FAILURE in 1h 31m 52s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/43dfcbf81e4f42d39af08dc0ce79f414 : SUCCESS in 32m 00s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/6cb2ddcc4d1c497d86729bc05cccfe83 : FAILURE in 1h 31m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/4b250d237f8149f2bd0cd1a422c7fff2 : FAILURE in 1h 37m 36s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/e675618c35e64f85b8b20f7b269f0431 : FAILURE in 1h 34m 58s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/d429b6d951164e12babc49cb3fe5f3c7 : SUCCESS in 34m 52s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/4289ff3fcd784390a0323523579cdd98 : FAILURE in 1h 32m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/f7f10b28b21149d890ae22b4b2df56a1 : FAILURE in 1h 48m 03s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/3f51be0dbf864742a628dfa3ab22061f : FAILURE in 1h 52m 18s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/f21f11dec0944124acd0d3f4b97ffe69 : SUCCESS in 39m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67a93ea7658141b08a33363b7ada7424 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"3e4b6b1a883fd348e6627077df0422da0e34ffaa","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-17 21:38:00.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"028ffe8bc9956c145702b25487153c464aefc0dd","author":{"_account_id":36716,"name":"Kyuyeong Lee","display_name":"Kyuyeong Lee","email":"kyu0.lee@samsung.com","username":"kyu0"},"date":"2024-08-22 01:29:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e66debb9c0d5576a0f78ead0295d86821dc2a16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36716,"name":"Kyuyeong Lee","display_name":"Kyuyeong Lee","email":"kyu0.lee@samsung.com","username":"kyu0"},"date":"2024-08-22 15:23:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6b491cc004f97e5cab1ab34f5bb3590a1faad7a9","author":{"_account_id":36716,"name":"Kyuyeong Lee","display_name":"Kyuyeong Lee","email":"kyu0.lee@samsung.com","username":"kyu0"},"date":"2024-08-22 15:24:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b10e83b1124f9b220a3ec52ffab1a322ed19ea5f","author":{"_account_id":36716,"name":"Kyuyeong Lee","display_name":"Kyuyeong Lee","email":"kyu0.lee@samsung.com","username":"kyu0"},"date":"2024-08-22 15:25:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1ab744d8f93474ca77d559b3e78cf0bfc4512e08","author":{"_account_id":36716,"name":"Kyuyeong Lee","display_name":"Kyuyeong Lee","email":"kyu0.lee@samsung.com","username":"kyu0"},"date":"2024-08-22 15:31:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bac387f7c854df49a151004a5941b49d4907b2b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 17:43:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f9976223f8043ed9a1f77ccc609295a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cba0b8fa69624e238bca903ec48396f0 : SUCCESS in 4m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/78c41b1779f5405e816b2b0ba62feaef : SUCCESS in 1h 26m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5b972ea833054b3b86f34b30361a4d12 : SUCCESS in 1h 58m 09s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4c3e351cd48f403d9b511e4b032a9305 : SUCCESS in 1h 24m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/df302605f5ea4215b05540ebaa59f8b4 : SUCCESS in 2h 14m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3224c42b659a462a9479686754290ddc : SUCCESS in 32m 08s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/818aea802f934e419749cf9528d6d32c : FAILURE in 1h 54m 30s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/3b07e93aa3784a15bc94ea843fd59dff : FAILURE in 1h 36m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/4026d0b9566e4bbebbaa458a813dd2fc : FAILURE in 1h 47m 33s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/7db2577908be48ea93ea033a2c243e68 : FAILURE in 1h 59m 55s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/dc3dba17d6554023bc59b4c260656e65 : SUCCESS in 41m 10s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/efe7baab0a554cf6820aeedb447ae729 : FAILURE in 1h 38m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/48572fc04d16444baac3564eed44ebf5 : FAILURE in 1h 39m 47s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/091a6cd09fa947d3a3cf9bc8b9999e67 : FAILURE in 2h 15m 28s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/24d97173f55a4c4e843b51a486172775 : SUCCESS in 42m 51s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/a0e1db5cbe46419b8092e0ffc1e46297 : FAILURE in 1h 48m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/a2c865fbdf83499f8356b85ceff1aae3 : FAILURE in 1h 44m 20s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/a140cbcfcda24d179bc854628a0ada1b : FAILURE in 1h 39m 29s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/db553f9fbf2047289e3e19692b86934f : SUCCESS in 31m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efa9b4579b4d43e59926e2f71227b316 : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":2},{"id":"e5ff78bbe8f99dd032e86bcf1952b56f207a1be4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-10-04 08:02:37.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2},{"id":"3c073fa6c21bf1c61bd05ba67c2f79387c4fa778","tag":"autogenerated:gerrit:restore","author":{"_account_id":36716,"name":"Kyuyeong Lee","display_name":"Kyuyeong Lee","email":"kyu0.lee@samsung.com","username":"kyu0"},"date":"2024-10-31 10:27:20.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"bcff4e97b5e17c4406dff348f624813ca5989ba3","author":{"_account_id":36716,"name":"Kyuyeong Lee","display_name":"Kyuyeong Lee","email":"kyu0.lee@samsung.com","username":"kyu0"},"date":"2024-10-31 10:43:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a12c438fa79be845db0cfa9a3384e7a848910d42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 12:18:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a5ef9a348504723b7ca62a082c41408\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab35b332312547acb675a1923e6dbeb0 : SUCCESS in 3m 31s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c5fd3b934e8845ddbb3e2b5c98141996 : SUCCESS in 1h 15m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/486cae8591934756bc3259db1de7a44b : SUCCESS in 1h 25m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6a899a8b66214337a3d706f786057de8 : SUCCESS in 1h 43m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2952f3a4157148d199b339559744b647 : SUCCESS in 28m 53s (non-voting)\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/29071b506d1a443a8d1be88ecbaec50a : SUCCESS in 1h 20m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/9ba8d77e77424aa98baf6480fa44a6f8 : SUCCESS in 1h 17m 17s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/08634524d5f440989850f9f8adcae1cb : SUCCESS in 1h 32m 46s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/94ce9e7f377b426bbe4f4e8a6f69e5b6 : SUCCESS in 33m 11s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/cf8a962cc9cb4198a5157d35cf1fa3d8 : SUCCESS in 1h 17m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/ad8981d8c14f4c5c8b3f92b2afab7682 : SUCCESS in 1h 26m 39s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/366676d70ca44b5896409b7d8b2325dc : SUCCESS in 1h 49m 23s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/200c66440bae45d4bddcc1d4bd97d935 : SUCCESS in 30m 29s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/b8629155f1e443cdb5c7d4a5abd01297 : SUCCESS in 1h 33m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/db87bfa082ee4373ad500e789d9b0c38 : SUCCESS in 1h 39m 11s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/89e54eb82d5944cfb11659e3b443e26c : SUCCESS in 1h 26m 06s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/f0ec0b4d05764ebaba6c39bd3a29ea15 : SUCCESS in 30m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88cc468b7fb04f84bb8a60386c03c2eb : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":2},{"id":"eb5bb59feeee035388abded3d372604304ebcdfb","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-10-31 16:10:49.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/4cb8c53d7a2545ea944cd59e80a7e372\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/197c9f83300b400f83a67f91a46624d9 : SUCCESS in 5h 06m 11s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/c16b955e9de44b2d87bd849d2bbfc08e : SUCCESS in 3h 21m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eee3c6aaa93a4167da91df9f4558ddf4f9751dff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-09-23 11:16:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c33ba6820ea1913d2d502c4d02097998c3a02e02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 13:36:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b587f39d5e14016be202229622fb5a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18adafe5c3454321a66a7d6317a614d5 : SUCCESS in 3m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8757c406f3ec4c1b957e1b9714e1667b : FAILURE in 1h 16m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/67282d89e85d42ee9dc87f001e6a5034 : SUCCESS in 1h 52m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ad2cf7bef8d4485fae4f0e824518ba73 : SUCCESS in 36m 52s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/260e4f12f13c4a5a838c06d256b41c50 : SUCCESS in 1h 07m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/3483ac89801d4d209b86428b46d0ffb6 : SUCCESS in 1h 19m 07s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/d3fbbfb19947439385e4acaba5bda634 : SUCCESS in 1h 33m 35s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/e19e783b71ca42f0b37516292e41051c : SUCCESS in 28m 38s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/093ce295fd344c9bac2fac262ce36662 : SUCCESS in 2h 00m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/6e347ada3e3b4e1a9d97be407def5d72 : SUCCESS in 1h 47m 18s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/5809f870cdd7465281a61d5ecc5de9a8 : SUCCESS in 2h 17m 58s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/6766708f8afe41d281f16cabd6cd4426 : SUCCESS in 28m 49s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/3c71bfc79f2046f9816232b602e9d788 : SUCCESS in 1h 42m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/72cc10fb2acc4cef8b4fd597bdac3ada : SUCCESS in 1h 55m 21s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/ccab1fb45c8342de86380133bee7c551 : SUCCESS in 2h 19m 03s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/b5ba0c4ceeea4015ade0961356bab877 : SUCCESS in 42m 32s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/d114ccbcc66041fbaaddfad261f683db : SUCCESS in 1h 10m 18s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/ec1cc4cd92d5441baca1299218e7f26a : SUCCESS in 1h 22m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c672c68670f644cb95e4d8e1db7c7307 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":3},{"id":"720cecfb7f1cd44a5c9ef6585bfff48ffd2d5fe0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-09-23 15:32:14.000000000","message":"Patch Set 3: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/68771ea5ab2a4c26b872e66432838a68\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/50a429f565a740b78f1087e3217673ea : SUCCESS in 4h 14m 50s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/b191d5b6a6f34a078909b4795f518d65 : SUCCESS in 3h 21m 50s","accounts_in_message":[],"_revision_number":3},{"id":"0c300b28480d0ea467df8dcdd46cdbd09fc7f666","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-09-23 21:41:07.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8153bbd540298f79ded3974cc7c9813dbf929b78","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-12-04 13:56:04.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e938921e102ca79c2515f5e46c5f0c56c2d2110c","revisions":{"14f895fe3eb5cb37e068aea1cb1ef87a7163e3ba":{"kind":"REWORK","_number":1,"created":"2024-08-16 12:21:34.000000000","uploader":{"_account_id":36716,"name":"Kyuyeong Lee","display_name":"Kyuyeong Lee","email":"kyu0.lee@samsung.com","username":"kyu0"},"ref":"refs/changes/69/926469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/69/926469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/926469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/926469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/926469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/926469/1"}}},"commit":{"parents":[{"commit":"5225abc6049e0b4912d6635df0a142b8bfccebca","subject":"Fix check of the revision number in the resonse from API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5225abc6049e0b4912d6635df0a142b8bfccebca"}]}],"author":{"name":"kyu0","email":"kyu0.lee@samsung.com","date":"2024-08-16 12:21:34.000000000","tz":540},"committer":{"name":"kyu0","email":"kyu0.lee@samsung.com","date":"2024-08-16 12:21:34.000000000","tz":540},"subject":"Add test \"test_sg_rule_quota_with_default_sg_rules\"","message":"Add test \"test_sg_rule_quota_with_default_sg_rules\"\n\nThis test checks that the SG rule quota is working properly when\ncreating an SG.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/921909\nRelated-Bug: #2067239\nChange-Id: I40d9f26a3d76d872de93d86ded5d40ccbd696f2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/14f895fe3eb5cb37e068aea1cb1ef87a7163e3ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/14f895fe3eb5cb37e068aea1cb1ef87a7163e3ba"}]},"branch":"refs/heads/master"},"6cd2f054f9fde9522098c365828b38aa0dede061":{"kind":"REWORK","_number":2,"created":"2024-08-22 15:23:26.000000000","uploader":{"_account_id":36716,"name":"Kyuyeong Lee","display_name":"Kyuyeong Lee","email":"kyu0.lee@samsung.com","username":"kyu0"},"ref":"refs/changes/69/926469/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/69/926469/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/926469/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/926469/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/926469/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/926469/2"}}},"commit":{"parents":[{"commit":"5225abc6049e0b4912d6635df0a142b8bfccebca","subject":"Fix check of the revision number in the resonse from API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5225abc6049e0b4912d6635df0a142b8bfccebca"}]}],"author":{"name":"kyu0","email":"kyu0.lee@samsung.com","date":"2024-08-16 12:21:34.000000000","tz":540},"committer":{"name":"kyu0","email":"kyu0.lee@samsung.com","date":"2024-08-22 15:23:21.000000000","tz":540},"subject":"Add test \"test_sg_rule_quota_with_default_sg_rules\"","message":"Add test \"test_sg_rule_quota_with_default_sg_rules\"\n\nThis test checks that the SG rule quota is working properly when\ncreating an SG.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/921909\nRelated-Bug: #2067239\nChange-Id: I40d9f26a3d76d872de93d86ded5d40ccbd696f2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6cd2f054f9fde9522098c365828b38aa0dede061"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6cd2f054f9fde9522098c365828b38aa0dede061"}]},"branch":"refs/heads/master"},"e938921e102ca79c2515f5e46c5f0c56c2d2110c":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2025-09-23 11:16:49.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/69/926469/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/69/926469/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/926469/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/926469/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/926469/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/69/926469/3"}}},"commit":{"parents":[{"commit":"6cac8c82e4916743d144a5a40aa78b21cdab13ce","subject":"Merge \"Add hacking check to the pre-commit configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6cac8c82e4916743d144a5a40aa78b21cdab13ce"}]}],"author":{"name":"kyu0","email":"kyu0.lee@samsung.com","date":"2024-08-16 12:21:34.000000000","tz":540},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-09-23 11:14:28.000000000","tz":120},"subject":"Add test \"test_sg_rule_quota_with_default_sg_rules\"","message":"Add test \"test_sg_rule_quota_with_default_sg_rules\"\n\nThis test checks that the SG rule quota is working properly when\ncreating an SG.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/921909\nRelated-Bug: #2067239\nChange-Id: I40d9f26a3d76d872de93d86ded5d40ccbd696f2a\nSigned-off-by: Kyuyeong Lee \u003ckyu0.lee@samsung.com\u003e\nCo-Authored-By: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e938921e102ca79c2515f5e46c5f0c56c2d2110c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e938921e102ca79c2515f5e46c5f0c56c2d2110c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
