)]}'
{"id":"openstack%2Fneutron-tempest-plugin~926508","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Id7048793d75640694299f340762f079111c81565","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2076328","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-05-14 15:46:26.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-04-24 15:34:00.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2025-04-24 17:24:42.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Id7048793d75640694299f340762f079111c81565","subject":"Unset \"router:external\" in subnet tests","status":"MERGED","created":"2024-08-19 08:53:16.000000000","updated":"2025-05-14 15:46:26.000000000","submitted":"2025-05-14 15:46:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926508-bug/2076328","meta_rev_id":"65de581605ecd649600099fc91b4175fcd410ce6","_number":926508,"virtual_id_number":926508,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-14 15:46:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2025-04-24 17:24:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2025-05-14 12:47:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2025-05-14 11:29:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2025-05-14 12:47:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-19 10:48:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-19 14:31:01.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2025-04-21 12:53:15.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2025-04-30 13:04:26.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2025-05-14 12:47:08.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"c7e23413d2375fd70bd67a11dc1437de01292782","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-19 08:53:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad35360689d19b55a2f079b08e4bb8fa39790c27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 10:48:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68e3b884ff6c4dcf95bda2f950c1629d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf06043ca9d5452fa515922b29174ca3 : SUCCESS in 3m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ba0b2bd9f45b419ea2211cd514a53586 : FAILURE in 1h 21m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cf50be1ad1794fddbf479aecfa006c3b : FAILURE in 1h 33m 10s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/de5aa53ba9804995b7484db318dfed00 : FAILURE in 1h 26m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/447a656d6baf466f9b7f62491735929b : FAILURE in 1h 53m 29s\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/a37ed982159946328723bf7f7ff35f00 : SUCCESS in 1h 29m 22s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/20aedca6890c48f1979a0f57324bd287 : SUCCESS in 1h 13m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/065b13c337c84ffda1e1c51301efa5c0 : SUCCESS in 1h 10m 10s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/e759dea39a214206b2b952fe38335a54 : SUCCESS in 1h 25m 32s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/07ebb4bc2e8b434eade1fb8a36179fcc : SUCCESS in 1h 07m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/c8950c3a3fad446ca5044f03484df7e5 : SUCCESS in 1h 18m 36s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/f9f7faacdbcb4fa79bcc6001ebf31397 : SUCCESS in 1h 35m 54s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/d76997ae070f4280b88150e73b2d121e : SUCCESS in 1h 13m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/24c4c442064445cc81e445336403340b : SUCCESS in 1h 40m 45s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/3b9aad67b4c74137aac1e0ce05b12061 : SUCCESS in 1h 24m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c232e1bc4844bf691f5396d6534e148 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a5cc661ce0ea24993833eb66ee622e6a99613066","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-02 07:34:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"d2353a1fb7009d5b443c267dc8bd3e44f083cc50","tag":"autogenerated:gerrit:restore","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-04-21 08:38:09.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"6d846afc2dcb4beaf23d0349ad4d6de18aae66a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-04-21 08:38:50.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5dcb00521d7954784ce790b0a88f77a587db14cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 11:20:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0158a8c0c8b4c1999316639cf3b1b74\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a29f32a74ba4770a009f06919627896 : SUCCESS in 5m 01s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c0d875164d7942a5b4ff4a0ce6a82747 : FAILURE in 1h 45m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7318d7ffedc74c5e94407ba42294e054 : FAILURE in 2h 40m 06s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/2153fe97b2274e0683815d1b6aec0694 : SUCCESS in 1h 37m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/3666dbd976e14c4282c738e0629e21c6 : SUCCESS in 1h 32m 28s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/21858f068e444bcfb5e088616d164444 : SUCCESS in 1h 52m 09s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/afdeb923f3ac4718b4cc6ba7b364a8ce : SUCCESS in 1h 34m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/c70056c394654688a6ef391aa5c9ad5f : SUCCESS in 1h 56m 48s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/2ea84d2e183845feb4add7b4e3555479 : SUCCESS in 1h 45m 09s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/75574a35e05b43a2a31df2c1cbe75b64 : FAILURE in 1h 27m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/6f135a60f1024571a8a418627f005dcb : FAILURE in 1h 16m 23s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/0cae28ba158049be80af060b43ec6af0 : FAILURE in 1h 45m 14s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/be99e25f99094604a3aca208517fd04b : FAILURE in 1h 09m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/f340e7a484c94aad851daaf060321d32 : FAILURE in 1h 43m 10s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/7406b49807dc4c569d76788982c17222 : FAILURE in 1h 22m 11s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/7454cbc5fc304ff098b69f3eef3cc3ff : FAILURE in 1h 09m 30s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/9f845e3f80804de4be1f483a3fd3258f : FAILURE in 1h 57m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/895688395e7f4659a9c3b2504b081222 : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":2},{"id":"1785d5ccdc101b4ec67fcef9c9679f4961e2ea27","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-04-21 12:53:15.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/07b38a169b8c495da1f854727bb8086c\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/28d8a18021a04c378d1841231485fbc8 : SUCCESS in 4h 13m 20s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/09c21d3a8df94b838ec6b01e46914228 : SUCCESS in 3h 15m 16s","accounts_in_message":[],"_revision_number":2},{"id":"ff066f283fe49e0a241b476cabec06c8c894f806","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-04-24 13:03:58.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"73a0e58fa0f56e555ac91f0801a95981acf69691","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 15:34:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35ea99534e8b434fbf55193a95fbadf8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1efde6cf9f4c4c6c9db19e0ca028dbf0 : SUCCESS in 4m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c45a2d9179ae491283f2d659b93f6d37 : FAILURE in 24m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5c1c6e6792d146c08dcab386d0380649 : FAILURE in 2h 26m 30s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/1e3bf99fb8bb45eaa69376aec066b57e : SUCCESS in 1h 01m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/7dfc8f5384564e3b8c496c155fcc390c : SUCCESS in 1h 17m 22s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/fb5f04952ead45f1b44e60bc7504d942 : FAILURE in 22m 18s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/4ccc8ae1dce040bcb13b2967a076a100 : SUCCESS in 1h 33m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/00ff5bb24a1649f08d0fe31ca4dd8c14 : SUCCESS in 1h 41m 10s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/c287fd4e866d4c40b5d4ddcc3201b457 : SUCCESS in 1h 28m 58s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/9534d4e7ffe541f49b82a96367084a5c : FAILURE in 1h 29m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/3d1f1e49966047b3927ff91289ee14c9 : FAILURE in 2h 01m 06s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/07feaa2acd1a4d3ea51fc6f03a88ecec : FAILURE in 1h 21m 30s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/46894120f8334e588fbdc371c94d239c : FAILURE in 2h 00m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/d1d7d64c32d24257ad0748b1bd0f4d75 : FAILURE in 1h 15m 22s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/98641ad397b146a59fffb7928817db4e : FAILURE in 2h 09m 29s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/962ba4e11ab54986975e27201997958b : FAILURE in 1h 01m 20s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/96617772c7e54be9b9bc82da47c7f8d0 : FAILURE in 1h 21m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ed48f70a972454fa3160e9e586eb501 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":3},{"id":"57cacd461eb639ccda9cfbdd6036372433e41788","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-04-24 17:24:42.000000000","message":"Patch Set 3: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/9494b08382d146d1b4ff7307e1607c26\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/fd01bf51600f4d23805d5560dd08e76c : SUCCESS in 4h 19m 07s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/e3f9fca5c3734337a18a3e258c76013c : SUCCESS in 3h 15m 13s","accounts_in_message":[],"_revision_number":3},{"id":"3dfac90c2f8a33b9da960fed947704fcfd749519","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-04-25 10:32:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0fa1c8b3c94e3b98b073f0acdfc1be60f17dce00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 12:56:42.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36d91351d8704103b430a04333328b20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55743bbc3b1b4bd6adffc274c5986b2b : SUCCESS in 3m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4cf55b7a510c427aad711bacd15b1ac2 : SUCCESS in 1h 09m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/729c1218292d466eaf901644f48ee781 : SUCCESS in 2h 17m 21s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/dc6599c739c74399893c7402322b4bec : SUCCESS in 49m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/beb70c97b5f0456592aa5a7125388bc6 : SUCCESS in 1h 43m 08s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/ae510d9bc3d6499b911389a1af111f53 : SUCCESS in 1h 46m 03s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/1ceea38de5bc4a3ba56cbe5edee4f993 : SUCCESS in 59m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/0e65acc5f20c4d1aa264aba0b1255d5a : SUCCESS in 1h 02m 54s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/e0582a670e4e49408f54ebdf3b767507 : SUCCESS in 2h 12m 03s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/c5ae34cd5c72463fb3fa8c340d6e4d39 : FAILURE in 1h 45m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/087a26b1c67a489d87437e881d42c184 : FAILURE in 2h 01m 17s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/9fe9257901164ac3b9f4416e9256e0a5 : FAILURE in 1h 17m 11s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/dc280252a1df4eaa80f84eb033720d8e : FAILURE in 1h 36m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/eec769e4382a4030876fb8b454bf345d : FAILURE in 2h 12m 46s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/75baa2884be443ad96f5315471501644 : FAILURE in 2h 04m 26s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/56ba6e3ff9b242c09d16191721af1141 : FAILURE in 1h 03m 38s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/e9999d0cef1142e0bef7178a95823601 : FAILURE in 1h 41m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e13da670a105449e802f28c19c495213 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":3},{"id":"29c59ec3ed62a56caeacca06764fd5fe539d2780","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-04-25 14:46:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/75f7876d94ff4eb99be505b7652fa9f5\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/dcd752fb15d44fac8db8cdba5187f008 : SUCCESS in 4h 12m 11s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/936c732a156746a8aaf05f980e1250af : SUCCESS in 3h 11m 03s","accounts_in_message":[],"_revision_number":3},{"id":"c921d086008cee806716eb2ab0f9b4d6661c1bb1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-04-30 13:04:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"851a1b8255a0ede2e5648dd786f7d5f18c8bcccb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-04-30 13:04:45.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"9fc3758854659ad5c906b01e7c777d80f5451a63","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-05 07:03:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7bf251abc58d7015293912102f88ef60d00e1212","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-13 12:48:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9742ab3c2c0e77f3e8aad8092beeb140932d50fc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-13 12:48:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"58779d6b284eb68d8c1900cd0ef9ef543feffd63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 15:15:49.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/20697f97d927436585f1ab4c6d90a1ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12cf5bb1f0de41ad92ddd645dcf0b77f : SUCCESS in 4m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5269c4dbcfc145ecbd14d87f2c417a6e : SUCCESS in 1h 57m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/02cf3a3be6604c8085853ac88bd05ca6 : SUCCESS in 1h 49m 34s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/b7d773e2ba4c4988a42d263f00139f8c : SUCCESS in 1h 05m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/da64ea0461554db181943947f5ffa0a6 : SUCCESS in 1h 23m 13s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/0bf8d6a669204ea898f36d63d7714573 : SUCCESS in 1h 45m 20s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/fcdc3c0956424c0fa5eae04577517315 : FAILURE in 22m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/cc1220de8f7f4345935dcd14156eba5f : SUCCESS in 2h 05m 09s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/bd313087ca8947ed88a3fc4da0a7c2a1 : SUCCESS in 1h 37m 15s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/b889b8dad0814fab91dacb21477dc8c7 : FAILURE in 17m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/5ed4d59a978348da85eb07622e599d8b : SUCCESS in 2h 19m 10s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/0f5fa851f9e24315965d0b567c668ffd : SUCCESS in 1h 34m 56s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/e1daa99270d54791a490fa389f3aa6ec : SUCCESS in 1h 28m 15s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/dce181f2c1f74d0a9a497c9376614ac4 : SUCCESS in 2h 07m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/583f89ae97c94680a1be5230fbfad8a3 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":3},{"id":"1718c464b440c99009096bf8fbdf176e553bb1ae","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-05-13 16:42:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/d8a7d627a6e1482b85cbabdde63c9028\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/4d2d2537bde740618b5ae9bf90f71245 : SUCCESS in 3h 51m 25s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/78dfda0118a74d40bef1f472133370a8 : SUCCESS in 3h 00m 00s","accounts_in_message":[],"_revision_number":3},{"id":"0472cdcc6498b4ead494e399c5f84cd4e0ae1132","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-14 06:01:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3b36c0a6aaa27d76aee244102cc92d17d3e133e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 08:46:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5daa99da93634c91aa6f9ea436870cbb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c82a78885b284b0c9f75eef8da0134a6 : SUCCESS in 5m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/aae994ca8b0c4943a6297cfe646bbfd4 : SUCCESS in 1h 35m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ec1659a8e3c041bebc334fdeae3edf8f : SUCCESS in 2h 38m 30s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/92e9db12427c4a65a7ae01c21c101d4b : SUCCESS in 1h 02m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/6b5d3bdb85c2451cb631b9a02875cf81 : SUCCESS in 2h 01m 28s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/64d45e3bdc804ff894d0a6d20bcb4ad2 : SUCCESS in 1h 47m 03s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/88244c2c0169446995cc2ef29e39ba00 : SUCCESS in 1h 24m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/0fc3d4596840405f8585b4e39fe38014 : SUCCESS in 1h 52m 40s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/d9c0bab47ead49849574c8dbf789de2e : SUCCESS in 1h 26m 20s\n- neutron-tempest-plugin-openvswitch-2025-1 https://zuul.opendev.org/t/openstack/build/65cf5fd7c4364a11b953bd25a51988b4 : SUCCESS in 1h 21m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2025-1 https://zuul.opendev.org/t/openstack/build/912a9864ec5846fd87c604009d94ed95 : SUCCESS in 1h 24m 10s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/9278762b3e174248979560a511ab8227 : SUCCESS in 2h 44m 08s\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/e408087eb49f42249f37dba978cf4830 : SUCCESS in 1h 21m 25s\n- neutron-tempest-plugin-ovn-ubuntu-jammy-2025-1 https://zuul.opendev.org/t/openstack/build/29ccb4c9fb6945aeafbb1b24e87d1793 : SUCCESS in 1h 45m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b60804ab8ff34b178cdfa6abd12161dd : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":3},{"id":"f2c3e9e28d9448acf003e6c3ad77e20775152b75","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2025-05-14 10:13:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/c4b19e19a1ba470a8542034cd4405054\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/339e99b990eb468882d304a7ca3e9ef3 : SUCCESS in 4h 10m 35s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/97bc723a9c6c4fb79afaefddd216ec2d : SUCCESS in 3h 15m 02s","accounts_in_message":[],"_revision_number":3},{"id":"a7819a6833c25d117274e847fe224cf7990de592","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-05-14 11:29:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a0d35b2237bbe4c2751419a0331f6cc23f942db9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-05-14 12:47:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"951aed775f07b09999d35c58ffb0b546d359ec31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 12:47:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d7faa6a00ce4c32e4485f65ff5f85ee2b1a8dbbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 15:46:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/095a99bf2d1c40d98ab6c2885d6ad2e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b360179c9f6f49bdba6caf957a3f80ab : SUCCESS in 6m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5844571eb0744f80b2bd63d6f9699368 : SUCCESS in 1h 56m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cfe97f279d124b9bb29203961dda84fc : SUCCESS in 2h 57m 23s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/cf7cce067e364620abdf19b267e2b5a9 : SUCCESS in 2h 19m 13s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/15053eb1c87d4efcbb448d8aaa6e2f07 : SUCCESS in 1h 46m 45s\n- neutron-tempest-plugin-ovn-2025-1 https://zuul.opendev.org/t/openstack/build/0d43dbc02672432f9f75253e7a985588 : SUCCESS in 1h 35m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74518751d0a842b8bca1d268581eb4a1 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":3},{"id":"65de581605ecd649600099fc91b4175fcd410ce6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 15:46:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f4fb68d5c295bde4ae4dadd83597a994b47823f0","revisions":{"a4a8c9f376cba07a9e0949a8936c6bd54babb269":{"kind":"REWORK","_number":1,"created":"2024-08-19 08:53:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/08/926508/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/08/926508/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/926508/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/926508/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/926508/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/926508/1"}}},"commit":{"parents":[{"commit":"5225abc6049e0b4912d6635df0a142b8bfccebca","subject":"Fix check of the revision number in the resonse from API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5225abc6049e0b4912d6635df0a142b8bfccebca"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-08-06 16:36:38.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-08-06 16:36:38.000000000","tz":0},"subject":"Unset \"router:external\" in subnet tests","message":"Unset \"router:external\" in subnet tests\n\nSince [1], Neutron will return also the subnets belonging to an external\nnetwork. By default, the class ``SubnetsSearchCriteriaTest`` will not\nconsider these networks, same as in the class\n``NetworksSearchCriteriaTest``.\n\n[1]https://review.opendev.org/q/topic:%22bug/2051831%22\n\nCloses-Bug: #2076328\nRelated-Bug: #2051831\nChange-Id: Id7048793d75640694299f340762f079111c81565\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a4a8c9f376cba07a9e0949a8936c6bd54babb269"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a4a8c9f376cba07a9e0949a8936c6bd54babb269"}]},"branch":"refs/heads/master"},"f52cd8a8202740851e069f690bc6b63b1121b1c6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-04-21 08:38:50.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/08/926508/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/08/926508/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/926508/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/926508/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/926508/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/926508/2"}}},"commit":{"parents":[{"commit":"5225abc6049e0b4912d6635df0a142b8bfccebca","subject":"Fix check of the revision number in the resonse from API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5225abc6049e0b4912d6635df0a142b8bfccebca"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-08-06 16:36:38.000000000","tz":0},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2025-04-21 08:38:50.000000000","tz":0},"subject":"Unset \"router:external\" in subnet tests","message":"Unset \"router:external\" in subnet tests\n\nSince [1], Neutron will return also the subnets belonging to an external\nnetwork. By default, the class ``SubnetsSearchCriteriaTest`` will not\nconsider these networks, same as in the class\n``NetworksSearchCriteriaTest``.\n\n[1]https://review.opendev.org/q/topic:%22bug/2051831%22\n\nCloses-Bug: #2076328\nRelated-Bug: #2051831\nRelated-Bug: #2082883\nChange-Id: Id7048793d75640694299f340762f079111c81565\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f52cd8a8202740851e069f690bc6b63b1121b1c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f52cd8a8202740851e069f690bc6b63b1121b1c6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f4fb68d5c295bde4ae4dadd83597a994b47823f0":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-04-24 13:03:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/08/926508/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/08/926508/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/926508/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/926508/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/926508/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/08/926508/3"}}},"commit":{"parents":[{"commit":"5225abc6049e0b4912d6635df0a142b8bfccebca","subject":"Fix check of the revision number in the resonse from API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/5225abc6049e0b4912d6635df0a142b8bfccebca"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-08-06 16:36:38.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2025-04-24 13:03:58.000000000","tz":0},"subject":"Unset \"router:external\" in subnet tests","message":"Unset \"router:external\" in subnet tests\n\nSince [1], Neutron will return also the subnets belonging to an external\nnetwork. By default, the class ``SubnetsSearchCriteriaTest`` will not\nconsider these networks, same as in the class\n``NetworksSearchCriteriaTest``.\n\n[1]https://review.opendev.org/q/topic:%22bug/2051831%22\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/948073\n\nCloses-Bug: #2076328\nRelated-Bug: #2051831\nRelated-Bug: #2082883\nChange-Id: Id7048793d75640694299f340762f079111c81565\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f4fb68d5c295bde4ae4dadd83597a994b47823f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f4fb68d5c295bde4ae4dadd83597a994b47823f0"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
