)]}'
{"id":"openstack%2Fneutron-tempest-plugin~926940","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ie2e49a53857009446e22300e2fff292355cef058","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-08-29 09:30:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie2e49a53857009446e22300e2fff292355cef058","subject":"Implement nested snat validation test scenario","status":"MERGED","created":"2024-08-22 18:11:08.000000000","updated":"2024-08-29 09:30:09.000000000","submitted":"2024-08-29 09:30:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926940","meta_rev_id":"a0d5585e5fc5b3b89d067523304c40a507949fdb","_number":926940,"virtual_id_number":926940,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-29 09:30:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-08-29 07:17:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-08-29 08:10:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2024-08-28 17:39:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2024-08-28 14:23:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-08-29 07:17:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-22 18:49:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-22 19:59:01.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-08-28 14:23:05.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2024-08-28 17:39:43.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-08-29 07:17:11.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-08-29 08:10:35.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"b2a8732f7e4b1bef35309ff564308b5cf508b973","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-22 18:11:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6169cf7c602929ac4fe069361be5fb8f87d105fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 18:49:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56505db55f52420c983663b0d93bbe08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/269d26ed2e43465481ad9f1580c5acb2 : FAILURE in 3m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/02b5c52c90cc4d35b9b8ed6f773f164e : FAILURE in 25m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/771a2444b3664e34aa368caca7841d8c : FAILURE in 26m 16s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/83c081189ea14fdaa88d0c6c79895806 : FAILURE in 24m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ae88b6caa18045b68490d654254da1cf : FAILURE in 34m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e5ce307efe2b4fe6aad17c98c4f946a7 : FAILURE in 34m 54s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/4bdf58678c4b43309635cdc7c8780cca : FAILURE in 31m 18s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/de33102b516a489db3ca82c7b6b07ce3 : FAILURE in 25m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/5c8c6f430b834ee1a59c8f2a7edcf08c : FAILURE in 31m 51s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/3a6f29238cc0408e9505ef7d08671ca6 : FAILURE in 33m 37s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/f0e49b881853472d8af3867c7961c63f : FAILURE in 31m 18s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/b4ec9f4d74b9453fab01153ba4b33426 : FAILURE in 24m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/15792eed2fbb4af299befe9f7721f720 : FAILURE in 26m 56s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/738af1671d174246abde2187ec1ce42d : FAILURE in 33m 30s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/6f2697bdb7c3445a866bcecf81822e86 : FAILURE in 31m 07s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/6e0aae7c7b7a416492e221e642ffa93f : FAILURE in 30m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/baa59fc03adf4616a864cdf2f7838d7d : FAILURE in 30m 43s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/3ea9978699b9414f971594946e2f042c : FAILURE in 30m 46s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/04fb0240c4df45d3b734a05158e1e18c : FAILURE in 34m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9e94ded6486457b97f5b29654971fad : FAILURE in 2m 39s","accounts_in_message":[],"_revision_number":1},{"id":"35c655682d70d70498afcec6641315bc0b3fcf1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-22 20:17:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2af69edd15eab459de52a5c13c9a62c276b119b1","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-22 20:18:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3455bf1f29f6c8200dca27e433e63c7c7c6457a8","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-22 21:03:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e90b1da7e603f9a2b220708d237cd59f195fe22","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-22 21:04:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a3abb5acb27d419d361fe6e6453e085641e61f4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 22:37:12.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c74c72eff2084e76a570eefcd081c4c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6db295c02f904735891af26f04a1e6b4 : SUCCESS in 3m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/76798acc20bd4239af7845fd4526dc8f : SUCCESS in 1h 20m 59s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/74c35936bfbd455f9a00773a174c3ad0 : SUCCESS in 1h 29m 46s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4458693f575e4477bc1b758a9711705e : SUCCESS in 1h 24m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/551c4bc711614b7f85f0c29aa72b4973 : FAILURE in 2h 01m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0aa31932a93a4d6eb6027de47b7ac215 : SUCCESS in 30m 16s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/a3918198e30a460e82305c1cec6b1754 : SUCCESS in 1h 35m 41s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/917f48766c874c438aaf0581e78da50c : SUCCESS in 1h 57m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/1fcb39b48b164b88a85a2af0c7020884 : SUCCESS in 1h 49m 25s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/88b97245bdde40f9a0d5902f5f777975 : FAILURE in 2h 06m 10s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/9394d74c02f743fe9a931ba57554f098 : SUCCESS in 45m 11s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/ed06768520eb477fa44a360b6d7d112d : SUCCESS in 1h 39m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/5f7b5507fce84261b77e8210a96066b5 : SUCCESS in 2h 10m 33s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/5e27ff0162dc411c880b36595bb5d8c1 : FAILURE in 2h 13m 49s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/ff5e599ec19e4df4aa4ec93901b11078 : SUCCESS in 46m 40s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/7bb1f62a7b164bc09035b19c6ca87774 : SUCCESS in 1h 50m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/646ebc7eb3c941d886992e93c979b288 : SUCCESS in 1h 48m 02s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/ddfa560d95ba48d8b1ac3c9e4ed6e74d : FAILURE in 1h 56m 54s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/bbbf6593737542849a64ce3ba4a3694f : SUCCESS in 36m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c24cf5b2e04542d5814a7ca461882add : FAILURE in 2m 46s","accounts_in_message":[],"_revision_number":2},{"id":"cc85a4ffaae781f14e95e484ee93759350e22aba","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 23:12:12.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65bff9e336fe49ecb6775f2e697bf4ea\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/259d79ad5e6343489fbb50f3093d93a0 : TIMED_OUT in 2h 49m 59s\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/1e894add2d404c35a69aa91f83a5c5f5 : SUCCESS in 1h 55m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/81680a966a584d08bdb4f1db44b3433d : FAILURE in 2h 03m 30s\n- neutron-tempest-plugin-api-ovs-wsgi https://zuul.opendev.org/t/openstack/build/19616b79fa004d4cbf3c4b750d1c3d5e : SUCCESS in 2h 00m 56s (non-voting)\n- neutron-tempest-plugin-api-ovn-wsgi https://zuul.opendev.org/t/openstack/build/da28cd5e04b3418ca5136d4a0f9dceb0 : FAILURE in 21m 11s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-2 https://zuul.opendev.org/t/openstack/build/162cd01048644033990cf600b6e40861 : SUCCESS in 2h 06m 43s\n- neutron-tempest-plugin-linuxbridge-2024-1 https://zuul.opendev.org/t/openstack/build/09cd8ac6dd8e4f758c2013cb3a3fc558 : SUCCESS in 1h 38m 51s","accounts_in_message":[],"_revision_number":2},{"id":"e362c669302489a21912081c2e3056f4d688e1ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-26 23:20:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e944ecc48383e2fea11688455a3d738d3ad6c058","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-26 23:34:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fa7ce5944c09381bea701abdced38b0d207005bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 01:47:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59540acecc8b44d093a114bdce5cd055\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bfe96ab483b4530aec0cb4fef7c16af : SUCCESS in 4m 26s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/106a82aa759a404e9e924aa5a4ddf64e : SUCCESS in 1h 47m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b65203309b53492ba72cc8bb81951025 : SUCCESS in 1h 40m 06s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/214f5381c7d74a408be89c0bee2c4b94 : SUCCESS in 1h 57m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/873eb85ab9034a5ea7dc7083e3bd8138 : SUCCESS in 2h 11m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a3c7cd8cb5024319b09e4c967441f7d5 : SUCCESS in 39m 32s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/3b1cb955d1b5463ab92ef9a6371584ff : SUCCESS in 1h 16m 09s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/ad6cb406d08843c6a2d7f786bec14c94 : SUCCESS in 1h 17m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/cbdbcb5e0b6648c8ab6422943ba40f2a : SUCCESS in 1h 24m 39s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/85288fc16e2448a396ba3d24368de431 : SUCCESS in 1h 58m 16s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/fe367477559744ebb1d356c8ee323644 : SUCCESS in 39m 53s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/2ead1b2ad5e6499795e16ea490a06f95 : SUCCESS in 1h 11m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/dd3712f5d8844ccbb2571ab059b6131c : SUCCESS in 1h 46m 14s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/fd46d9cdecd34adfa84befa7391b21a3 : SUCCESS in 1h 46m 22s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/9a40d62e611146b2b4706682e55bd6b5 : SUCCESS in 40m 30s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/41df2beb227f4440904c2c4a4a7daf85 : SUCCESS in 1h 36m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/1600a975edb24e3391c2c71c3742da5d : SUCCESS in 1h 40m 35s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/ea5160f64c8342c8850e1d996976f2d2 : SUCCESS in 1h 54m 11s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/4eecf02f27c74997b256a8b60685f3ad : SUCCESS in 38m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/533fc9a83e834533af3edbc9f1b3741a : SUCCESS in 2m 48s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/5e7c70b5da6740e692057ea936036d23 : SUCCESS in 1h 56m 05s\n- neutron-tempest-plugin-sfc-2023-1 https://zuul.opendev.org/t/openstack/build/8146d3f952c443ba92a90b09ae4f58c6 : SUCCESS in 1h 36m 47s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/56385813af9b4ea2bcf04aa5d1833dd6 : SUCCESS in 1h 40m 30s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/0cfd5828eabb4aa2bbfa558a5213d901 : SUCCESS in 1h 11m 19s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/50e98103ac604722afaf23b8a5dfa946 : SUCCESS in 1h 26m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/7747ff12f9c74ca6b35a212a90c416d1 : SUCCESS in 1h 07m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/2909348983cc485d9adbb185809b79f1 : SUCCESS in 1h 24m 47s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/6b666fc35ed841adb4789e3ec8246c2c : SUCCESS in 1h 43m 08s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/3282a0f9e4624fe8aa997db1d95ac4c8 : SUCCESS in 57m 57s\n- neutron-tempest-plugin-dynamic-routing-2023-1 https://zuul.opendev.org/t/openstack/build/91401bb9aa194208af94dbc0e2357d62 : SUCCESS in 57m 47s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/94ad6196a79d49119fc59733c92217e7 : SUCCESS in 55m 38s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/3a5468767b134d09854874e9639f81aa : SUCCESS in 52m 36s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/5c5fcca3b25b4b00afff5fccfe6a8d4d : SUCCESS in 43m 14s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/eae4f30f7414452ab02f2ac39bc379d4 : SUCCESS in 44m 08s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/d3f9daee91a94276947aed6e64c79a38 : SUCCESS in 45m 39s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/c9e7a5301690417ebea9b421851c7317 : SUCCESS in 41m 31s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/1c9a3f90e8074c65898fae78ed601e3c : SUCCESS in 57m 13s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/12c333656e62450dbc1708755bf0179f : SUCCESS in 49m 14s\n- neutron-tempest-plugin-vpnaas-2023-1 https://zuul.opendev.org/t/openstack/build/37e548cf7afd43abafcec1db35bae4f1 : SUCCESS in 40m 09s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/e15d2cf3588e4d62a1d4c2b8843b0a4b : SUCCESS in 46m 42s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/ac2be9e5d9124afca9fb8896d16e61cc : SUCCESS in 50m 21s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/ef55f824b55e42edb2c95f4ffff18212 : SUCCESS in 1h 17m 57s\n- neutron-tempest-plugin-tap-as-a-service-2023-1 https://zuul.opendev.org/t/openstack/build/14eb38bc8b884f6e932b38fbb9676bff : SUCCESS in 1h 28m 10s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/0c28c5b2eaf74ebebe83b66b6ca648d1 : SUCCESS in 1h 29m 21s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/5321a68ef7f349569cb7e1fbd93f8f6f : SUCCESS in 1h 08m 21s","accounts_in_message":[],"_revision_number":4},{"id":"d5f4b0b17474debc0163bfe66c5ea1a605d987a0","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-27 13:31:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"87955f35d3c5326354a77ef8f6210e28bac71062","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 15:47:04.000000000","message":"Patch Set 4:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d3fdcce3e18498e8497d906b74d5452\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d6f5b65d0ef94069bab6a80a75b8cfcf : SUCCESS in 1h 50m 23s\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/66c22df450a5491d9e29ce86438a064a : SUCCESS in 1h 16m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/2a834b49965a435fbeab8757c899f5c8 : FAILURE in 1h 41m 57s\n- neutron-tempest-plugin-api-ovs-wsgi https://zuul.opendev.org/t/openstack/build/2894118c13e74e0f9e4db969191a5222 : SUCCESS in 1h 32m 46s (non-voting)\n- neutron-tempest-plugin-api-ovn-wsgi https://zuul.opendev.org/t/openstack/build/174aa70230254064926338055ceb36e8 : FAILURE in 16m 03s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-2 https://zuul.opendev.org/t/openstack/build/88302e1f6e9648c8bb7a24ff7373c82f : SUCCESS in 2h 13m 23s\n- neutron-tempest-plugin-linuxbridge-2024-1 https://zuul.opendev.org/t/openstack/build/7d130b4b7499420581328c6af1080606 : SUCCESS in 2h 07m 16s","accounts_in_message":[],"_revision_number":4},{"id":"0633d54f9c292114a18f0e153f2c2c9579e058d2","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2024-08-28 14:23:05.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"48fed5f4b4692eecd96256d88c50fcf2c42432ad","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-28 17:39:43.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0bbc265cff69c066c26d27b7beaf6f3f3402ac98","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-28 19:36:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"73e270c5924c939b408bddeda85fb4288a64e6d3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-29 07:17:11.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b38a52e7f6dfff4a9b75866f9342d79584ba9910","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 07:17:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"43697144e2c48caf493e76278e91d7f408c1ddb6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-29 08:10:35.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"17bf73451c0812af07eae37166001eb772237ef4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 09:30:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00d76fe9f5f44b0fbb45b5f9ba854789\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e37e8aa366b412b991c2053e8a57ab0 : SUCCESS in 3m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e8bc6dd22db2485a9e889c8df6676852 : SUCCESS in 1h 37m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/21e031328c18419e83562f95bcb48386 : SUCCESS in 2h 10m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/47ef0ef6b492407b8065189ea9f9360f : SUCCESS in 2h 03m 58s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/3f1019df74cf431a9f0de89151bca167 : SUCCESS in 1h 56m 57s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/b8428d2e4faf4525ba8e00e96abe9d43 : SUCCESS in 1h 47m 12s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/d1521477e50e40fab9c487b73655312c : SUCCESS in 1h 57m 42s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/f11ad5c27ec14a9091e5a4044af453bf : SUCCESS in 1h 56m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d3a4e3d1f6a4d778762cdb94f3a63d9 : SUCCESS in 3m 02s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/b39afb7086e1475885f8d6c93dc500f7 : SUCCESS in 1h 57m 35s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/dd81ee93e8c64f20b64c7140b27d44a9 : SUCCESS in 1h 42m 34s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/65c887d960fc4a2bbe1291b50122d982 : SUCCESS in 59m 19s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/5e1166b586f8436aaf01532ff9c7f7bf : SUCCESS in 44m 10s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/cfcc548256094257b0efdd6ccefe48f4 : SUCCESS in 48m 58s","accounts_in_message":[],"_revision_number":4},{"id":"a0d5585e5fc5b3b89d067523304c40a507949fdb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 09:30:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"889566aa9735cade0842af45b2e040c23a60c8a1","revisions":{"013062a663038a97bbaac7bc3a8386c1601b3b63":{"kind":"REWORK","_number":1,"created":"2024-08-22 18:11:08.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/40/926940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/40/926940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/926940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/926940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/926940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/926940/1"}}},"commit":{"parents":[{"commit":"54e4ebe02d7034d77732beba926aae5cb172a308","subject":"Merge \"Test metadata query over IPv6 only network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/54e4ebe02d7034d77732beba926aae5cb172a308"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-08-22 18:08:29.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-08-22 18:10:55.000000000","tz":0},"subject":"WIP Implement nested snat validation test scenario","message":"WIP Implement nested snat validation test scenario\n\nI expect it to fail for ovn right now. We\u0027ll have to disable the\nscenario for ovn jobs by default. For now I just want to see it fail in\ngate.\n\nRelated-Bug: #2051935\nChange-Id: Ie2e49a53857009446e22300e2fff292355cef058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/013062a663038a97bbaac7bc3a8386c1601b3b63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/013062a663038a97bbaac7bc3a8386c1601b3b63"}]},"branch":"refs/heads/master"},"d91e832a4db8ed065b1f0feb112ec85e44870717":{"kind":"REWORK","_number":2,"created":"2024-08-22 20:17:39.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/40/926940/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/40/926940/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/926940/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/926940/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/926940/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/926940/2"}}},"commit":{"parents":[{"commit":"54e4ebe02d7034d77732beba926aae5cb172a308","subject":"Merge \"Test metadata query over IPv6 only network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/54e4ebe02d7034d77732beba926aae5cb172a308"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-08-22 18:08:29.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-08-22 20:17:35.000000000","tz":0},"subject":"WIP Implement nested snat validation test scenario","message":"WIP Implement nested snat validation test scenario\n\nI expect it to fail for ovn right now. We\u0027ll have to disable the\nscenario for ovn jobs by default. For now I just want to see it fail in\ngate.\n\nRelated-Bug: #2051935\nChange-Id: Ie2e49a53857009446e22300e2fff292355cef058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d91e832a4db8ed065b1f0feb112ec85e44870717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d91e832a4db8ed065b1f0feb112ec85e44870717"}]},"branch":"refs/heads/master"},"dd11269ad5d370252b55ec161727ef6f3e73f6f9":{"kind":"REWORK","_number":3,"created":"2024-08-26 23:20:48.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/40/926940/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/40/926940/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/926940/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/926940/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/926940/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/926940/3"}}},"commit":{"parents":[{"commit":"54e4ebe02d7034d77732beba926aae5cb172a308","subject":"Merge \"Test metadata query over IPv6 only network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/54e4ebe02d7034d77732beba926aae5cb172a308"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-08-22 18:08:29.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2024-08-26 23:18:44.000000000","tz":-240},"subject":"Implement nested snat validation test scenario","message":"Implement nested snat validation test scenario\n\nIt\u0027s disabled for ovn because it\u0027s not, yet, implemented for the driver.\n\nSee: https://review.opendev.org/c/openstack/neutron/+/926495\n\n(The feature also requires special configuration for the driver that we\nprobably won\u0027t enable in the gate. This can be discussed though.)\n\nRelated-Bug: #2051935\nChange-Id: Ie2e49a53857009446e22300e2fff292355cef058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dd11269ad5d370252b55ec161727ef6f3e73f6f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dd11269ad5d370252b55ec161727ef6f3e73f6f9"}]},"branch":"refs/heads/master"},"889566aa9735cade0842af45b2e040c23a60c8a1":{"kind":"REWORK","_number":4,"created":"2024-08-26 23:34:43.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/40/926940/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/40/926940/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/926940/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/926940/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/926940/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/40/926940/4"}}},"commit":{"parents":[{"commit":"54e4ebe02d7034d77732beba926aae5cb172a308","subject":"Merge \"Test metadata query over IPv6 only network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/54e4ebe02d7034d77732beba926aae5cb172a308"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-08-22 18:08:29.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2024-08-26 23:34:36.000000000","tz":-240},"subject":"Implement nested snat validation test scenario","message":"Implement nested snat validation test scenario\n\nIt\u0027s disabled for ovn because it\u0027s not, yet, implemented for the driver.\n\nSee: https://review.opendev.org/c/openstack/neutron/+/926495\n\n(The feature also requires special configuration for the driver that we\nprobably won\u0027t enable in the gate. This can be discussed though.)\n\nRelated-Bug: #2051935\nChange-Id: Ie2e49a53857009446e22300e2fff292355cef058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/889566aa9735cade0842af45b2e040c23a60c8a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/889566aa9735cade0842af45b2e040c23a60c8a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
