)]}'
{"id":"openstack%2Fneutron-tempest-plugin~927103","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I37c1a3855cb5c37aadec5cda26c89cd3f99e074f","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2076328","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-08-26 11:35:24.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-08-28 23:08:15.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-08-28 23:08:15.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-08-28 23:08:15.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-08-28 18:51:54.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I37c1a3855cb5c37aadec5cda26c89cd3f99e074f","subject":"[pagination tests] Do not fail on zero resources","status":"MERGED","created":"2024-08-26 11:14:16.000000000","updated":"2024-08-28 23:08:15.000000000","submitted":"2024-08-28 23:08:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"927103-bug/2076328","meta_rev_id":"e01d89554e1116b6bbcc06620eeb8f8f70d30042","_number":927103,"virtual_id_number":927103,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-28 23:08:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-08-26 11:35:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-08-28 18:51:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-08-28 18:51:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-26 11:28:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-08-26 13:43:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-27 04:19:27.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-08-27 04:19:27.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-08-27 04:19:27.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"62e5c2a15133b96c0176182d3f6e08946d3fc194","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-08-26 11:14:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8836edfaec8cbd34e61fb12bb3e45e4f000d86e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-26 11:28:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7034addc7e8db428ecf6f3eb52c3147823a8f05f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-08-26 11:34:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"6cdbc1c64d6c731aa9bf51be6a0cd570db2bc5ea","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-08-26 11:34:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f89c21a3b04fa78f30f8d433c79e02ff466f3aa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-26 11:35:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ee88901d31c1439253baef7c3ccedc513246e04e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 13:43:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1a2379f359f4413bbb08c9914a9c352\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4356da67b5448cfa6daa858c64e8886 : SUCCESS in 3m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b52a7e6ffb26467c9ddbadfbd6c3dd77 : SUCCESS in 1h 23m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/90cb48fce96a4f9bba719c0f3a391386 : SUCCESS in 1h 55m 20s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/f9858b56cc0847e3aae8de0d594e26f5 : SUCCESS in 1h 32m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7d803c931def418eaad7251ace0f0f6d : SUCCESS in 2h 06m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/13a00b7fde5245feb104de58e42390db : SUCCESS in 32m 28s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/4f45f02fffa04e9fbbfc2ed1ecf65379 : SUCCESS in 1h 45m 07s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/c507a7b6dbaf45be97127d19dd3ff376 : SUCCESS in 1h 31m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/73edec3453114298bf166a181f6472e4 : SUCCESS in 1h 29m 53s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/9a530e5faef54950bb1cbb9af99523a7 : SUCCESS in 1h 23m 39s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/5d898f0ba74549d985dccd239e15335d : SUCCESS in 31m 17s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/3c783e630bf341dd94008bee9066465e : SUCCESS in 1h 10m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/843471c3bcc34ced8bd6241ae20c8d10 : SUCCESS in 1h 41m 53s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/8a6cf57cdee34e1aae12bbaeede3d5fb : SUCCESS in 1h 37m 49s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/71397ac4db874fcc9c6470228fd7734e : SUCCESS in 30m 54s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/11c079f289b3465f8fbd0542120ab971 : SUCCESS in 1h 04m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/382e478a040a47f9a0d4d8d1ebfef73d : SUCCESS in 1h 25m 51s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/8cd33861ab3a4c2e878fb3d9204a4f36 : SUCCESS in 1h 32m 59s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/545d9e11a31042b29d123a93e3a29260 : SUCCESS in 31m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc6f54a881094987bf76d3a9cc9aa63f : SUCCESS in 3m 05s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/e492aa39b0034e4fa0f51ac32280369d : SUCCESS in 1h 38m 51s (non-voting)\n- neutron-tempest-plugin-sfc-2023-1 https://zuul.opendev.org/t/openstack/build/531b694580c24b32b243b2ad9ae41613 : SUCCESS in 1h 36m 54s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/d5829e3e034740c99fe9aabdf1c48a4b : SUCCESS in 1h 26m 05s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/bd91541cba7e4f669cc9bb3eda8fdb77 : SUCCESS in 1h 31m 27s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f3c6ac31f33642189bb6fdb9f578d462 : SUCCESS in 1h 42m 50s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/64ee78061ae64dfe9b58c394a2cdb7c0 : SUCCESS in 1h 19m 16s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/77b2293205004aac9bd4b9fbd577171f : SUCCESS in 1h 32m 36s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/a8d412eb4d804e78bd510077955d5f79 : SUCCESS in 1h 49m 48s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/f81c7a20c7804b3c901e9edfac42ee97 : SUCCESS in 53m 10s\n- neutron-tempest-plugin-dynamic-routing-2023-1 https://zuul.opendev.org/t/openstack/build/98c9c197517f47daa748083a3a6efcf1 : SUCCESS in 58m 31s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/0d4745cbd6184fdeb98fd2d2257f27e9 : SUCCESS in 1h 06m 58s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/7cd6f435519d4acc82c2e088b7a17dab : SUCCESS in 54m 14s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/8d51228125624eeca27d2729af87cce1 : SUCCESS in 47m 00s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/b13553989ff84b76bb290dd51de25322 : SUCCESS in 46m 03s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/4e90319b2d2d4a47a44dd283434cfd14 : SUCCESS in 57m 34s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/d6f2919b6b2b4422a39f86a185622b99 : SUCCESS in 57m 47s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/f957e1de8227459288833eeb12aeb0be : SUCCESS in 57m 40s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/a44f897c1f9043b49772360662c21e09 : SUCCESS in 53m 12s\n- neutron-tempest-plugin-vpnaas-2023-1 https://zuul.opendev.org/t/openstack/build/4574f2b357274fdb9ed3aad6437662f0 : SUCCESS in 48m 34s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/419fb246d31b4bdd82f854cb5006b881 : SUCCESS in 52m 36s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/4cea874fb1cc490ca92d1fb45ea9ae5f : SUCCESS in 37m 34s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/0f62674d098849cba6b075adb2d0020b : SUCCESS in 1h 32m 23s\n- neutron-tempest-plugin-tap-as-a-service-2023-1 https://zuul.opendev.org/t/openstack/build/828253ac461845a696c4ae97606712ab : SUCCESS in 1h 10m 45s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/ad462517c9a547dfbc69dd148fca81ee : SUCCESS in 1h 18m 31s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/3f471507e18e42ef81769af8d6445976 : SUCCESS in 1h 26m 39s","accounts_in_message":[],"_revision_number":2},{"id":"0d9ed797b84752989c7ca1a615e0717f0f5cd43f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-08-28 18:51:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"460a1de7c78e485bb1d83f91077dbef499f312cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 18:53:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ddc1b4390691f7b5111007ec54d9458cdd9e4fcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 23:08:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb690ec3996949db9a3c35668641ba72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c4b3efcd8be4167b809fe85b868ec97 : SUCCESS in 3m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8ee2f0c5889547d8a57ec14cb52fbd09 : SUCCESS in 1h 19m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8d8be16b126f49cd946e4a3bf0402ca0 : SUCCESS in 2h 01m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/966a3c2dc8484d199d49189bc3eec53b : SUCCESS in 1h 45m 45s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4f78e285e4af40a18af89fe6889149e5 : SUCCESS in 2h 00m 55s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/40bf297f8427402f80b7abdcbf3ed63d : SUCCESS in 1h 59m 31s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/b91274a3324e48ab815f97638071c95f : SUCCESS in 2h 08m 05s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/9110b2cfda1a43c19f026c7bc316660b : SUCCESS in 1h 34m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/244ea4a3a4f6443da85748ff1ec2cff0 : SUCCESS in 3m 06s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/4d7c99f2599141ec9d2da12fd07b4cd9 : SUCCESS in 1h 48m 02s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/f2ada7d207234de28a1f2b7a21fb7a84 : SUCCESS in 1h 32m 47s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/f31a1ba2b75d4a398c1175458ad25253 : SUCCESS in 53m 57s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/29c636b8ae424e70a65d55eb85ac4d1b : SUCCESS in 39m 11s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/26b8ab4ec4244a00a859321e5438929b : SUCCESS in 44m 26s","accounts_in_message":[],"_revision_number":2},{"id":"e01d89554e1116b6bbcc06620eeb8f8f70d30042","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 23:08:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a8c221ca48d69d8df74713dcbf589559dab62496","revisions":{"84767524f2a3a59882579c5975b5c03084baa280":{"kind":"REWORK","_number":1,"created":"2024-08-26 11:14:16.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/03/927103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/03/927103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/927103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/927103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/927103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/927103/1"}}},"commit":{"parents":[{"commit":"20dec4e823d8217c45f2c7f7891e5b78ec04596e","subject":"Merge \"Switch networking-sfc job to be run with eventlet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/20dec4e823d8217c45f2c7f7891e5b78ec04596e"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-08-26 11:10:38.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-08-26 11:13:24.000000000","tz":330},"subject":"[pagination tests] Do not fail on zero resources","message":"[pagination tests] Do not fail on zero resources\n\nSince the tests running in parallel, when resources\nare deleted by concurrent tests, GET call can return\n0 resources, adjust tests to not fail when there are\n0 items returned.\n\nThis is in continuation of previous patch[1].\n\n[1] https://review.opendev.org/926201\n\nRelated-Bug: #2076328\nChange-Id: I37c1a3855cb5c37aadec5cda26c89cd3f99e074f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/84767524f2a3a59882579c5975b5c03084baa280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/84767524f2a3a59882579c5975b5c03084baa280"}]},"branch":"refs/heads/master"},"a8c221ca48d69d8df74713dcbf589559dab62496":{"kind":"REWORK","_number":2,"created":"2024-08-26 11:34:33.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/03/927103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/03/927103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/927103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/927103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/927103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/03/927103/2"}}},"commit":{"parents":[{"commit":"20dec4e823d8217c45f2c7f7891e5b78ec04596e","subject":"Merge \"Switch networking-sfc job to be run with eventlet\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/20dec4e823d8217c45f2c7f7891e5b78ec04596e"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-08-26 11:10:38.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-08-26 11:34:10.000000000","tz":330},"subject":"[pagination tests] Do not fail on zero resources","message":"[pagination tests] Do not fail on zero resources\n\nSince the tests running in parallel, when resources\nare deleted by concurrent tests, GET call can return\n0 resources, adjust tests to not fail when there are\n0 items returned.\n\nThis is in continuation of previous patch[1].\n\n[1] https://review.opendev.org/926201\n\nRelated-Bug: #2076328\nChange-Id: I37c1a3855cb5c37aadec5cda26c89cd3f99e074f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a8c221ca48d69d8df74713dcbf589559dab62496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a8c221ca48d69d8df74713dcbf589559dab62496"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
