)]}'
{"id":"openstack%2Fneutron-tempest-plugin~927199","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I0b8f856b68397edcbe1244856fdd99e22cf6f90f","project":"openstack/neutron-tempest-plugin","branch":"master","hashtags":[],"change_id":"I0b8f856b68397edcbe1244856fdd99e22cf6f90f","subject":"DNM: introduce a config knob to determine if nested snat is on","status":"ABANDONED","created":"2024-08-26 17:59:31.000000000","updated":"2024-08-26 23:21:50.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c00f5a3b86e6a97428ef61ec579f1e7e279bfbbb","_number":927199,"virtual_id_number":927199,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-26 22:27:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-26 22:27:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0576c5865a10cb87b07daee7249dd349e960844","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-26 17:59:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37f69e873d885617e19dce3b74ac2f3fc337d02b","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-26 19:41:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a407765fe09bb3fc5da3eba23ff58ab4da976f1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-26 20:01:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d7f987a2bae005fb32062bd40b51b567278fff13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 22:27:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce927bebb1514b989701d70c3b1ec4d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6db79a5f580a480cacf5b178ed6f6d83 : SUCCESS in 4m 32s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/46103ec644d54ec1a84424676a572c9f : SUCCESS in 1h 41m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/966687572bfe468f91ff9e78ae6ee6c3 : SUCCESS in 2h 18m 54s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/91a54bb606e648029dc370d152ee066f : SUCCESS in 2h 03m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8b8f9b8e55cd43aa8c1d00330fb82e09 : SUCCESS in 2h 22m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eb41b66b4ff7413bac7d77e6acdd061f : SUCCESS in 43m 38s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/9a59b7e8b3e84cc4a2754befca61875b : SUCCESS in 1h 58m 12s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/73a7730b3566498ab3fa7e1bdaa1f589 : SUCCESS in 1h 35m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/7b5712ab5f584ad3b5274faf8fe5f675 : SUCCESS in 1h 49m 00s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/bb677ab0c3e3436c98e3a55790c8ba48 : SUCCESS in 2h 17m 26s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/9be136307f4440cbaedd57d5f8851866 : SUCCESS in 43m 00s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/b22465bea8144313b9c748d97e9d992f : SUCCESS in 1h 18m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/e97b1ef8d320476785ff29b5b98235f1 : SUCCESS in 1h 21m 27s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/b3ad91913f084ce283ac79098ae87416 : SUCCESS in 1h 19m 56s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/890e1902d99f4e2bb0fa026711b2c8a3 : SUCCESS in 33m 08s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/f45e5c5b703a4aff90864fb40767d0b4 : SUCCESS in 1h 46m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/7b6c97e68f7a410aaad3396bace76a5a : SUCCESS in 1h 17m 10s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/aae931ea587c4c36b7e87f1bdd3b316a : SUCCESS in 1h 46m 00s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/c87cba545e5948ea8bd3e211c6b86a8c : SUCCESS in 34m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4eacfe600e544d98a0a2e11f9ebb208 : SUCCESS in 3m 20s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/48c30e15f322464eb2e67fda66aaf858 : SUCCESS in 1h 25m 37s (non-voting)\n- neutron-tempest-plugin-sfc-2023-1 https://zuul.opendev.org/t/openstack/build/9d82c7085d114cfeb9f27e5b46ff99d1 : SUCCESS in 1h 13m 53s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/14368841a93e4d048291abbff4dab101 : SUCCESS in 1h 33m 42s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/8cab6508c4544bb097cfe1a4ff195fb1 : SUCCESS in 1h 14m 43s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/b23f4403ede54ff8a8fc227fcbceb3d9 : SUCCESS in 1h 46m 20s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/da7e47c989d4495e84442e41ada5db82 : SUCCESS in 1h 10m 28s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/cb04c62c45c44ab28f8713867146d5fc : SUCCESS in 1h 27m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/309dd87ef96c409d91454c5d81f2dca7 : SUCCESS in 1h 39m 54s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/a76e634ebb3d4af1ba46346eb449aa24 : SUCCESS in 52m 01s\n- neutron-tempest-plugin-dynamic-routing-2023-1 https://zuul.opendev.org/t/openstack/build/4265233bea30446b815d0ca2d2835877 : SUCCESS in 59m 25s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/5fc6bd88a5174c5a9f20d6d24987981c : SUCCESS in 1h 06m 16s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/dfe79497118c44468f7607ae77825656 : SUCCESS in 1h 00m 13s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/e9da39029ab64dde8bd8bd6e2a9dcb9f : SUCCESS in 52m 04s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/e3107a9ed06646868b1e22e6decdbf4b : SUCCESS in 47m 29s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/7d463e9d9b23454ea57caa784f73f0b5 : SUCCESS in 47m 30s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/e35efedb36e74b3380cc01a04e877ae9 : SUCCESS in 36m 06s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/c05c8b3358904d7a8b744c783556c0e9 : SUCCESS in 43m 48s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/b1a5e3dd5bc04e5c8d235c9af92dae84 : SUCCESS in 42m 21s\n- neutron-tempest-plugin-vpnaas-2023-1 https://zuul.opendev.org/t/openstack/build/e8e0d5727c4e4c349fea6980c7543470 : SUCCESS in 48m 31s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/4c0205e2bc3a4952a022e084062587d3 : SUCCESS in 40m 50s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/9747c248ea9242e08f6f945f51d3b25a : SUCCESS in 53m 14s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/833b4de43e3640c5bab00d80298e1638 : SUCCESS in 1h 27m 58s\n- neutron-tempest-plugin-tap-as-a-service-2023-1 https://zuul.opendev.org/t/openstack/build/dfa456eaf92049d192b49b50b290d84a : SUCCESS in 1h 32m 30s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/adf5f6b6c61846cb915ef2f9e91dae73 : SUCCESS in 1h 07m 36s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/cec75573a38a41ed9d3738a703efa2a4 : SUCCESS in 1h 25m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c00f5a3b86e6a97428ef61ec579f1e7e279bfbbb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-08-26 23:21:50.000000000","message":"Abandoned\n\nSquashed into the parent patch.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fa5e80b56366ef40d8d28e5c4f4ca5b5aad67fbc","revisions":{"a710f51a8d07f47f0fd061e4ea7997f4ad54badd":{"kind":"REWORK","_number":1,"created":"2024-08-26 17:59:31.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/99/927199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/99/927199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/99/927199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/99/927199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/99/927199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/99/927199/1"}}},"commit":{"parents":[{"commit":"d91e832a4db8ed065b1f0feb112ec85e44870717","subject":"WIP Implement nested snat validation test scenario","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d91e832a4db8ed065b1f0feb112ec85e44870717"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-08-26 17:57:49.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-08-26 17:59:01.000000000","tz":0},"subject":"DNM: introduce a config knob to determine if nested snat is on","message":"DNM: introduce a config knob to determine if nested snat is on\n\nml2/ovs and ml2/linuxbridge should implement it already. ml2/ovn\ndoesn\u0027t, yet (and will require special configuration anyway.)\n\nPosting a DNM on top to not disrupt other check jobs in gate. Will\nsquash in later.\n\nChange-Id: I0b8f856b68397edcbe1244856fdd99e22cf6f90f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a710f51a8d07f47f0fd061e4ea7997f4ad54badd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/a710f51a8d07f47f0fd061e4ea7997f4ad54badd"}]},"branch":"refs/heads/master"},"fa5e80b56366ef40d8d28e5c4f4ca5b5aad67fbc":{"kind":"REWORK","_number":2,"created":"2024-08-26 20:01:24.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/99/927199/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/99/927199/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/99/927199/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/99/927199/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/99/927199/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/99/927199/2"}}},"commit":{"parents":[{"commit":"d91e832a4db8ed065b1f0feb112ec85e44870717","subject":"WIP Implement nested snat validation test scenario","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d91e832a4db8ed065b1f0feb112ec85e44870717"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-08-26 17:57:49.000000000","tz":0},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2024-08-26 20:01:10.000000000","tz":0},"subject":"DNM: introduce a config knob to determine if nested snat is on","message":"DNM: introduce a config knob to determine if nested snat is on\n\nml2/ovs and ml2/linuxbridge should implement it already. ml2/ovn\ndoesn\u0027t, yet (and will require special configuration anyway.)\n\nPosting a DNM on top to not disrupt other check jobs in gate. Will\nsquash in later.\n\nChange-Id: I0b8f856b68397edcbe1244856fdd99e22cf6f90f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fa5e80b56366ef40d8d28e5c4f4ca5b5aad67fbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/fa5e80b56366ef40d8d28e5c4f4ca5b5aad67fbc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
