)]}'
{"id":"openstack%2Fneutron-tempest-plugin~928461","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I88423b89f30ef2c4cb4738bbee1246e019796d73","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"log-on-error-sg-test-connection","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-09-10 00:51:22.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2024-09-09 18:09:55.000000000","reason":"removed on reply"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-09-10 00:51:22.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-09-10 00:51:22.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-09-09 16:19:57.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I88423b89f30ef2c4cb4738bbee1246e019796d73","subject":"refactor: Remove unused code from utils and tests","status":"MERGED","created":"2024-09-06 17:07:19.000000000","updated":"2024-09-10 00:51:22.000000000","submitted":"2024-09-10 00:51:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928461-log-on-error-sg-test-connection","meta_rev_id":"a480c71c395048f40b2989b661960bc9e76eae73","_number":928461,"virtual_id_number":928461,"owner":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-10 00:51:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"disliked":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":-1,"date":"2024-09-09 18:09:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-09-09 16:19:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2024-09-09 16:09:45.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-06 18:24:45.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-09-06 19:46:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-06 21:10:50.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2024-09-09 07:34:25.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-09-09 13:36:09.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"2eb12608c1b13e3e02f3e1fef6d263bca14f768d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-09-06 17:07:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3b28ff662a0db61f2dbe2e3e142714cacb1d5ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-09-06 17:27:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"43c09b7b682c94c61b90cd696bd91ea6951a824c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-09-06 18:24:45.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fb1038298dc8389fc0f94adb642df4704e9d89c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 19:46:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f20324e690048babdaa61668273eb36\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a2e24dc3a49462e854530379e7dec1e : SUCCESS in 2m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e9b8d4102c364881b2ab5df4061b3b75 : SUCCESS in 2h 15m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ac4ad8af84744fea949fed29c8239c02 : SUCCESS in 2h 08m 39s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/6605cb9076cc4e878e5e1382751f4189 : SUCCESS in 1h 58m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/10658835ac2648a7bf2df40298a6fcdd : SUCCESS in 2h 16m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/19aabe0841b2485cb71471337bc34202 : SUCCESS in 43m 53s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/fc64fce178b34d889153c635f450383e : SUCCESS in 2h 14m 27s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/9741384288e645ba8e78e624b27b0870 : SUCCESS in 1h 41m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/18419d1078834eefafd8c3d1765ca812 : SUCCESS in 1h 44m 25s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/358ed6796bc64d2aa42be2e3f9259522 : SUCCESS in 1h 31m 48s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/4f6941fe588a41eaa059b1d47953c6e3 : SUCCESS in 46m 23s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/bba0560805dd4037aa2f5f888dd8c218 : SUCCESS in 1h 09m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/c444074b4a8c4a81821e294be061ae24 : SUCCESS in 1h 26m 33s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/1a588a5924c142238201b6d641360b3d : SUCCESS in 1h 33m 05s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/edd780b77e2544eb820f545bf4caf7ae : SUCCESS in 29m 32s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/95feb19a52464258919989d800aabf05 : SUCCESS in 1h 55m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/b532c7f443a544ea8e7fb1d17a80b819 : FAILURE in 2h 09m 00s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/f5595dc1d6a74e669b11da9cc1c4562a : SUCCESS in 2h 00m 40s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/3b8f65fab8e34c8abfa79678405832a1 : SUCCESS in 49m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aee68fb3c38d4f349bff80473ea74fb0 : SUCCESS in 2m 57s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/83ebbd45fc5e419e9ca80a91c33da7ca : SUCCESS in 1h 47m 31s\n- neutron-tempest-plugin-sfc-2023-1 https://zuul.opendev.org/t/openstack/build/852200fc37574efbacf4019a5257340b : SUCCESS in 1h 34m 25s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/6d514a5cbc934093a1e7c2a81ee3c6b5 : SUCCESS in 1h 00m 06s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/5c01ca6226594a6a9028d00542cb1792 : SUCCESS in 1h 38m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/4949ee1aa45945068b16bfc996ae2fe2 : SUCCESS in 1h 36m 52s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/e53ec522cef342ffb9d5f03ed966044f : SUCCESS in 56m 57s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/c44cfe08ffc34375a69f7cfe58ab36f8 : SUCCESS in 1h 30m 37s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/2d6f67dd8e65422f8f428111c5582c30 : SUCCESS in 1h 26m 43s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/45e67cd63a224b05806880eb42877832 : SUCCESS in 41m 54s\n- neutron-tempest-plugin-dynamic-routing-2023-1 https://zuul.opendev.org/t/openstack/build/d9d88068e6bf4e1f8b86577e16be500a : SUCCESS in 1h 02m 36s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/56501b630c9f4596b97a99dd9f6267c6 : SUCCESS in 45m 51s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/9377a5ea166a4eda83e30a02ff45cf45 : SUCCESS in 44m 08s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/dd1e70a5534b41318e97922dd75d09df : SUCCESS in 31m 50s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/19d4c8d089894b6ea683f6bee17a3852 : FAILURE in 30m 40s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/867d068960fc4df8ab7fbd36f2aaf6b0 : SUCCESS in 33m 30s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/3bca8e29f7f842b1af9302b5657f6e19 : SUCCESS in 29m 42s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/29ea67985460496c9fc557f20764605e : SUCCESS in 45m 17s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/8491b4634bbd401abbd53034c013080d : SUCCESS in 34m 20s\n- neutron-tempest-plugin-vpnaas-2023-1 https://zuul.opendev.org/t/openstack/build/e65b368b709c40729ea71f0d71804954 : SUCCESS in 48m 04s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/4e00c275ff7a40deb0c2017deb87eb4d : SUCCESS in 45m 32s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/5a973b56c3bf43fb9a0be2d4ffdef9b0 : SUCCESS in 46m 52s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/7f98d006fc84400697cf40155862028d : SUCCESS in 1h 29m 16s\n- neutron-tempest-plugin-tap-as-a-service-2023-1 https://zuul.opendev.org/t/openstack/build/93e273fed8d8492a80b304ba2b62ce85 : SUCCESS in 1h 17m 26s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/9218a95b397547919429684d42c4ec6c : SUCCESS in 1h 23m 53s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/d0bc47e6f75b4fb189c4d929111fb62c : SUCCESS in 1h 18m 19s","accounts_in_message":[],"_revision_number":2},{"id":"4ac8ae86f2b67c2acecc181b1c2672327087ee9e","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-09-06 21:10:50.000000000","message":"Patch Set 2: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/c3a86e8358e74093a13270e12cfdc817\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/d918a0edea334075a63d0cae44e1f3ed : SUCCESS in 3h 42m 39s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/298d12b5515d4aa19ac3949a3e0e00b7 : FAILURE in 2h 41m 56s","accounts_in_message":[],"_revision_number":2},{"id":"14d21a93498cf5e72ebc575e61f5ca4b69392226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-09-07 00:04:51.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"099a76280f8a1425b5396b53f7d659633f54a000","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-09-07 01:00:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e7eed530851898711b9255f8398dd2871f166c9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-07 01:59:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b79bb59ac84414589d7bc16e8e6603d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f23f2dcef954194ade1dfc140faed5f : SUCCESS in 2m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/480826493de1486a8a1987d6f67edee4 : SUCCESS in 1h 36m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c070c224d89543efb4daf73633f20dff : SUCCESS in 1h 41m 55s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/627339d6851643cdb8d3fa8137e17214 : SUCCESS in 1h 50m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/307d073ef6534f2fbc9c94e6c61a1314 : SUCCESS in 1h 53m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d5c43fb8873949238713b5db26a98bbd : SUCCESS in 37m 18s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/8019f7a7e3e34cdab969facc0d0ebc0d : SUCCESS in 1h 35m 32s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/72ab3065d7d1439f95a434bf1e0715d2 : SUCCESS in 1h 12m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/3ddb81fa51524a0181f2c1c0ce9941c6 : SUCCESS in 1h 11m 00s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/5da8b4256e6f4151ab01a9b07e0ad3b5 : SUCCESS in 1h 29m 39s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/589562a326fc40f4a54f80e4251722ef : SUCCESS in 32m 22s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/9e994f2f3ad84b30ba25da1c1295d06e : SUCCESS in 1h 05m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/346b641586234bd6bd4c3971bea028e0 : SUCCESS in 1h 28m 30s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/9aa0e1d3cc6a4fa794d54f374d105b79 : SUCCESS in 1h 50m 26s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/0afa6f7f96294b1e88d0398a52b0d9f3 : SUCCESS in 31m 16s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/62fa4f1e3ef24183a08901cb1d02579c : SUCCESS in 1h 10m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/0f58b59c99ac44c0a3ab967e5a6339b3 : SUCCESS in 1h 44m 34s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/4d7e563779a542b7b38686add455ef9e : SUCCESS in 1h 34m 26s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/9f787c1f493e44dda125b404b0cca5a7 : SUCCESS in 32m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfce52091b3949af8e63cf7dfdaadb50 : SUCCESS in 2m 46s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/132bfe766132446e8085477486149133 : SUCCESS in 1h 49m 59s\n- neutron-tempest-plugin-sfc-2023-1 https://zuul.opendev.org/t/openstack/build/3509e61d4cd640248a9b11a162ec87be : SUCCESS in 1h 02m 25s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/cbd4bb6f810d49a19c4fb36246eba46f : SUCCESS in 1h 35m 06s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/0bf31d979bd747cdb66e21c6d244903f : SUCCESS in 1h 32m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/534e66793ce34b0ea636784fb0f0ddb6 : SUCCESS in 1h 18m 59s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/6cb7e5abb7064aad949052408a7531a7 : SUCCESS in 1h 06m 19s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/5b584e78781c4c7aa2a0f688f37c85d2 : SUCCESS in 1h 02m 43s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/9c3750894008467fb403bf00d243bb1b : SUCCESS in 1h 11m 03s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/e69464b2f7ae45029e28020af0a22400 : SUCCESS in 42m 33s\n- neutron-tempest-plugin-dynamic-routing-2023-1 https://zuul.opendev.org/t/openstack/build/9736d17190c7417cb1f9adf9cb02073a : SUCCESS in 42m 00s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/01acbfc37d89493ca010a5db06eb1d1e : SUCCESS in 43m 35s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/f36624ee07a245aeaa1c08ad52fe0da5 : SUCCESS in 56m 30s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/5c29f87464164dc4b66f56d8e8b2c0b9 : SUCCESS in 29m 09s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/5ab1d328cf404a3a82e9c0a443026bcd : SUCCESS in 43m 13s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/4331200f36814e8c904334aaf82bad15 : SUCCESS in 42m 25s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/d5293c5ac2964c9f9082f4ae9dacf211 : SUCCESS in 30m 56s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/25f1ab60f865412590fa505e382e3e11 : SUCCESS in 44m 43s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/02b123dc88e442db816a98f488c187c3 : SUCCESS in 35m 00s\n- neutron-tempest-plugin-vpnaas-2023-1 https://zuul.opendev.org/t/openstack/build/b4976f12919a4cef9e631c28cbdc1f26 : SUCCESS in 52m 20s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/72f1af3af25c4633b6713020e4739dba : SUCCESS in 33m 53s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/436e4299d1c640d7bb9aa73e6d86a7d6 : SUCCESS in 47m 39s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/d447930276f44a4290f04dd077a2a278 : SUCCESS in 1h 12m 57s\n- neutron-tempest-plugin-tap-as-a-service-2023-1 https://zuul.opendev.org/t/openstack/build/c88d1fe9fbbb40f1b00970d4cf5e4520 : SUCCESS in 1h 17m 58s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/d64fdc5285654cadbbd922cce8cddefe : SUCCESS in 57m 37s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/a2be7bd0d02a4c7781714be4713f3789 : SUCCESS in 1h 24m 19s","accounts_in_message":[],"_revision_number":3},{"id":"6f59dd9bf8bbc9f03ac7e9cf1f029905c00f2023","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-09-07 04:26:14.000000000","message":"Patch Set 3: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/3d3c53dfd56e4a6ebd8aff1e9f5d7433\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/139fb83ef7b9419fb8129e9295f30c01 : SUCCESS in 4h 20m 24s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/21ca3a6438934105b2695c5662413bf6 : FAILURE in 3h 04m 04s","accounts_in_message":[],"_revision_number":3},{"id":"42f544715a31ef1c074099598766f2a5b962749f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-09-09 00:11:09.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1899900f229c3405877e0353e552bedc04f1b763","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-09 07:34:25.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"afbfc9f78ed20fe9869d838d9946afee53c1895e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-09-09 13:36:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c827a3bfa50d2b44d5a9006bb319d25fc7735689","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 15:51:54.000000000","message":"Patch Set 3: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron-tempest-plugin for 928461,3","accounts_in_message":[],"_revision_number":3},{"id":"2e63e22dd407f8523716d182691428b0e9c9afc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-09-09 16:02:53.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"9009981ea3201b4230332348271a9ffefbb4438e","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-09-09 16:09:45.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"defa11f56b0069aff27a51ae6d16ec73198b8f69","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-09-09 16:19:57.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"71eb5e7d7857ede596b9d9a14f466b3f5a2d9b58","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-09-09 18:09:55.000000000","message":"Patch Set 4: Code-Review-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/aec2a39571bd4ee796d1025676f892d2\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/c644b8b91d784ee8aef2bb4d4b799232 : SUCCESS in 2h 01m 29s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/09e7eb2ebad1485fbaa3d672f02ba20f : FAILURE in 18m 30s","accounts_in_message":[],"_revision_number":4},{"id":"f6be1e2370244c36d088a2f99a4205d2b10ec54c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 18:10:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c21a30b2dc104d85b422ae25f9c371e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5046a3d9e16c47c2baf94a69c81cff91 : SUCCESS in 3m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3dcd18d2ac63484180924e40e0d5c9ec : SUCCESS in 1h 42m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bb209f4eeeef4abeaf22dc3236e3210e : SUCCESS in 1h 40m 40s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/801efc7d803e4e9b9b1660ed5c717b07 : SUCCESS in 1h 16m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/53e2632f93b54ed5ad592d59068a951b : SUCCESS in 1h 51m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6c8c56a3bfb14123afbddc6fbf40de5d : SUCCESS in 33m 46s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/bc398f2f887648a8bba830eefd9ff1c3 : SUCCESS in 1h 20m 17s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/262737a28fcc4d97a64241e1e3244ae2 : SUCCESS in 1h 19m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/fb1af711e37745a0974cc5a9f8daff18 : SUCCESS in 1h 20m 27s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/81fbeb761aa3445bba35495fd3427a19 : SUCCESS in 1h 33m 07s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/7d1d076651974f38b4c73faf130bfa1f : SUCCESS in 32m 10s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/36538a2d790e494c92b08a75eebe9eff : SUCCESS in 1h 06m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/b2b5a83889c14070a1eb1e65243328ec : SUCCESS in 1h 51m 24s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/aeb48d44331247598588f526580e948f : SUCCESS in 1h 23m 52s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/73ae22f9c7c04e70a3d1a40390489f51 : SUCCESS in 37m 46s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/eba8acd476554b84a0fbd99f91df296e : SUCCESS in 1h 36m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/a5766d8fa5be4f99b6ea335ca0ea4c1d : SUCCESS in 2h 00m 05s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/c92ca993062a480f861e43fba45a39ed : SUCCESS in 2h 05m 11s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/25d23fd9f5ef43cf8edf261bcea01828 : SUCCESS in 39m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd87bd50b4234cdba3cd0ee952912a26 : SUCCESS in 3m 48s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/f1ae5592998f49cb85cd9d8e6116cc58 : SUCCESS in 1h 39m 43s\n- neutron-tempest-plugin-sfc-2023-1 https://zuul.opendev.org/t/openstack/build/235d252af62e4372b2aaa1b27611d411 : SUCCESS in 1h 24m 29s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/43d0803dc31447aba97ec87ef203bd21 : SUCCESS in 1h 09m 42s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/5cf2d9a3ff7e46e59a1f4229e0444193 : SUCCESS in 1h 19m 05s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/6604d61f3b7942e1beefefcf5ae7a423 : SUCCESS in 1h 12m 21s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/7c3da9d731fc4dc5a930ceac7e25abb5 : SUCCESS in 57m 36s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/9e0bd90ab03d425fb618bbea0a2983f9 : SUCCESS in 1h 13m 20s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/2f83c3059eb547689ec513367fff17d6 : SUCCESS in 1h 21m 18s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/2a9a91d5d32f41809a0ce2afdddd105e : SUCCESS in 36m 50s\n- neutron-tempest-plugin-dynamic-routing-2023-1 https://zuul.opendev.org/t/openstack/build/d1a9d8e2a4864f3ca49df07b5e1e93d6 : SUCCESS in 52m 31s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/76c1e0fa739a44afbb5a8397a4bc895c : SUCCESS in 58m 57s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/08bae24fcfe1460b96f82872d7f3c3e0 : SUCCESS in 40m 29s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/1bb9ee14a5064465b6602c5175a9504e : SUCCESS in 41m 29s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/ee2a8005dbd94bd4a16efcc04b576240 : SUCCESS in 50m 56s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/db8ffd1d1b0a423fa5e9a33f7be5d420 : SUCCESS in 41m 45s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/64557f4f7c30448ba209345dd7cd03c0 : SUCCESS in 52m 20s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/5e965fe8271c42d7bbbc131f138337bc : FAILURE in 49m 31s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/d8e98cdd221a44fba6aa69211c186da0 : SUCCESS in 53m 05s\n- neutron-tempest-plugin-vpnaas-2023-1 https://zuul.opendev.org/t/openstack/build/c9ec35ac22cc402188a3d2a6215f3543 : SUCCESS in 45m 46s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/1cb2d0ad08334f5caa2b76dc85bae886 : SUCCESS in 51m 29s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/b24f04e531dd418ebedd311f2b52a7f5 : SUCCESS in 45m 47s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/6a01b2c803274002b0a8733e299070b4 : SUCCESS in 1h 40m 43s\n- neutron-tempest-plugin-tap-as-a-service-2023-1 https://zuul.opendev.org/t/openstack/build/9ed1b8b0efcd482ca1b6e0258c7497db : SUCCESS in 1h 29m 15s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/2342546c72084af7ba9022bf148507b9 : SUCCESS in 1h 17m 14s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/1f986bf03cc2483f9c583c698b0a04f5 : FAILURE in 42m 35s","accounts_in_message":[],"_revision_number":4},{"id":"a094e4863f1ed837e5396507d2afafc7e6bdac3d","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-09-09 20:20:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3fd2736c8949e2be0ff0181a07097f351cf17407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 22:40:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9f3ef2dba3d44ae87225e87d3d278b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffc586898efc46e08ee0fc326435607b : SUCCESS in 4m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a8d6a29bedfb48bf8cf81f1ac93870dd : SUCCESS in 1h 39m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e234c51b9bc74788a0e85d013dc4cf61 : SUCCESS in 2h 12m 13s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/b8be3506484e4d1eb0cedb93a7b99599 : SUCCESS in 1h 35m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/331873d3ee2c4d5aac793522f8dcb07e : SUCCESS in 2h 04m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/54e7ec5824f34e74beb8adc67ff5c6f5 : SUCCESS in 34m 53s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/c97a2e8968fa4d318d4b656358be05c3 : SUCCESS in 1h 49m 56s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/7e92e24237d44cdd8c7a7c37c725b774 : SUCCESS in 1h 23m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/503501fe8ac14a09915d4992bcd37627 : SUCCESS in 1h 52m 43s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/9aa4e21cd5a04861bf29eb8b11d43a23 : SUCCESS in 1h 34m 38s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/4da4635857d34c0588a30e1fbcb69d03 : SUCCESS in 31m 01s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/859e4032a9d248a2b28fdbf5dc51d719 : SUCCESS in 1h 28m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/57d543e04c874c54bb146022a22cb514 : SUCCESS in 1h 26m 36s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/089cec790745425c98c1a02ae4f3ff3c : SUCCESS in 1h 42m 23s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/da14c5565c3943919ae30f7fc1f1277a : SUCCESS in 31m 20s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/dfc78a2cec37454bba9f494739cba1e8 : SUCCESS in 1h 25m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/9a239beb8e12472bb42cbd232623bc4a : SUCCESS in 1h 54m 22s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/0ccd77b3071840129c5fc6b6a891ccd5 : SUCCESS in 1h 29m 49s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/5288e24868754482b7823fe116197dd0 : SUCCESS in 36m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf3cefd24644a6a93feb5dd3ec5a21a : SUCCESS in 2m 35s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/265744e7ae9b43188a39d8e789933008 : SUCCESS in 1h 46m 32s\n- neutron-tempest-plugin-sfc-2023-1 https://zuul.opendev.org/t/openstack/build/d5a723f763fd4a6ba45f5aff4f9732e9 : SUCCESS in 1h 05m 12s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/59be0f2da10d4d61b0723f8a173004d9 : SUCCESS in 1h 20m 37s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/4d5abc3e25a047ff9d45df049bcbafbf : SUCCESS in 1h 07m 15s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7ebfb5972da340f2a84a196ca50e6ead : SUCCESS in 1h 12m 00s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-1 https://zuul.opendev.org/t/openstack/build/37df133aa34a4408a4c73c826c96e2fd : SUCCESS in 1h 09m 49s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/9370141b40364a81a46db93068586f6f : SUCCESS in 1h 30m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/964a309fd09e4b1d9533325f07558618 : SUCCESS in 1h 09m 13s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/df2c2b0c80fe4ebd888c478d1c9da971 : SUCCESS in 59m 49s\n- neutron-tempest-plugin-dynamic-routing-2023-1 https://zuul.opendev.org/t/openstack/build/d352872edf8d4392bd2b82f2e63642a4 : SUCCESS in 43m 07s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/436d9e9047f948af869573f989f3344b : SUCCESS in 48m 46s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/785a1487ff294fa4987cafe781819ed5 : SUCCESS in 59m 21s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/473dd4b232324a78a94fb542ce8e790f : SUCCESS in 49m 23s\n- neutron-tempest-plugin-fwaas-2023-1 https://zuul.opendev.org/t/openstack/build/a500537681ba4229bad1dcfb43a109ff : SUCCESS in 36m 59s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/8e767f07f011408185ce37e100d0216b : SUCCESS in 48m 06s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/b8d5e26bde9242e1ac402229acec83cc : SUCCESS in 53m 15s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/38be6360cc1a420694bdf029f1770b0c : SUCCESS in 49m 36s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/42a51b378ab1465f9cd7f3e3e3b702b8 : SUCCESS in 54m 11s\n- neutron-tempest-plugin-vpnaas-2023-1 https://zuul.opendev.org/t/openstack/build/2838d987a9c749d9aabdd4bd10b724e6 : SUCCESS in 53m 40s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/45bfdfd0c9c04fde951ae925c92dccff : SUCCESS in 49m 01s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/62b858a99d0a49b293c21a7ddfda1ad1 : SUCCESS in 51m 56s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/0e893bebfe8f468fab8ed98713f3afda : SUCCESS in 1h 35m 34s\n- neutron-tempest-plugin-tap-as-a-service-2023-1 https://zuul.opendev.org/t/openstack/build/4e563f542abd47e190eef1aee3e0c9db : SUCCESS in 1h 04m 27s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/5dee21362c7e4edf9076757177fe4130 : SUCCESS in 1h 06m 07s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/72dc6d369fa548dfae82463eb31ad96d : SUCCESS in 1h 19m 22s","accounts_in_message":[],"_revision_number":4},{"id":"5d02aabf1b703a452e63ae405dbc6f3ce2d2e529","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 22:40:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7bfd9677d96f0c336fb1846995b340c5201813e7","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-09-10 00:35:01.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/47414f5c372649d2b9287ba914529b65\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/aea06ccd84804a5d8c5bcdce2d92da42 : SUCCESS in 4h 12m 54s\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/9a468bdcb14c4e44a0fdd7690c14fb76 : FAILURE in 2h 56m 07s","accounts_in_message":[],"_revision_number":4},{"id":"f707479df4c9b089e6339a8e194025ba4446e8bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 00:51:21.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/895e84e37a7f424185fddc7affe72000\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/204b2bb273d644f6a8193bab8cba1a77 : SUCCESS in 3m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/76a77cd8f2124947b0554749a34d33ef : SUCCESS in 1h 49m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2818a31e318f4b1c8c7de18ca91e28a6 : SUCCESS in 1h 59m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e5739df9b28946d2891126d0ed6b0b5b : SUCCESS in 1h 54m 27s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c42c3cbf92c24d859d6732b5224f1d9d : SUCCESS in 1h 44m 12s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/889b136764d84829aa8e213226267de9 : SUCCESS in 1h 46m 28s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/e480950e2faf4ad692f4424f8aa5f2bf : SUCCESS in 1h 34m 18s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/02887a3ad162448b942b648da1ebfee9 : SUCCESS in 2h 09m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f8529bada664498ba08d72c2a86e423 : SUCCESS in 2m 50s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/ffe74af998934a32a06c068b03f010d1 : SUCCESS in 1h 57m 04s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/65f7c416eccd493abb33e2e8344281a7 : SUCCESS in 1h 42m 13s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/85ad9ae9e1ab477c9795ca993a95cc3e : SUCCESS in 47m 23s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/8c7fe8ce66144357a3b20d56f102fd03 : SUCCESS in 31m 21s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/bb09a5a0295743de89f906a3574b897f : SUCCESS in 55m 47s","accounts_in_message":[],"_revision_number":4},{"id":"a480c71c395048f40b2989b661960bc9e76eae73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 00:51:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"86ddcd2fa5822c556e2e224c231cd82698535b15","revisions":{"dbdad8672d02c4d17264f39b3b8e4b8c1fc11e32":{"kind":"REWORK","_number":1,"created":"2024-09-06 17:07:19.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/61/928461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/61/928461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/928461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/928461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/928461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/928461/1"}}},"commit":{"parents":[{"commit":"1f538b55513fd21a54b037f58fadf18e071c5a6e","subject":"Merge \"Revert \"Switch networking-sfc job to be run with eventlet\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1f538b55513fd21a54b037f58fadf18e071c5a6e"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2024-09-06 17:06:48.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2024-09-06 17:06:48.000000000","tz":-240},"subject":"refactor: Kill _verify_http_connection (dead code)","message":"refactor: Kill _verify_http_connection (dead code)\n\nSince Iaca6fd3e6ed3c64ab3ca22817ad461479ecfa189, this is no longer used\nanywhere.\n\nChange-Id: I88423b89f30ef2c4cb4738bbee1246e019796d73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dbdad8672d02c4d17264f39b3b8e4b8c1fc11e32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/dbdad8672d02c4d17264f39b3b8e4b8c1fc11e32"}]},"branch":"refs/heads/master"},"e2882faadc9c8e67a96fefde2e211e96182b5240":{"kind":"REWORK","_number":2,"created":"2024-09-06 17:27:09.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/61/928461/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/61/928461/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/928461/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/928461/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/928461/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/928461/2"}}},"commit":{"parents":[{"commit":"1f538b55513fd21a54b037f58fadf18e071c5a6e","subject":"Merge \"Revert \"Switch networking-sfc job to be run with eventlet\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1f538b55513fd21a54b037f58fadf18e071c5a6e"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2024-09-06 17:06:48.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2024-09-06 17:27:00.000000000","tz":-240},"subject":"refactor: Kill dead code from utils and tests","message":"refactor: Kill dead code from utils and tests\n\nSince Iaca6fd3e6ed3c64ab3ca22817ad461479ecfa189, _verify_http_connection\nis no longer used anywhere. Its removal also exposes two more utils\nfunctions that are also not used anywhere.\n\nChange-Id: I88423b89f30ef2c4cb4738bbee1246e019796d73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e2882faadc9c8e67a96fefde2e211e96182b5240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e2882faadc9c8e67a96fefde2e211e96182b5240"}]},"branch":"refs/heads/master"},"f1c300ded225565913ea47ce8aa5af01e01f6b9e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-09-07 00:04:51.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/61/928461/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/61/928461/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/928461/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/928461/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/928461/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/928461/3"}}},"commit":{"parents":[{"commit":"1f538b55513fd21a54b037f58fadf18e071c5a6e","subject":"Merge \"Revert \"Switch networking-sfc job to be run with eventlet\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1f538b55513fd21a54b037f58fadf18e071c5a6e"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2024-09-06 17:06:48.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2024-09-07 00:04:51.000000000","tz":0},"subject":"refactor: Remove unused code from utils and tests","message":"refactor: Remove unused code from utils and tests\n\nSince Iaca6fd3e6ed3c64ab3ca22817ad461479ecfa189, _verify_http_connection\nis no longer used anywhere. Its removal also exposes two more utils\nfunctions that are also not used anywhere.\n\nChange-Id: I88423b89f30ef2c4cb4738bbee1246e019796d73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f1c300ded225565913ea47ce8aa5af01e01f6b9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f1c300ded225565913ea47ce8aa5af01e01f6b9e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"86ddcd2fa5822c556e2e224c231cd82698535b15":{"kind":"REWORK","_number":4,"created":"2024-09-09 16:02:53.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/61/928461/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/61/928461/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/928461/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/928461/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/928461/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/61/928461/4"}}},"commit":{"parents":[{"commit":"f70313a236421d3852b97ef094d8522326833737","subject":"Merge \"zuul: remove obsolete job definitions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f70313a236421d3852b97ef094d8522326833737"}]}],"author":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2024-09-06 17:06:48.000000000","tz":-240},"committer":{"name":"Ihar Hrachyshka","email":"ihar.hrachyshka@gmail.com","date":"2024-09-09 16:02:08.000000000","tz":-240},"subject":"refactor: Remove unused code from utils and tests","message":"refactor: Remove unused code from utils and tests\n\nSince Iaca6fd3e6ed3c64ab3ca22817ad461479ecfa189, _verify_http_connection\nis no longer used anywhere. Its removal also exposes two more utils\nfunctions that are also not used anywhere.\n\nChange-Id: I88423b89f30ef2c4cb4738bbee1246e019796d73\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/86ddcd2fa5822c556e2e224c231cd82698535b15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/86ddcd2fa5822c556e2e224c231cd82698535b15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
