)]}'
{"id":"openstack%2Fneutron-tempest-plugin~929633","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I7f690b1a260ff70e12bcfac82f63e36ddca7cfbc","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"remove_geneve_from_writable_mtu_test","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-09-19 06:28:47.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-09-20 16:43:07.000000000","reason":"Change was submitted"},"29350":{"account":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"last_update":"2024-09-20 10:48:06.000000000","reason":"\u003cGERRIT_ACCOUNT_29350\u003e replied on the change","reason_account":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"}},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2024-09-18 14:36:08.000000000","reason":"removed on reply"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-09-18 13:49:36.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"33341":{"account":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"last_update":"2024-09-20 16:43:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7f690b1a260ff70e12bcfac82f63e36ddca7cfbc","subject":"Skip NetworkWritableMtuTest when driver is ML2/OVN","status":"MERGED","created":"2024-09-17 12:15:44.000000000","updated":"2024-09-20 18:50:58.000000000","submitted":"2024-09-20 16:43:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929633-remove_geneve_from_writable_mtu_test","meta_rev_id":"6d4ba73287c37fd1af0fa18d7015ab0a7b5fe03a","_number":929633,"virtual_id_number":929633,"owner":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-20 16:43:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-09-18 09:21:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-09-18 14:36:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2024-09-18 13:49:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2024-09-18 13:49:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-17 14:28:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-17 14:37:34.000000000","updated_by":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-09-17 14:37:34.000000000","updated_by":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-09-17 14:37:34.000000000","updated_by":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"reviewer":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"state":"REVIEWER"},{"updated":"2024-09-17 17:11:49.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-09-17 19:40:40.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"0367f6237e8403b71c8412dec7bdf4cb6dcce0e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2024-09-17 12:15:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75cb40e647249f962a8f560c05b8381bc262b191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 14:28:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ce0597b5d8d4ea286e0f927fe321506\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/592b686af6a142a0bd7bf69708c782a6 : SUCCESS in 4m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f4115308181b4f1495c29e921d796362 : SUCCESS in 2h 01m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8249218529584d23a11af34dba67cec3 : SUCCESS in 2h 10m 11s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/5b23fc1ddb784f9e935ace39aae4d4fe : SUCCESS in 1h 57m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/30ee75641c0d45428a18ecc581fba413 : SUCCESS in 2h 09m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1cb4c23bee4b43a4a70ef74fd4b67aa3 : SUCCESS in 28m 33s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/8cfb76b8795a48bb92c1b61f810dc23d : SUCCESS in 1h 22m 47s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/c9884f47f3bf4f4ca73c3f81e5f38fd8 : SUCCESS in 1h 10m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/ffa1c8d7a5b64b84aa00d1646bf11180 : SUCCESS in 1h 14m 36s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/9f327b31ffb24bd3a01d90f54e504b1c : SUCCESS in 1h 24m 51s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/d37bd7ba6d074e81b0b163f0fc45dd3e : SUCCESS in 30m 11s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/d88b6b02711d49e4aab81a15527faf1e : SUCCESS in 1h 14m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/1d804518ed914f8b918800bbca400bfb : SUCCESS in 1h 08m 24s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/b3626dccdf49454287f3ab3e87f0b9cf : SUCCESS in 1h 33m 28s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/0100066f7688478388b3e579773996f7 : SUCCESS in 35m 50s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/6d1f419548de4d39a01c84135dc3431f : SUCCESS in 1h 31m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/95d76800edc4485a91bf25df8b5226b6 : SUCCESS in 1h 57m 19s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/3eaba4475cc540949229268d035f4d95 : SUCCESS in 1h 56m 39s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/a4856fa66aa04b9a9db6414765747933 : SUCCESS in 40m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d1eb71dcec04423948b3b83c52b083c : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"eab0cc6d76acaebad8ec0e5df131b3b4f7720ce4","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2024-09-17 14:39:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b482a58da73daa5e8d6f410596cb1e409f865783","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-09-17 17:11:49.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f5716f0df2f0bc42e795535b1d1256c5ce8e024","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-09-17 19:40:40.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/690c7129a6c748778c52aa60db3ae8c2\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/57019c449de441c69fcf168f731f700e : SUCCESS in 6h 43m 23s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/13ee361f7aae4d949714ae4f0d1d9a85 : FAILURE in 3h 02m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9be4bbb5064d24fc71ddb01c9a095104e05bdbd0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-18 07:04:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bbf80535e5ea39379b9407b7cd5c1a38ada4bf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2024-09-18 09:18:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5a14358a0e03b3ff8047837aeaa5096a1149e459","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-18 09:21:08.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"304496bdf1fa1580f44d4d1f83ed0c483adc8ca2","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2024-09-18 09:25:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1058b4138a09a344de678426fc782dec793c8210","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2024-09-18 09:31:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fc3bd75641f21d4df3e57092fd0c85d84f93a7ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 11:48:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95f84f48dcbb4442b8bf7bf58d512b93\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffb7037b6728402485fd41450abdd7c0 : SUCCESS in 3m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5594df9cf69948b1ba3df06a55f532a3 : SUCCESS in 1h 46m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9a6938dad7ae437185c6c902d663e946 : SUCCESS in 2h 25m 13s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c8d81ac52dd442959a77646cdb54de81 : SUCCESS in 1h 33m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8cfd1fdb93724b16bbf312402eeaae11 : SUCCESS in 1h 47m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/57618775d23348b88dc3f21a0be79176 : SUCCESS in 30m 32s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/35d56ff57a844fe68e2c06409c505247 : SUCCESS in 1h 22m 39s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/e92172439b6249049fb51281130c38f3 : SUCCESS in 1h 28m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/3d859359076c4931ad2c5ff8f438f4ae : SUCCESS in 1h 43m 52s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/b7beb3cc4fec430aa92f118cbf66d725 : SUCCESS in 1h 35m 28s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/6ea85ce92d664beeb1ec9cecebee4feb : SUCCESS in 32m 46s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/9a7c65cf175543cf90a151f8bcc22ab1 : SUCCESS in 1h 27m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/b74ead4dff7542d4ae1f9e191ff3d193 : SUCCESS in 1h 52m 31s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/2d87a3f37c704a6a8dcff86702d563fe : SUCCESS in 1h 19m 40s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/e18c36d22cd647c8b7c38376e2bb5c25 : SUCCESS in 39m 28s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/02cea77de02e4b7ea6c557837543b89d : SUCCESS in 1h 40m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/8babc13893b44f5c86be453a30d1b27d : SUCCESS in 1h 51m 50s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/165f798557aa44f09ca9d638c9a2c736 : SUCCESS in 1h 37m 52s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/97efc3ef187c4f16b3445dc1811aca21 : SUCCESS in 40m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b17826dc6a054db5a2c3d3055b268aba : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":2},{"id":"bd04bae5a93cf4ec24437208584deafdbab738ce","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-09-18 13:49:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1d6751e0bd49d8384f7eb9acc7a4721341c88784","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 13:50:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6b77ce2fb94c6166ee893b1d636e75bc9061f645","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-09-18 14:36:08.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/85645ef3afe04e66a993c4293fc67607\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/09c57f5369a64e44a0a70c6710b727d2 : SUCCESS in 4h 15m 16s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/7bc78e0917244a08958aa3c64fccf436 : FAILURE in 3h 13m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dd93f15627b01d548e4dcbcc7b5ace8b8f71befd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 16:35:11.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1c485d3ac65410d846922d1275b10f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e410e8316a834e4a9da22af551c40353 : SUCCESS in 4m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0a138d727eee42668dcc9e576f65b6c3 : SUCCESS in 1h 31m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bad63d370a5b46bcaaafbcd699a996f6 : SUCCESS in 2h 37m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a6be1ef6bfba414bbbf49814e3863c3a : SUCCESS in 2h 43m 24s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/95309b16ecac4edabb7197c36dcb33e3 : SUCCESS in 1h 47m 42s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/74386ac247944bd2ba9a9a33b09c1e97 : SUCCESS in 2h 01m 17s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/7db84d8dd093492ca791b207b4f3ff53 : SUCCESS in 1h 25m 51s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/8743a230e24e4284919a179b070e0f6e : FAILURE in 1h 57m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a7f3c1b04f14e7ebe84c1b91f4a33e5 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":2},{"id":"d61bf2148908d2d6ddbd2f847a5663a993f0415e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-19 06:28:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4232e3ed68b6cf5baba0facce4e507ff1a6636fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 08:46:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab75dbfc41b640de93d8ba666494f57e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ccb427cb76044b2ae9cb96bd7fcdfd2 : SUCCESS in 5m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d33acd60b13546e6b7031386ebb28c49 : SUCCESS in 1h 43m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/842740fddc324a56b775cbe248b2e186 : SUCCESS in 2h 03m 19s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/27ec3446fd344c8f864ae9f8bd729f74 : SUCCESS in 1h 59m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c8ab2b4dfbf74e29b28ba8a7e128a22c : SUCCESS in 2h 15m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6307200b743e43fab2bd8e286c66f657 : SUCCESS in 27m 10s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/078abe2313ac49ac8450dfba25178240 : SUCCESS in 2h 14m 47s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/736df9ebc1ce40aa86fcf3bb1b435ef8 : SUCCESS in 1h 15m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/412ded0261424dd5ab648306c1482ca4 : SUCCESS in 2h 01m 10s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/712f8da4c22241219a082a80a654ca52 : SUCCESS in 1h 30m 51s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/a149fd85ec5240b794248f19c3145395 : SUCCESS in 41m 01s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/4032eb88f4314b2f8ba0255db856700e : SUCCESS in 59m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/46a76661892f4120b45648e80262ca32 : SUCCESS in 2h 05m 22s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/e2a03c8e913b4e30b4dfc9318180ccca : SUCCESS in 2h 14m 24s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/b19969ad98364ae3aac9da469ed9925e : SUCCESS in 44m 12s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/7dec3776419a4333b3642bb6c268ebd6 : SUCCESS in 1h 43m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/aff78b2d0152456db698d8372f183b0d : SUCCESS in 1h 47m 12s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/7ee48d9c0d9f4a9995ea8c8b94747150 : SUCCESS in 2h 01m 39s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/c1029fffd34f471cad18588ecdce55dd : FAILURE in 28m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a096f1a999949debb41b9403eb5bbc1 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":2},{"id":"c00c2ef0a6f5d86dc2c3333d39cd1e369c9d640b","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-09-19 10:40:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/66bac920afeb408abd3bc28e3ce0521f\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/7956462272bf44919c6d330706b075be : SUCCESS in 4h 10m 29s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/d33424257e01458e9dbf99abe9bc858b : SUCCESS in 3h 15m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"174ccbdcb4c79b12c79395c2f20986873ef9b0fb","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2024-09-19 14:19:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0fb111e9e8423b9f82b0f9da71781e6c8e2b180a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 16:42:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e26ee511d9f64a5184f8c9bbc27c9ca3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0199c86808fb41cbb5d3ae88a0dda797 : SUCCESS in 3m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0d2ae0e6adde40239483799ed7ccff73 : SUCCESS in 1h 49m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5c970fe8b7764da492ea76d3d7841618 : SUCCESS in 2h 04m 13s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/ab4a73007a1a4a51bfe3e78f59ab6187 : SUCCESS in 1h 36m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ded3a10348e64423a07fa9e1c4a753df : SUCCESS in 2h 20m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/260b5e7c882140e89cda9bf6d873b76f : SUCCESS in 31m 49s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/5f04cbe71fa74d1c862d2e4ad662a77d : SUCCESS in 1h 25m 56s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/e58a9f6c6f18457db9dc893ad63e0ed3 : SUCCESS in 1h 10m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/a26c573dac1c488cbd0ede0286a40b38 : SUCCESS in 1h 16m 53s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/101b1d082cf643b996c485fceb819328 : SUCCESS in 1h 37m 01s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/46e0fc4f35b941f8a1a86d61dbdb89a4 : SUCCESS in 30m 42s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/f16b5b36a41446b1a1670b4f1b55abb3 : SUCCESS in 1h 09m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/80e991e943ea47f98cf3dc017683d847 : SUCCESS in 1h 26m 38s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/bc78fec2caab4ef898a057504dc631fb : SUCCESS in 1h 46m 18s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/e9432af099f4481597cfe28851d65965 : FAILURE in 28m 40s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/65bc297cd1194b549bf66d31c06ad2cc : SUCCESS in 1h 12m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/1f45835424764e92bab4c38da6180379 : SUCCESS in 1h 23m 27s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/0a8c92586c7d4ff580a9478bedb16254 : SUCCESS in 1h 26m 45s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/c359c83e544f491d91faa9e0c2623c66 : SUCCESS in 38m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b022e8fe61b64aaa8ff664b46f901d5a : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":2},{"id":"179ba9120c00a974fad32fc0e90567fddefe68ce","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-09-19 18:45:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/b3c1b39c17fc4c4d96a9201abc9f7c7e\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/c7ca12836bc248ecb1e35b7518ca7a7d : SUCCESS in 4h 24m 25s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/31dbe2b1989b4074b93e83cc551c5aae : SUCCESS in 3h 18m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5687287ab1f42155cb26e50faea01657237662a1","author":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"date":"2024-09-20 10:48:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"67829119b69beffd2a5d48fd2dd60bd3dccc5e60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 13:30:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d38f0bf97d641549f38718ee309342f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a01b9db2fae04e4cafc3486d7a2b923c : SUCCESS in 3m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/22c0755a8023458cbdb3d1a2e568f143 : SUCCESS in 1h 42m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bea413e8c6904c39a20e834348878ae5 : SUCCESS in 2h 40m 48s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c0944c008cf34afbb3ce0c905b3fb16f : SUCCESS in 1h 47m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fca06ab447034b3184bb7f22d9700d57 : SUCCESS in 2h 05m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/38521f3ff6bd46cb912dcec32a0eb568 : SUCCESS in 41m 08s (non-voting)\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/a2b78a0649604c8784d9b29a7d96d68e : SUCCESS in 1h 43m 06s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/2ee9165fbbdf4bb2828333e12792170d : SUCCESS in 1h 11m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/970e27ffd3f341e7a9ac012aade3a1cb : SUCCESS in 1h 15m 53s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/0e03dc5be56440a2afe25b6e3bab5a02 : SUCCESS in 1h 43m 42s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/a645c63046634bf08b3ae4998fbccf53 : SUCCESS in 32m 07s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/cde66c9d228e4544b735991f678706ee : SUCCESS in 1h 26m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/75bde51592c841b0b56ab40d7d4cb995 : SUCCESS in 1h 32m 17s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/8e32c431e7a64d648e544586fdb357c1 : SUCCESS in 1h 38m 19s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/ddedf0d2f7ec4d288a16c21a20bd0fbf : SUCCESS in 28m 01s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/e12dca8b93464b24bcd1bb35e530a9de : SUCCESS in 1h 16m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/6375e9984c67430f91bca0b4f7621f78 : SUCCESS in 1h 28m 03s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/e33220b365e743d08fbe8d1bfa7a8a4f : SUCCESS in 2h 12m 14s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/9e3ead7e5bd945bb9835fde6757759d5 : SUCCESS in 36m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44cabaabddd04b7b87df5b33da2a0f80 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"e7b5aee5797cbaf1229490066e499447b9999528","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 13:31:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e64e89421926f85301341a528cd1db0e110391f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 16:43:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4317d3f7b5c54372a5ca234187055464\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/910f853d3bb742929d7e4291dafb3eb8 : SUCCESS in 4m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/28701133b60944ba8f7549a6d1b82e1a : SUCCESS in 2h 04m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/927b8bde42fa41489f6200ba42e38d48 : SUCCESS in 3h 02m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f807f0f8d02a4ac8ac1569973fb205b0 : SUCCESS in 2h 37m 01s\n- neutron-tempest-plugin-ovn-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/e07be16bdd3844b9a5dd1daba557612b : SUCCESS in 2h 20m 25s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/7343e6ae8a1443c1b7dc43c76c2599fa : SUCCESS in 2h 16m 41s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/c1c46cd45ad0416ab3d394b6cfebc4e8 : SUCCESS in 1h 13m 21s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/d6a06d8a7e37402abadd5dedd80f3591 : SUCCESS in 1h 47m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de0aaaa4ccc34d039f4d2b0b7481d9f4 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"2b57ce5c64d985ea8b9de4925cddfcb4b0e4be15","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-20 16:43:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6d4ba73287c37fd1af0fa18d7015ab0a7b5fe03a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-09-20 18:50:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/7b52e8c5b9d349cb94607de4b843e8b2\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/682c63386dc74751b3e25c3a7a601e1b : SUCCESS in 4h 44m 00s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/427b8d7c747043f09d9af82634752268 : SUCCESS in 3h 27m 01s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9be9375131ad08c016d0f2de2c8ddc5e4b3eab23","revisions":{"0e31266f4f75c5b5ddd6aecea0572872261ae5a7":{"kind":"REWORK","_number":1,"created":"2024-09-17 12:15:44.000000000","uploader":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"ref":"refs/changes/33/929633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/33/929633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/33/929633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/33/929633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/33/929633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/33/929633/1"}}},"commit":{"parents":[{"commit":"26adc41889a9093924b7928292c74588bdc3bab3","subject":"Merge \"Add test for extension \"quota-check-limit-default\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/26adc41889a9093924b7928292c74588bdc3bab3"}]}],"author":{"name":"Roman Safronov","email":"rsafrono@redhat.com","date":"2024-09-17 11:57:48.000000000","tz":180},"committer":{"name":"Roman Safronov","email":"rsafrono@redhat.com","date":"2024-09-17 12:15:00.000000000","tz":180},"subject":"Remove geneve from type drivers for e/w mtu test","message":"Remove geneve from type drivers for e/w mtu test\n\nEast/west icmp fragmentation is not supported with geneve since\nthere is no production use case, as mentioned in [1].\nRemoving geneve from available_type_drivers list of\nNetworkWritableMtuTest will cause the test be skipped\nautomatically and thus ci jobs configs can be simplified\nby removing exclude list.\n\n[1] https://docs.openstack.org/neutron/latest/ovn/gaps.html\n\nChange-Id: I7f690b1a260ff70e12bcfac82f63e36ddca7cfbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0e31266f4f75c5b5ddd6aecea0572872261ae5a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/0e31266f4f75c5b5ddd6aecea0572872261ae5a7"}]},"branch":"refs/heads/master"},"9be9375131ad08c016d0f2de2c8ddc5e4b3eab23":{"kind":"REWORK","_number":2,"created":"2024-09-18 09:18:55.000000000","uploader":{"_account_id":29350,"name":"Roman Safronov","email":"rsafrono@redhat.com","username":"rsafrono"},"ref":"refs/changes/33/929633/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/33/929633/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/33/929633/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/33/929633/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/33/929633/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/33/929633/2"}}},"commit":{"parents":[{"commit":"26adc41889a9093924b7928292c74588bdc3bab3","subject":"Merge \"Add test for extension \"quota-check-limit-default\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/26adc41889a9093924b7928292c74588bdc3bab3"}]}],"author":{"name":"Roman Safronov","email":"rsafrono@redhat.com","date":"2024-09-17 11:57:48.000000000","tz":180},"committer":{"name":"Roman Safronov","email":"rsafrono@redhat.com","date":"2024-09-18 09:10:01.000000000","tz":180},"subject":"Skip NetworkWritableMtuTest when driver is ML2/OVN","message":"Skip NetworkWritableMtuTest when driver is ML2/OVN\n\nEast/west icmp fragmentation is not supported with ML2/OVN since\nthere is no production use case, as mentioned in [1].\nWith this change NetworkWritableMtuTest will be skipped\nautomatically when ML2/OVN agents found and thus ci jobs\nconfigs can be simplified by removing exclude list.\n\n[1] https://docs.openstack.org/neutron/latest/ovn/gaps.html\n\nChange-Id: I7f690b1a260ff70e12bcfac82f63e36ddca7cfbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9be9375131ad08c016d0f2de2c8ddc5e4b3eab23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/9be9375131ad08c016d0f2de2c8ddc5e4b3eab23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
