)]}'
{"id":"openstack%2Fneutron-tempest-plugin~934027","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I3f6b033123ced2d715c8f4e2cd172393f202000e","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2086602","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-11-11 07:53:09.000000000","reason":"Change was abandoned"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-11-11 07:53:09.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-11-11 07:53:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3f6b033123ced2d715c8f4e2cd172393f202000e","subject":"Increase the number of available VLAN IDs to 200 for \"foo\" network","status":"ABANDONED","created":"2024-11-04 09:47:39.000000000","updated":"2024-11-11 07:53:09.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2feca7b76cf16ac798699bf3040e172441f4c12e","_number":934027,"virtual_id_number":934027,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-11-04 11:59:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-11-04 15:18:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"date":"2024-11-04 13:50:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2024-11-04 13:50:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-04 09:49:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-11-04 09:49:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-11-04 09:49:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2024-11-04 09:49:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2024-11-04 09:56:05.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-11-04 11:59:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-04 15:18:33.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"ccca004947af1890d1eac289ad9b804757b935e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-04 09:47:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5890df3e6345a92c833609e4861e220b40915cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-04 09:48:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"49c1098fb2e80cf61c24c5b6935e2be36593b650","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-11-04 09:56:05.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7010acee5dbe7a404e4ad87f6343d024368786e5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-11-04 10:54:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"58ec7f1ae3c293ff86dd21eaf57177ca3d64a506","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 11:59:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0486a8628dd145818b315434b97a92e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/877a438c4cf04ee9a408d6e1ab915956 : SUCCESS in 3m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9fc1990532874a06b3224304bbf1666f : SUCCESS in 1h 53m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/38a045eb2a62441aa8343bec0e416d2c : SUCCESS in 1h 48m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d1124df630724a51a90105ed43f0cf05 : SUCCESS in 1h 59m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/92abdba7b7374bb8949433e0d3808041 : SUCCESS in 34m 59s (non-voting)\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/92d28d30c22244a28779cf8d5e0e1153 : SUCCESS in 57m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/d01779902ded4713a7e627f8637e04af : SUCCESS in 1h 39m 23s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/d09c4d2d4f184b7aa09318c94ce14f3d : SUCCESS in 1h 25m 06s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/868f7a0e00cb4428975eb143fd04bb7e : SUCCESS in 40m 37s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/959fadcaf5df4c849d46030cd55a3b5f : SUCCESS in 1h 28m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/ee1a652da9ff45709230ba71a7f848a4 : SUCCESS in 1h 17m 19s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/ad9a525688544cb7b3a378bd7af0c985 : SUCCESS in 2h 04m 12s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/b50cda1c955e48aa9679bc14d99a68b0 : SUCCESS in 34m 27s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/58cd283396eb47da8f63544b8e3c062b : SUCCESS in 1h 30m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/5078f5b0ff664031afe5f4b5a0de1914 : SUCCESS in 2h 09m 18s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/df39ef587a2745f58dd77ea6f86a4ebe : SUCCESS in 1h 50m 15s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/c84d85a16b584271803b5393cdd37b4c : SUCCESS in 28m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dce659223104be4aee62521b962e1fd : SUCCESS in 3m 16s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/2deadc06dc584a0e8cbe786cec06264e : SUCCESS in 1h 57m 41s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/07a898f97bd34d169a1535b8e1555abc : SUCCESS in 1h 41m 22s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/a17d26820b354f07936d45b5148a780f : SUCCESS in 1h 38m 58s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/a4fdd57ae0b6443985d08c0d35426abe : SUCCESS in 1h 32m 23s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9cd6a0b3a6dd417b85218a16bf38b9bc : SUCCESS in 1h 45m 07s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/b76f7184bed8498e8964680ae91a8a6f : SUCCESS in 1h 28m 31s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/ce7717d965964adbac55e47b3f5a2fc1 : SUCCESS in 1h 44m 30s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/43486b07f543401ebe2e9355126748ca : SUCCESS in 1h 48m 44s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/2df741ecda0d412bbc40028515e31e96 : SUCCESS in 56m 44s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/21f8ce089b064b87b45b2714fa0b903c : SUCCESS in 59m 41s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/e41d34e177c841da84533b42f740c826 : SUCCESS in 54m 55s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/cb1826ec2fd64260bbc2574a2497aff4 : SUCCESS in 57m 41s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/23496670049a4529a759e73d86151630 : SUCCESS in 52m 42s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/43b2858ce8d04ee9baadaa2acf600e59 : SUCCESS in 47m 22s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/2223d39689fa4734ae3afcc784a3f3d1 : SUCCESS in 47m 58s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/5bfa9d73ed394bb7a5e860188d497b29 : SUCCESS in 46m 14s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/7383501e7fec40b58ae31c5d7c086221 : SUCCESS in 37m 30s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/eff4f9a9191a42c7b3c5bea0c719dcf6 : SUCCESS in 35m 08s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/3a27278cf85545e6a7c87a2981c7e4dd : SUCCESS in 48m 51s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/85c86f41c6a4403981b6378ea34100b3 : SUCCESS in 36m 36s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/010badd02fcc47a5a88bad69f7b9f9cc : SUCCESS in 51m 26s","accounts_in_message":[],"_revision_number":2},{"id":"071e237ce5cb08074956c465984e301a8980880b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-04 13:50:42.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1719c89abaf5feba48329ac6a1890cfe7d1eb3b1","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-04 13:50:45.000000000","message":"Removed Code-Review+2 by \u003cGERRIT_ACCOUNT_8313\u003e\n","accounts_in_message":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"_revision_number":2},{"id":"724b59624dc19cf716d111faeca47b4c2bd4a7b3","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-11-04 15:18:33.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/f1520d33f39e4e3c91a67aa503b13d32\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/eb49e454347a47e9befaa0ca6afc81b2 : SUCCESS in 5h 28m 40s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/45c79c2be51a41b89f34930ab702f02c : SUCCESS in 3h 19m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2feca7b76cf16ac798699bf3040e172441f4c12e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-11 07:53:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1bec0ce5e1c3788d463fb2f998260ac933b01beb","revisions":{"d7594d426336b1649b1fac1e5934011402b10b34":{"kind":"REWORK","_number":1,"created":"2024-11-04 09:47:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/27/934027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/27/934027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/27/934027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/27/934027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/27/934027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/27/934027/1"}}},"commit":{"parents":[{"commit":"505a7981278f9f0ea80f72fcea99514aef383833","subject":"Merge \"Fix trunk transition waiting message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/505a7981278f9f0ea80f72fcea99514aef383833"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-04 09:46:53.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-04 09:46:53.000000000","tz":0},"subject":"Increase the number of available VLAN IDs to 200 for \"foo\" network","message":"Increase the number of available VLAN IDs to 200 for \"foo\" network\n\nCloses-Bug: #2086602\nChange-Id: I3f6b033123ced2d715c8f4e2cd172393f202000e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d7594d426336b1649b1fac1e5934011402b10b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d7594d426336b1649b1fac1e5934011402b10b34"}]},"branch":"refs/heads/master"},"1bec0ce5e1c3788d463fb2f998260ac933b01beb":{"kind":"REWORK","_number":2,"created":"2024-11-04 09:48:52.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/27/934027/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/27/934027/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/27/934027/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/27/934027/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/27/934027/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/27/934027/2"}}},"commit":{"parents":[{"commit":"505a7981278f9f0ea80f72fcea99514aef383833","subject":"Merge \"Fix trunk transition waiting message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/505a7981278f9f0ea80f72fcea99514aef383833"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-04 09:46:53.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-04 09:48:46.000000000","tz":0},"subject":"Increase the number of available VLAN IDs to 200 for \"foo\" network","message":"Increase the number of available VLAN IDs to 200 for \"foo\" network\n\nCloses-Bug: #2086602\nChange-Id: I3f6b033123ced2d715c8f4e2cd172393f202000e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1bec0ce5e1c3788d463fb2f998260ac933b01beb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/1bec0ce5e1c3788d463fb2f998260ac933b01beb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
