)]}'
{"id":"openstack%2Fneutron-tempest-plugin~935368","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ib93fb642f067b7de5746d25418cd6c14c7aa9204","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2088285","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-01-02 09:17:11.000000000","reason":"Change was abandoned"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2024-11-18 12:39:20.000000000","reason":"removed on reply"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-01-02 09:17:11.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-01-02 09:17:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib93fb642f067b7de5746d25418cd6c14c7aa9204","subject":"Change references from Quagga to FRR","status":"ABANDONED","created":"2024-11-15 16:26:36.000000000","updated":"2025-01-02 09:17:11.000000000","total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"860742cd26c73d42b8f06fd68d4a5974914ae55c","_number":935368,"virtual_id_number":935368,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-18 10:33:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2024-11-18 12:39:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-15 16:36:16.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-11-15 17:28:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-15 20:39:38.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"b0eb9c579ad88cfa102e2b30de713e33f705912c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-11-15 16:26:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e954cc8299cbe4ea324e557f43bea39c33220dcc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-15 16:36:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"21a08b82fee114b2843b5db5f75080b234a68b47","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-15 16:37:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ec6ed5ba2d0d4506afa9aee9749d30d7cf181b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 17:28:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8defae32192e418c9c81e74a292fee93\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd4c91bfba0408280a649c34c8a9196 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50aa9bd6751146c190c6dfbd899b2191 : SUCCESS in 3m 07s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/c69346579c124bac993b71a9808f76aa : FAILURE in 34m 37s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/2a6d4a7ca9f34d2392ba16d2016459ef : FAILURE in 1h 00m 49s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/366fd97cbdd0400e9017ec9d436ede4e : FAILURE in 45m 47s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/8f5ebaebcb75460b8354ff6fdd84b580 : FAILURE in 31m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f3e2c046f1e51b7b9f417e7e26ad3edb55d94ce2","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-11-15 20:39:38.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/9a76fc61030f48d396f77f3268ff291b\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/ec4d2e866abe40ffb1486042e49f32f3 : SUCCESS in 2h 19m 57s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/7a53bd3285934e74adec2799420b9d44 : FAILURE in 1h 16m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8122f2a8ca13190bee3ba93f368f6ddadf05baa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-18 06:10:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6b7400d2dc3dba7d167b00aced7420d584768d8f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-18 06:11:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c8222400bfd29de7d0c43a6f8c0f4738a3bd69f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-18 06:14:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cd11758e021504f57a405d19abb4495048390923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 06:58:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c596f31206c4cefb8cf7f9d84822925\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db938b35a9a43cc847a811be648bdff : SUCCESS in 3m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9aa1ab1e02104d5e9e0d6223ecaf0b4e : SUCCESS in 1m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a155cece45948e28d865a2321a76d36 : SUCCESS in 2m 52s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/6d8bc2296d7b4bb798eb801c0bccf265 : FAILURE in 25m 46s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/832681edd100428cb2a939d1984515b2 : FAILURE in 27m 45s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/e77499520e07452eb4d3f6e446e2b3e9 : FAILURE in 26m 12s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/a29551448f914839b638a949e28ca11f : FAILURE in 42m 41s","accounts_in_message":[],"_revision_number":3},{"id":"f4b0c2f87218546e2dfa1c6713f9e609fe69d6b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-18 07:03:38.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"38137801300f15529432489d46ee463e387b4999","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-11-18 07:32:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"71ddf691a4707df4978aa91c359a1f4f5987d74b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-11-18 07:39:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7a5a03ab25c4897eaaf59672a81e831ea3cdec03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-18 07:55:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"118e12d6a904e5dfec5bbca6ecc76726ee4cda86","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-11-18 08:06:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e9667fea145c635651c9d00f22f3a1de8c2f9d3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-11-18 08:11:01.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bc8aa17a3d59cff314a20254768d29d0e7b6a2e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 09:07:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8010cdbd9434fb0ba9548d3479541ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1441227d824643189b1db9d62d7a0b2c : SUCCESS in 3m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a42b2bc885f74175b4ee185b5d35e58c : SUCCESS in 2m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ece2463c473b4ea489a50e08629d656b : SUCCESS in 2m 59s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/130e6460e8974746b2ad29d8a8fff067 : FAILURE in 44m 17s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/03abe4e14fc8488a9095c07109741a67 : FAILURE in 48m 20s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/e8146c28330749e78e4044bc03ba06e0 : FAILURE in 50m 16s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/2aaceda9045d4c55b6814a595e8d7b1b : FAILURE in 44m 18s","accounts_in_message":[],"_revision_number":6},{"id":"2a56756156ca257d1eae9821e2396b7e66ffed6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-18 09:38:01.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"342a1123e18e3a408dba16297ad4fec54a1a72f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 10:33:56.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/244fdf20802d48528373e614c78b2de0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df87aaadacad4234b843410d2f48e357 : SUCCESS in 3m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0b48b05707a749c28664ff9da45158fe : SUCCESS in 1m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a2a5cecd0fe4c0fa56b96b33f10bd1e : SUCCESS in 2m 32s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/bb1b518ba5f1490d8f21c63553e13be1 : FAILURE in 36m 02s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/59ae1784fa2d44039e87d5be01d56e52 : FAILURE in 27m 09s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/45f6ffcf2fd54fce96e640717eef8cf1 : FAILURE in 46m 41s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/28f6f8660fff4ae992039e4ad2e73a87 : FAILURE in 28m 48s","accounts_in_message":[],"_revision_number":7},{"id":"b621d18699216a6a8b022708bb0c329a4fccd483","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-11-18 12:39:20.000000000","message":"Patch Set 7: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/9b5ec9f4fc8d43ec81baf5abfaca0cb6\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/b9f18c5fc80a4636b225dc64a015883b : SUCCESS in 2h 31m 23s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/242a6e416f7343ebb7c3c1fa07afc0cd : FAILURE in 1h 12m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"860742cd26c73d42b8f06fd68d4a5974914ae55c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-01-02 09:17:11.000000000","message":"Abandoned\n\nAs https://review.opendev.org/q/04be5a3949175f64056d56c89a4d1bdcdd6c6c7e was reverted this is not necessary now","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6315c192fcdafdfb408c2feaa8c40d723295abdf","revisions":{"06cb9f5266b8ff2f7e4e5c5bdbc7f0081591bf07":{"kind":"REWORK","_number":1,"created":"2024-11-15 16:26:36.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/68/935368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/68/935368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/1"}}},"commit":{"parents":[{"commit":"e52927e7677c7376b4e2263da551f018e01bca7c","subject":"nit: Drop mention of postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e52927e7677c7376b4e2263da551f018e01bca7c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-11-15 16:24:07.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-11-15 16:24:07.000000000","tz":60},"subject":"Change references from Quagga to FRR","message":"Change references from Quagga to FRR\n\nIn os-ken Quagga was changed to FRR in [1], let\u0027s\nchange the references in n-t-p also to FRR module\nfrom os-ken.\n\n[1]: https://review.opendev.org/c/openstack/os-ken/+/891677\n\nCloses-Bug: #2088285\nChange-Id: Ib93fb642f067b7de5746d25418cd6c14c7aa9204\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/06cb9f5266b8ff2f7e4e5c5bdbc7f0081591bf07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/06cb9f5266b8ff2f7e4e5c5bdbc7f0081591bf07"}]},"branch":"refs/heads/master"},"e2f2c1e1b1a214303c414b5da47c8d31eb5e5a7b":{"kind":"REWORK","_number":2,"created":"2024-11-18 06:10:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/935368/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/68/935368/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/2"}}},"commit":{"parents":[{"commit":"e52927e7677c7376b4e2263da551f018e01bca7c","subject":"nit: Drop mention of postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e52927e7677c7376b4e2263da551f018e01bca7c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-11-15 16:24:07.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 06:10:29.000000000","tz":540},"subject":"Change references from Quagga to FRR","message":"Change references from Quagga to FRR\n\nIn os-ken Quagga was changed to FRR in [1], let\u0027s switch the usage in\nn-t-p also to FRR module from os-ken.\n\n[1]: https://review.opendev.org/c/openstack/os-ken/+/891677\n\nCloses-Bug: #2088285\nChange-Id: Ib93fb642f067b7de5746d25418cd6c14c7aa9204\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e2f2c1e1b1a214303c414b5da47c8d31eb5e5a7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e2f2c1e1b1a214303c414b5da47c8d31eb5e5a7b"}]},"branch":"refs/heads/master"},"7767352715a6532a5b24a6b3885ebe8e615acc45":{"kind":"REWORK","_number":3,"created":"2024-11-18 06:14:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/935368/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/68/935368/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/3"}}},"commit":{"parents":[{"commit":"e52927e7677c7376b4e2263da551f018e01bca7c","subject":"nit: Drop mention of postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e52927e7677c7376b4e2263da551f018e01bca7c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-11-15 16:24:07.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 06:14:13.000000000","tz":540},"subject":"Change references from Quagga to FRR","message":"Change references from Quagga to FRR\n\nIn os-ken Quagga was changed to FRR in [1], let\u0027s switch the usage in\nn-t-p also to FRR module from os-ken.\n\n[1]: https://review.opendev.org/c/openstack/os-ken/+/891677\n\nCloses-Bug: #2088285\nChange-Id: Ib93fb642f067b7de5746d25418cd6c14c7aa9204\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7767352715a6532a5b24a6b3885ebe8e615acc45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/7767352715a6532a5b24a6b3885ebe8e615acc45"}]},"branch":"refs/heads/master"},"8021a0a3f3287919265b93c3059a14c4d5c46e04":{"kind":"REWORK","_number":4,"created":"2024-11-18 07:03:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/935368/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/68/935368/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/4"}}},"commit":{"parents":[{"commit":"e52927e7677c7376b4e2263da551f018e01bca7c","subject":"nit: Drop mention of postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e52927e7677c7376b4e2263da551f018e01bca7c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-11-15 16:24:07.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-18 07:03:26.000000000","tz":540},"subject":"Change references from Quagga to FRR","message":"Change references from Quagga to FRR\n\nIn os-ken Quagga was changed to FRR in [1], let\u0027s switch the usage in\nn-t-p also to FRR module from os-ken.\n\n[1]: https://review.opendev.org/c/openstack/os-ken/+/891677\n\nCloses-Bug: #2088285\nChange-Id: Ib93fb642f067b7de5746d25418cd6c14c7aa9204\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8021a0a3f3287919265b93c3059a14c4d5c46e04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/8021a0a3f3287919265b93c3059a14c4d5c46e04"}]},"branch":"refs/heads/master"},"66969cfa7ed32569e5c0b1a33cc4f877cccd125f":{"kind":"REWORK","_number":5,"created":"2024-11-18 07:55:15.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/68/935368/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/68/935368/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/5"}}},"commit":{"parents":[{"commit":"e52927e7677c7376b4e2263da551f018e01bca7c","subject":"nit: Drop mention of postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e52927e7677c7376b4e2263da551f018e01bca7c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-11-15 16:24:07.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-18 07:55:05.000000000","tz":0},"subject":"Change references from Quagga to FRR","message":"Change references from Quagga to FRR\n\nIn os-ken Quagga was changed to FRR in [1], let\u0027s switch the usage in\nn-t-p also to FRR module from os-ken.\n\n[1]: https://review.opendev.org/c/openstack/os-ken/+/891677\n\nCloses-Bug: #2088285\nChange-Id: Ib93fb642f067b7de5746d25418cd6c14c7aa9204\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/66969cfa7ed32569e5c0b1a33cc4f877cccd125f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/66969cfa7ed32569e5c0b1a33cc4f877cccd125f"}]},"branch":"refs/heads/master"},"6b00606b0db17f4043b2d8f5cba0c3582056e7d7":{"kind":"REWORK","_number":6,"created":"2024-11-18 08:11:01.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/68/935368/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/68/935368/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/6"}}},"commit":{"parents":[{"commit":"e52927e7677c7376b4e2263da551f018e01bca7c","subject":"nit: Drop mention of postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e52927e7677c7376b4e2263da551f018e01bca7c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-11-15 16:24:07.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-11-18 08:10:51.000000000","tz":60},"subject":"Change references from Quagga to FRR","message":"Change references from Quagga to FRR\n\nIn os-ken Quagga was changed to FRR in [1], let\u0027s switch the usage in\nn-t-p also to FRR module from os-ken.\n\n[1]: https://review.opendev.org/c/openstack/os-ken/+/891677\n\nCloses-Bug: #2088285\nChange-Id: Ib93fb642f067b7de5746d25418cd6c14c7aa9204\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6b00606b0db17f4043b2d8f5cba0c3582056e7d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6b00606b0db17f4043b2d8f5cba0c3582056e7d7"}]},"branch":"refs/heads/master"},"6315c192fcdafdfb408c2feaa8c40d723295abdf":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2024-11-18 09:38:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/68/935368/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/68/935368/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/68/935368/7"}}},"commit":{"parents":[{"commit":"e52927e7677c7376b4e2263da551f018e01bca7c","subject":"nit: Drop mention of postgresql","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e52927e7677c7376b4e2263da551f018e01bca7c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-11-15 16:24:07.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-11-18 09:38:01.000000000","tz":0},"subject":"Change references from Quagga to FRR","message":"Change references from Quagga to FRR\n\nIn os-ken Quagga was changed to FRR in [1], let\u0027s switch the usage in\nn-t-p also to FRR module from os-ken.\n\n[1]https://review.opendev.org/c/openstack/os-ken/+/891677\n\nDepends-On: https://review.opendev.org/c/openstack/os-ken/+/935490\n\nCloses-Bug: #2088285\nChange-Id: Ib93fb642f067b7de5746d25418cd6c14c7aa9204\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6315c192fcdafdfb408c2feaa8c40d723295abdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/6315c192fcdafdfb408c2feaa8c40d723295abdf"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
