)]}'
{"id":"openstack%2Fneutron-tempest-plugin~936036","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I37f46cf8f7991d49966d5e4a4598ed17ae7d392a","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2037002","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-11-27 12:53:22.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-11-27 12:53:22.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-11-27 10:13:41.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2024-11-23 17:34:19.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I37f46cf8f7991d49966d5e4a4598ed17ae7d392a","subject":"Use the admin client for the QoS tag tests","status":"MERGED","created":"2024-11-22 16:03:14.000000000","updated":"2024-11-27 12:53:22.000000000","submitted":"2024-11-27 12:53:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936036-bug/2037002","meta_rev_id":"e92bd053f5da78d96d85057aceaa7644db78d8a1","_number":936036,"virtual_id_number":936036,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-27 12:53:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2024-11-23 17:34:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2024-11-27 10:13:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2024-11-25 14:58:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2024-11-27 10:13:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-22 17:19:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-11-22 18:08:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-22 21:21:35.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2024-11-25 14:58:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-11-26 15:47:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-11-26 15:47:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"8a6ae4e0b91e40402ea20fc17fac60d931e39aee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-22 16:03:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"284895ccc4c5b2d40f9a827d38e387ef70317d15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-22 16:13:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"344bdcca15b6fc2df1ae8da8d2a6a201d48a12f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 18:08:22.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c82abdf7597c4034864a819bc82efcd2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efb72d93da964878b5e8c38ed22679d2 : SUCCESS in 4m 03s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/50e456bc17a94fe4a7d96656dbd124ca : SUCCESS in 1h 20m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e6a75e068cbe4cf6b75e1fb90b622104 : SUCCESS in 1h 33m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b8aa67eeb54b447bb3e2f3e2fb9dbbbc : SUCCESS in 1h 46m 54s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/56dac0869f244b71a055cb1ff9866249 : SUCCESS in 1h 04m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/a64c01561da444d096ed713d212ef800 : SUCCESS in 1h 37m 17s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/3af4fe33dbcd435c96628266de026002 : SUCCESS in 1h 26m 36s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/4ab648bafdab45799df17ebf33c44797 : SUCCESS in 1h 23m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/d101570e3d574ad1a6ebd59521c25393 : SUCCESS in 1h 29m 12s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/02819c60075a4c70862591ab33c0d6f5 : SUCCESS in 1h 50m 52s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/abf0957336ce49fd9caeaade4fed5315 : SUCCESS in 1h 35m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/45a3145594404afc9d4e59a214084e2f : SUCCESS in 1h 47m 28s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/02b6555063224d61af223fe6cd63968a : SUCCESS in 1h 49m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/176dec63b33c4018b8529e8b636abf7c : FAILURE in 2m 50s","accounts_in_message":[],"_revision_number":2},{"id":"36d5f6e2b73b74228eac4679db49acbc1e1763c3","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-11-22 21:21:35.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/145a021b65d64c4fb163b1521ec32f74\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/5a34f2aaeeaf4688af3eb7f857b2929e : SUCCESS in 4h 26m 18s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/f920eeafeb1b45629bb24f7dc776a0d8 : SUCCESS in 3h 30m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"633c97390ec4602e5fc709b162dd8585e3ab47e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-23 13:01:31.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"81e00a695de94a4e8246ee2ef4a2426d7cf289fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 15:14:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1664f37487a4ff9bf3dc63e31af6c8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e4a0b8b272a43dd9b611f155dbaf5f5 : SUCCESS in 4m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3193b412a67a4793b2efbbc8b9f5dd91 : SUCCESS in 1h 22m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/370de6f7166d4dd8b227a33800cf2acd : SUCCESS in 1h 58m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2c83e5c0b5084d40a3f5865089271cc7 : SUCCESS in 2h 10m 34s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/960fc94a2e134bb590ddebff28bdc07b : SUCCESS in 1h 12m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/0ce646d49d9b49aaaf899cfbb18749e6 : SUCCESS in 1h 16m 02s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/885615b9bac24dbf81dca7bea32914cc : SUCCESS in 1h 29m 16s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/2bdb524e7a6b4dbdb569aa50244e9ea8 : SUCCESS in 53m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/e311fc59fbe64e4d9954fefc8e3574b5 : SUCCESS in 1h 31m 08s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/2d9684a376e5482ab710ca1dc2a4be56 : SUCCESS in 1h 49m 04s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/e1a1f195bf5540eb90bab117c8dd1dc1 : SUCCESS in 1h 26m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/e7770f490bac470390f1cde318447bfa : SUCCESS in 2h 11m 25s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/315f96e2c05d42c8b279dbb86419f786 : SUCCESS in 1h 42m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c072f2f4ccba4ade95b81b9f038780fe : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":3},{"id":"236d47c41cf13f35c943501f38ffd1635853498b","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-11-23 17:34:19.000000000","message":"Patch Set 3: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/03eef489f5b4442780ccf970a49b603d\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/124e7dc240dc448ebce8e273fbf2d6c8 : SUCCESS in 4h 31m 23s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/d1891e6ded194af88c4601f309a3c7a1 : SUCCESS in 3h 36m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"702a1183e49b6643c7625164f833d8f4cd3f5bd5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-11-25 14:58:54.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3da92da0d64872922a75005db5d64da26ecfe11b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-11-27 10:13:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ee225cce66bb389bc9e2c2ffdcb25d128781195b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 10:14:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e77873ae6ef6996c293a63b8e85c1c02361c27a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 12:53:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2db1be96f344dca823628d5a2f2c640\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4eed75c00bb4ef6b316ab879c36c64b : SUCCESS in 3m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a0a47c1f622f4724ad71a348483208af : SUCCESS in 1h 50m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5388cbf8f237402a9f8f4c7ac6618160 : SUCCESS in 2h 35m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e34a297892ff4005acac34c6111fb071 : SUCCESS in 2h 15m 09s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/5c51f5894ee6408d88f7eec200265db6 : SUCCESS in 1h 14m 23s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/012262b77c17406689ee24e1da90e10b : SUCCESS in 2h 00m 14s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/6170e3d5622e4a3db06158af5916f6de : SUCCESS in 2h 36m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91862c25b85c4e458d5e3c92a1b40caa : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":3},{"id":"e92bd053f5da78d96d85057aceaa7644db78d8a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 12:53:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4b97bdbc804aab5f21d3094b1537a310bbe3202c","revisions":{"897388a03bf7ba03b900e8645aa7c9c86f5190a2":{"kind":"REWORK","_number":1,"created":"2024-11-22 16:03:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/36/936036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/36/936036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/36/936036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/36/936036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/36/936036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/36/936036/1"}}},"commit":{"parents":[{"commit":"20e6666c816a51dbd89543280415258229965394","subject":"Change the default image in the NDR tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/20e6666c816a51dbd89543280415258229965394"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-22 13:34:49.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-22 16:03:01.000000000","tz":0},"subject":"Use the admin client for the QoS tag tests","message":"Use the admin client for the QoS tag tests\n\nBy default, the network QoS policies can be created, updated\nand deleted only by an administrator. This patch changes the client\nused for these tests.\n\nRelated-Bug: #2037002\nChange-Id: I37f46cf8f7991d49966d5e4a4598ed17ae7d392a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/897388a03bf7ba03b900e8645aa7c9c86f5190a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/897388a03bf7ba03b900e8645aa7c9c86f5190a2"}]},"branch":"refs/heads/master"},"451d1b8b4ecf4d7eec85b670a0ca9c8703756722":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-22 16:13:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/36/936036/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/36/936036/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/36/936036/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/36/936036/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/36/936036/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/36/936036/2"}}},"commit":{"parents":[{"commit":"4ffa1e660682ad5af0d09a5894d7acae2cdd0299","subject":"Improve the exclude file list for the stadium projects jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4ffa1e660682ad5af0d09a5894d7acae2cdd0299"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-22 13:34:49.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-11-22 16:13:58.000000000","tz":0},"subject":"Use the admin client for the QoS tag tests","message":"Use the admin client for the QoS tag tests\n\nBy default, the network QoS policies can be created, updated\nand deleted only by an administrator. This patch changes the client\nused for these tests.\n\nRelated-Bug: #2037002\nChange-Id: I37f46cf8f7991d49966d5e4a4598ed17ae7d392a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/451d1b8b4ecf4d7eec85b670a0ca9c8703756722"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/451d1b8b4ecf4d7eec85b670a0ca9c8703756722"}]},"branch":"refs/heads/master","description":"Rebase"},"4b97bdbc804aab5f21d3094b1537a310bbe3202c":{"kind":"REWORK","_number":3,"created":"2024-11-23 13:01:31.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/36/936036/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/36/936036/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/36/936036/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/36/936036/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/36/936036/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/36/936036/3"}}},"commit":{"parents":[{"commit":"4ffa1e660682ad5af0d09a5894d7acae2cdd0299","subject":"Improve the exclude file list for the stadium projects jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4ffa1e660682ad5af0d09a5894d7acae2cdd0299"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-11-22 13:34:49.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2024-11-23 13:01:31.000000000","tz":0},"subject":"Use the admin client for the QoS tag tests","message":"Use the admin client for the QoS tag tests\n\nBy default, the network QoS policies can be created, updated\nand deleted only by an administrator. This patch changes the client\nused for these tests.\n\nRelated-Bug: #2037002\nChange-Id: I37f46cf8f7991d49966d5e4a4598ed17ae7d392a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4b97bdbc804aab5f21d3094b1537a310bbe3202c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/4b97bdbc804aab5f21d3094b1537a310bbe3202c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
