)]}'
{"id":"openstack%2Fneutron-tempest-plugin~936045","triplet_id":"openstack%2Fneutron-tempest-plugin~master~Ia019cc1c1db65f8206727c4184561c44fc847395","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-12-17 00:50:19.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-12-17 00:50:19.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-12-16 22:00:57.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ia019cc1c1db65f8206727c4184561c44fc847395","subject":"Deprecate options to enable/disable plugin tests","status":"MERGED","created":"2024-11-22 17:25:57.000000000","updated":"2024-12-17 00:51:18.000000000","submitted":"2024-12-17 00:50:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"936045","meta_rev_id":"df36637210827ff0386c292458a6ce38c164e9f2","_number":936045,"virtual_id_number":936045,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-17 00:50:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2024-11-23 13:35:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2024-12-16 08:22:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2024-12-16 22:00:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2024-12-16 22:00:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-22 17:29:47.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-11-22 20:43:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-22 22:02:33.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2024-11-27 15:11:50.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-12-16 08:22:29.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-12-16 22:00:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"273070e04dbdf867271d13d671090c8cfd59d4e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 17:25:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62db194d8dedba875b181da7da180eba1534b31e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-11-22 17:29:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"adea18f023636b9b78e7dcc3f301f8e1283f6abe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 17:34:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"61d0254141e18781cad065afc7075dca3eac2299","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 17:35:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7cd596cdff2f1091a9a83f32fa1ce70ce8d4aded","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 20:43:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27105c3f3abb45f680698dae99a3267c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dd9c574c64c445292c3f68db5cc3732 : SUCCESS in 3m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a7aa735e28134336a38e1188dd306c25 : SUCCESS in 1h 20m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f68393569a5e427db15268b6b07d4139 : SUCCESS in 1h 52m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fa6b3101fbc34866b1873b6eefae9da5 : SUCCESS in 2h 41m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e8849b2ae3194592b8f09a0f8caf1043 : SUCCESS in 36m 55s (non-voting)\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/37125c479ddf4cfb9cc413b3fc68abd4 : SUCCESS in 1h 25m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/2b77b5df040a4b18ae5bd2e6e516fe35 : SUCCESS in 1h 30m 16s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/8529138a89ab4442a51d251516d3d299 : FAILURE in 1h 23m 04s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/7bed4828543c4545a5c5942ca2118f3f : SUCCESS in 28m 55s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/4b4461faa22b4ab884438bd7d9cee90e : SUCCESS in 1h 16m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/5ccaaed5b5f542db996121271243575b : SUCCESS in 1h 22m 23s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/732dc3bb6dc94878bac97e96bdeb322a : SUCCESS in 2h 06m 19s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/97efe29d04864f948fe2ce83479e20ee : SUCCESS in 34m 06s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/ff7aa09807cd4b408395e748cd552d2d : SUCCESS in 2h 06m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/f10bce3b03c34946b81f69f0e0552709 : SUCCESS in 1h 51m 56s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/6f7081bcdb19430ea92d1159d956eb9f : SUCCESS in 1h 47m 25s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/ed3d631bfad049f883cc916fbffe5a26 : SUCCESS in 25m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/036f88ce04414b2abb4863070751610f : SUCCESS in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb4b71172e2d4cb781855bfd7475444c : SUCCESS in 2m 13s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/814f375923544e2db2aa5731f330c094 : SUCCESS in 1h 47m 44s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/ba9f45daede948a0b7b4cab52b78ee72 : SUCCESS in 1h 33m 33s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/1143f156630f411596d1117ff01fbb9a : SUCCESS in 1h 37m 56s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/94a834b8e5114dd899ed4e846bd10696 : SUCCESS in 1h 39m 54s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9ed428a986214021b5e6eb4693f4580f : SUCCESS in 1h 30m 43s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/b45f952e2bf948a2ba33d1b7a48c9981 : SUCCESS in 1h 18m 45s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/65de70b6f87748eb9bcc7b6b202d7dfb : SUCCESS in 1h 28m 14s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/ec469c11dfa5436889dda58514a4c68f : SUCCESS in 1h 38m 11s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/19c6d90646524628a736eac0e21fe05a : FAILURE in 59m 31s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/f28695d73db3495f8d31afeb5b2ce1f4 : SUCCESS in 1h 00m 11s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/e296915e917b4c1c94869a2f67ff679d : SUCCESS in 50m 00s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/f80e264aa5a44f368338c1f33075ff30 : SUCCESS in 58m 11s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/02cdcf5b1e944368b09cd637281cad7f : SUCCESS in 51m 33s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/28591cfd1743464f9f979d2a7f14084a : SUCCESS in 45m 47s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/67791d83960843fba21bac306123293d : SUCCESS in 22m 27s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/40569cacfbbb45219ef543baf0b085fb : SUCCESS in 46m 06s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/4c2f5a700df745a591da232b7491a056 : SUCCESS in 42m 55s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/9d869958e3704fc098841b763b18f8fb : SUCCESS in 37m 28s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/d5af5d4e0b504d40a5ff08075f1d78f0 : SUCCESS in 54m 53s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/e06f81db88294426939a646220cd4e2f : SUCCESS in 38m 02s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/ea267fbe0f744681b9ba52b5c1c290ac : SUCCESS in 47m 58s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/ed2f98a257d84717b4d76e1ccaa70940 : SUCCESS in 1h 34m 51s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/ddd7b2550cea496cb596c1f9e51f7d8b : SUCCESS in 1h 31m 31s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/4adbf2d9a88e4b0fbc947e1937dccbc6 : SUCCESS in 1h 28m 55s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/8a31abe90df7442a86302d1928c773ab : SUCCESS in 1h 39m 13s","accounts_in_message":[],"_revision_number":2},{"id":"0c8de309e9efb890512b20f5d6d7594ca1518a61","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-11-22 22:02:33.000000000","message":"Patch Set 2: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/b2ac806d5d914264880d6d88a7e5d25b\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/b0c0648849324d6bb7b943c4612344b1 : SUCCESS in 4h 26m 31s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/df4b19f650b646a58da86fcd04df7699 : SUCCESS in 3h 22m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9b1e3674bf70151b4517ca647c9375d3e3293adf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-23 08:18:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"79747c844ddf80a224a93794a041869f1b088096","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-23 10:46:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dda100100f04dd49860893ea437804d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0188bc796634b1b919280f7b81063e1 : SUCCESS in 2m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9191b2620d474192983d5757a4958d26 : SUCCESS in 1h 50m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/42b30e5a20004b27863f8a18b9ee1148 : SUCCESS in 1h 35m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c624c0a96b6446c1aeb7650a1dbccb0a : SUCCESS in 2h 26m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/692dea787d334c51940dd64ca9684d66 : SUCCESS in 34m 15s (non-voting)\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/98cd570d8f9944d1a6454c3f9fad7d62 : SUCCESS in 58m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/aab769dca6c848048b2daea22beb261a : SUCCESS in 1h 43m 35s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/042a15f51d8648ccbf30b8ad907de7dc : SUCCESS in 1h 11m 15s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/be8be054fbdc44aeb128c3b2158670bb : SUCCESS in 34m 13s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/cf61c0491587468ca5f10db3e0340390 : SUCCESS in 1h 06m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/938d4f56bb0a4ad9b22167a2079f649f : SUCCESS in 1h 37m 36s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/d1295803a6e14bf092030c72b26e0191 : SUCCESS in 1h 27m 18s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/057ee96d649a4fbeaec71e0cbcb3dc20 : SUCCESS in 35m 01s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/8cedb278970e4f799a8133ac09f451d6 : SUCCESS in 1h 54m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/b6341eda0c364a50b68d3f943a2f5d74 : SUCCESS in 1h 46m 40s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/6d0c8d17e77e4893a416de6a7f22d02f : SUCCESS in 1h 37m 01s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/c70667d4b81e4714a8bcf172b7575758 : SUCCESS in 30m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a53c6b86dab141dbbb91b01f3808f22b : SUCCESS in 2m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fffcc1c07cd4f0dbf88b2d760086468 : SUCCESS in 2m 23s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/0002b82118724ee4b659e50bbff725f2 : SUCCESS in 1h 42m 59s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/c63aebfad48341bfbd932694284a935d : SUCCESS in 1h 16m 33s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/31f5cf1f88834491a0ce2be1b62bb047 : SUCCESS in 1h 47m 20s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/3306a267b4574812868bed3fc67fdf7b : SUCCESS in 1h 47m 26s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/df666d6d30634f3fba8626c9f11e1177 : SUCCESS in 1h 44m 03s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/d8d0e166d4274252ac0a7bf16ed6a9d5 : SUCCESS in 1h 28m 58s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/e8af36dde23348ca803f4988fed930c9 : SUCCESS in 1h 23m 54s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/2323492f01ed48a4ac854f0f5f5a7c7f : SUCCESS in 1h 51m 01s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/304da96efc6142239a5222fe40c7c152 : SUCCESS in 48m 32s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/77cfaa3080b54e0eba28c4245f28684e : SUCCESS in 30m 29s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/9358dec60f794b0b89d8636e492bb468 : SUCCESS in 55m 20s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/2a4dd3da18a14da3b2a47b667a71eddf : SUCCESS in 1h 04m 38s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/fe80e5c4167c4d4aabd29cf3f4f634d8 : SUCCESS in 42m 32s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/3bf710a5d3b940a888126557d456918f : SUCCESS in 23m 52s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/262e792a266d48dcab115acf371130fc : SUCCESS in 50m 52s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/a0bea2adb6144e7fb2ee5b648cb2b838 : SUCCESS in 29m 02s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/54adc7a94c684bc5b757bce035118ece : SUCCESS in 45m 13s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/737581b4b00f49549414745e0162f1ae : SUCCESS in 26m 43s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/30b32d784ebf47f2b079eb20e2987e59 : SUCCESS in 52m 30s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/4afe8a8fde7f4ddbad0d6ec8edc330f1 : SUCCESS in 34m 59s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/884d6021d3894cc99d624ea748f7c7ed : SUCCESS in 32m 55s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/50ab19d357184ea7bb12183c3d833c19 : SUCCESS in 1h 14m 22s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/6dfcd7a129d741dc81482408ba3ed807 : SUCCESS in 42m 04s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/a7e70e5d715c4a6bbe43338fa5b99a9f : SUCCESS in 1h 22m 40s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/76206318399a41cf8024d4cb1b70d321 : SUCCESS in 1h 38m 16s","accounts_in_message":[],"_revision_number":3},{"id":"c9caba29bfb329bc02e53fb84f529a17585ef4e0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-11-23 13:35:16.000000000","message":"Patch Set 3: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/3e6e0f88ef814ead9c4bccd7e3926f20\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/71310db4734f4df882762ed651f9e226 : SUCCESS in 4h 18m 21s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/15292fae82ae4cdd9621e1f704275755 : SUCCESS in 3h 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e37f1c1e71b4955bd5513e7ac7d17e7177f2afa9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-11-27 15:11:50.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d906bca5de45637f93b5ec4f38ea52649e2f1d39","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-27 16:00:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8606af9c9a5c9d602c4d42ecb293fb6091abaad8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-12-16 08:22:29.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8735b349b292b7194aefaa3f359c2e16054ce1b4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-12-16 22:00:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dc10490a37a28e5f6e1769b77ea5fa2293ba6e45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 22:01:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0e6fd3ed796f43a4cad3c85250dca3a468898ef7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 00:50:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcd94340e0bd4759ba1db74532a85e39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4563d5bb68d4e058aabe8d04fcd72ba : SUCCESS in 2m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7e324c36a2084865bdc42e3cdb605dbf : SUCCESS in 1h 16m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c604c54704bc4183bab930ba8b9dea57 : SUCCESS in 1h 51m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c7875185634e48cb99f415494af68956 : SUCCESS in 1h 26m 28s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/011d5590f1174b6db9b00fd3dbe0a281 : SUCCESS in 1h 34m 13s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/cb72fbc5bd9a4784ae313d4d5cd5234a : SUCCESS in 1h 27m 50s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/0b0ef782b2a24e2eaf482a4b13ce21ce : SUCCESS in 1h 56m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be055a31fdc24a56bf77ba94ea404688 : SUCCESS in 2m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51ef4b0acae04554aaf331383248e4a0 : SUCCESS in 2m 43s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/662874899fdb47e6aab70919a86ba8a2 : SUCCESS in 1h 31m 20s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9feaaadafe784197a07b1bcf121e2593 : SUCCESS in 1h 30m 26s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/6e8de1cf5404420ab61b93ba69dcd7a1 : SUCCESS in 45m 29s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/d4502de157bc484eb24ec30321d914ac : SUCCESS in 33m 58s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/c01df565bba347ffa42f1a2f3ac09d83 : SUCCESS in 37m 42s","accounts_in_message":[],"_revision_number":3},{"id":"20cb7058ab9a2273cb6d8274dc050476b73dc436","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 00:50:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"df36637210827ff0386c292458a6ce38c164e9f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 00:51:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/938f11a851ca4d82975673beb8021e91\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2004cefde0f94bb1b7497291e02ed12c : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f8d254746e0f126fbd7e74011d7612967cbad61d","revisions":{"e7fa5f7ae8f8c287359a414f61b94c3aba4e3118":{"kind":"REWORK","_number":1,"created":"2024-11-22 17:25:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/936045/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/936045/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/936045/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/936045/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/936045/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/936045/1"}}},"commit":{"parents":[{"commit":"20e6666c816a51dbd89543280415258229965394","subject":"Change the default image in the NDR tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/20e6666c816a51dbd89543280415258229965394"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 17:24:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 17:24:14.000000000","tz":540},"subject":"Deprecate options to enable/disable plugin tests","message":"Deprecate options to enable/disable plugin tests\n\nThese were added to merge plugin tests into neutron-tempest-plugin in\nTrain and haven\u0027t been necessary since Stein was transitioned to EOL.\n\nDeprecate these options so that we can remove these right after 2025.1\nrelease.\n\nChange-Id: Ia019cc1c1db65f8206727c4184561c44fc847395\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e7fa5f7ae8f8c287359a414f61b94c3aba4e3118"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/e7fa5f7ae8f8c287359a414f61b94c3aba4e3118"}]},"branch":"refs/heads/master"},"cc5b769fc2c48e2eea0254decf21f6ea46bca3e5":{"kind":"REWORK","_number":2,"created":"2024-11-22 17:34:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/936045/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/936045/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/936045/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/936045/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/936045/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/936045/2"}}},"commit":{"parents":[{"commit":"20e6666c816a51dbd89543280415258229965394","subject":"Change the default image in the NDR tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/20e6666c816a51dbd89543280415258229965394"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 17:24:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 17:34:50.000000000","tz":0},"subject":"Deprecate options to enable/disable plugin tests","message":"Deprecate options to enable/disable plugin tests\n\nThese were added to merge plugin tests into neutron-tempest-plugin in\nTrain and haven\u0027t been necessary since Stein was transitioned to EOL.\n\nDeprecate these options so that we can remove these right after 2025.1\nrelease.\n\nChange-Id: Ia019cc1c1db65f8206727c4184561c44fc847395\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cc5b769fc2c48e2eea0254decf21f6ea46bca3e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/cc5b769fc2c48e2eea0254decf21f6ea46bca3e5"}]},"branch":"refs/heads/master"},"f8d254746e0f126fbd7e74011d7612967cbad61d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-11-23 08:18:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/936045/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/45/936045/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/936045/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/936045/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/936045/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/45/936045/3"}}},"commit":{"parents":[{"commit":"3261fb3ae11084ac8f959ef8dd2392e00d630662","subject":"Merge \"Improve the exclude file list for the stadium projects jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/3261fb3ae11084ac8f959ef8dd2392e00d630662"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 17:24:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-23 08:18:47.000000000","tz":0},"subject":"Deprecate options to enable/disable plugin tests","message":"Deprecate options to enable/disable plugin tests\n\nThese were added to merge plugin tests into neutron-tempest-plugin in\nTrain and haven\u0027t been necessary since Stein was transitioned to EOL.\n\nDeprecate these options so that we can remove these right after 2025.1\nrelease.\n\nChange-Id: Ia019cc1c1db65f8206727c4184561c44fc847395\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f8d254746e0f126fbd7e74011d7612967cbad61d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/f8d254746e0f126fbd7e74011d7612967cbad61d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
