)]}'
{"id":"openstack%2Fneutron-tempest-plugin~936048","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I5d13f9146b00b2fcd5c2e96071ab914b8a56a089","project":"openstack/neutron-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-25 15:44:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5d13f9146b00b2fcd5c2e96071ab914b8a56a089","subject":"Remove transition logic for old tempest option renaming","status":"MERGED","created":"2024-11-22 17:34:38.000000000","updated":"2024-11-25 15:44:43.000000000","submitted":"2024-11-25 15:44:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936048","meta_rev_id":"22c834bd6903d9f6d3b362309bac6a716e8f8f16","_number":936048,"virtual_id_number":936048,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-25 15:44:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-11-25 13:40:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-11-22 22:15:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2024-11-23 21:39:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-11-25 13:40:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-22 19:59:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-22 22:15:31.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2024-11-23 21:39:40.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-11-25 13:40:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"5ee629eea34ad529d91e7c2271343ada70039088","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 17:34:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fd440562fcbeb949becdbad855197ed413ec7a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 19:59:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25825eca0bf74ecbb16efb39be30e6fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79f528f385084a099bd86558fffa0d40 : SUCCESS in 4m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/27d3f06670f741ec8352d92cf99276c6 : SUCCESS in 1h 57m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e5646a129e10409e837416ba4b8f3573 : SUCCESS in 1h 35m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d548809f95f54f148f8c56cc2f3b0d3d : SUCCESS in 1h 46m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/11723cbd6ee043cf933f95d28f448640 : SUCCESS in 31m 38s (non-voting)\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/0c4408640c934f679f230b2592484f1d : SUCCESS in 1h 22m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/f15dd39863514cb0ad0e161393993233 : SUCCESS in 1h 20m 05s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/006df8ed3f1747928d2122285d741c24 : SUCCESS in 1h 33m 42s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/81f7a4e9aef943ab8d9a4cde5eaf12e2 : SUCCESS in 31m 07s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/08baa83445d74a5a922048a2f75275e2 : SUCCESS in 1h 10m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/445af652eec7487ba47a56cd0e87042a : SUCCESS in 1h 21m 22s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/b2e6545f394746bebfe36724d73c7e5b : SUCCESS in 1h 41m 44s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/2d95512883034190b688787a2e109fe6 : SUCCESS in 28m 59s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/151341df4833464ea07b3da81aa9fdf2 : SUCCESS in 1h 16m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/0f3015800dd04b89921aac0b6be56dd5 : SUCCESS in 1h 37m 19s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/a5bd6fb5b3f64302b26de5282f51a16c : SUCCESS in 1h 38m 43s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/538e057a38754acc9356698d1e4548c3 : SUCCESS in 31m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67436041d2ae4305ba8e17a02ecdb996 : SUCCESS in 3m 04s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/3f54e79007204521acc44ffe2479efb0 : SUCCESS in 1h 39m 28s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/f2ee0ee6509945169a8dcc309cf9d9e6 : SUCCESS in 1h 20m 13s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/76e4f4b1e1a9422fa5090ccd6576e6e2 : SUCCESS in 1h 44m 04s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/494e2dfd186c4852a0c2a84bce990263 : SUCCESS in 1h 45m 18s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/ac61df369a0c470687b56443dafade45 : SUCCESS in 1h 39m 44s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/d24637bd947145ac9c1af1125a27c186 : SUCCESS in 1h 17m 56s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/82b047acbb8442d8b56eed1c1a78f8e5 : SUCCESS in 1h 13m 38s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/8bb5666aeb02467ba1cc32747b3ca5ab : SUCCESS in 1h 28m 23s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/a671f1911f0e4f2ab838b70c2af09d87 : SUCCESS in 53m 59s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/a4ae94f194134543bbc175b1a232b5bd : SUCCESS in 53m 18s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/b0a2c035972248cca51cd932eca34cae : SUCCESS in 47m 10s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/b733a9919b8b4a1e8eadc8913912bae0 : SUCCESS in 43m 37s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/021fe4489b5e48ae97bb0b2053c101bf : SUCCESS in 47m 53s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/1a37b6469f534e94ad6dd4081afe821b : SUCCESS in 52m 23s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/af3b426a6b834d79ab88b1028643c619 : SUCCESS in 43m 43s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/d2a20fed90784d5c992eef2367e982ae : SUCCESS in 47m 06s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/2feef31a5bbb49599143de47d20ec110 : SUCCESS in 45m 41s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/c29531d2ca5f4ca38ae66028feeacd4a : SUCCESS in 46m 32s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/a995aa47dde04146bc312d711baa51da : SUCCESS in 53m 34s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/f97c25ffba774c0e9deeb69c5f1ec7b1 : SUCCESS in 49m 56s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/f7d3b166360c40bc88f290cb0325ff69 : SUCCESS in 30m 40s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/46874180fbbd462cabf923bc07ec1511 : SUCCESS in 1h 14m 16s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/93a95a20fe904099baa15bf809a3a2e0 : SUCCESS in 1h 19m 00s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/4f70114ad65840f7af365e446df300db : SUCCESS in 1h 18m 00s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/705c335e773c4c0d80f17420bdb57765 : SUCCESS in 1h 26m 35s","accounts_in_message":[],"_revision_number":1},{"id":"15abf5c073d00cf226ea08cb22adec3f8b85f7ab","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-11-22 22:15:31.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/fadbb9d9da624c39b51779a01dfb6dab\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/1cc75072d1cf4564bf6835e0b1ac97d5 : SUCCESS in 4h 39m 43s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/d59359f306e94973b3f56b2f95f9e6aa : SUCCESS in 3h 36m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0439df6630b96e26b5bc9b1c6c8dac1cddb28ab9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-11-23 21:39:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9e5eda6c7017be2b31c2b046cf2d79650069da94","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-25 13:40:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"568b3c0b388dc4089190068b8af514ad51615068","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 13:41:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"79ee0b699c9dec9a00895332fb1f291b93d893dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 15:44:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a23c9df268ff4ec583ce28acf38dd598\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b303839aa31c41faa75826bbcb597291 : SUCCESS in 3m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9790b36656f24b31a0fd4ef1fc9cd302 : SUCCESS in 1h 12m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ad2c6016af8d4ee48261295a0b667fc8 : SUCCESS in 1h 54m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5297c5f64ae14ce7a412570fee22d141 : SUCCESS in 1h 41m 23s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/644522e26d7a47de92b3251dd4db93a2 : SUCCESS in 1h 23m 51s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/c63ea8dc017f4f8ea56310e8463dd1b0 : SUCCESS in 1h 36m 43s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/96df3577d723428ab277cd37575d61a7 : SUCCESS in 2h 00m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f807c7ac5bcd451da128e9a5711ce1a2 : SUCCESS in 4m 06s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/67f74527398c4abea71dd1652c149b38 : SUCCESS in 1h 57m 47s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9365aadf88d240049e316de41be885b4 : SUCCESS in 1h 32m 50s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/52adc136142048b79a20bf56c77fc003 : SUCCESS in 1h 01m 52s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/2a1dc50f279540ef9b024c755a4a9e07 : SUCCESS in 49m 32s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/8093ad63e34848bd8b4804f4d0f2b628 : SUCCESS in 56m 36s","accounts_in_message":[],"_revision_number":1},{"id":"22c834bd6903d9f6d3b362309bac6a716e8f8f16","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 15:44:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"938f2c7fc8ac23d65fa5d3e234f5d97727cdf093","revisions":{"938f2c7fc8ac23d65fa5d3e234f5d97727cdf093":{"kind":"REWORK","_number":1,"created":"2024-11-22 17:34:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/936048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/48/936048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/48/936048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/48/936048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/48/936048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/48/936048/1"}}},"commit":{"parents":[{"commit":"20e6666c816a51dbd89543280415258229965394","subject":"Change the default image in the NDR tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/20e6666c816a51dbd89543280415258229965394"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 17:33:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 17:34:24.000000000","tz":540},"subject":"Remove transition logic for old tempest option renaming","message":"Remove transition logic for old tempest option renaming\n\nThis logic was required to adopt to the renaming in tempest[1] which\nwas done in 12.0.0 release. Now the minimum tempest version required\nis much higher so we no longer need this compatibility consideration.\n\n[1] ed6e586b9f8f0ada10af7711f297afa01b2b7754\n\nChange-Id: I5d13f9146b00b2fcd5c2e96071ab914b8a56a089\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/938f2c7fc8ac23d65fa5d3e234f5d97727cdf093"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/938f2c7fc8ac23d65fa5d3e234f5d97727cdf093"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
