)]}'
{"id":"openstack%2Fneutron-tempest-plugin~936830","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I6c34e8af701102fb56ea8f02def8d129c4ac63dd","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"bug/2045549","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-12-02 16:58:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6c34e8af701102fb56ea8f02def8d129c4ac63dd","subject":"Bump cirros version to 0.6.3","status":"MERGED","created":"2024-12-02 12:59:48.000000000","updated":"2024-12-02 18:02:41.000000000","submitted":"2024-12-02 16:58:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936830-bug/2045549","meta_rev_id":"dbd6b3ae9245240b22a92f967efc901c86106aab","_number":936830,"virtual_id_number":936830,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-02 16:58:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-12-02 13:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-12-02 18:02:41.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2024-12-02 14:40:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2024-12-02 14:40:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-02 13:17:57.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-12-02 14:40:09.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-12-02 14:52:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-02 18:02:41.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"93d99f271f0c83081ddb14b7cd576ba39d412b94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-12-02 12:59:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50b05201869b2c06790c44800d69852f14c94e7e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-02 13:17:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"686a2c56fc66e19dce9fb0e6864bf48cacb4c5ab","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-12-02 14:40:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4a114be2b172668c2a87ca19c5f906a12b0f45e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 14:52:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15d9f0082acc4017ae6d44e8c592beaf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e40f7947fe54c08aa040cff8d8c3457 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acc2181339914c41b9950da08ae45e50 : SUCCESS in 3m 33s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/d7f97734f4ff4d30bc7a322c207b4c51 : SUCCESS in 1h 52m 08s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/41a39bf18ec240d9879589ea9804641b : SUCCESS in 1h 28m 37s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/12ca1d5ed7684f96b17e05a4549e23ca : SUCCESS in 51m 42s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/8824ea3da8eb447e949155d0bccf1a08 : SUCCESS in 1h 35m 24s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/67ca51eb2c304ea492b0edfc1a6ed19b : SUCCESS in 33m 23s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/3493c427d4fe4c43abd0906e1015f109 : SUCCESS in 59m 04s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/7784bbee1fec46a2b86eeb694e2df2f0 : SUCCESS in 50m 17s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/3f45db5461b44583b915a6d246bbe511 : SUCCESS in 43m 35s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/a6915c53347d4140be9283e677394b53 : SUCCESS in 46m 53s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/4cf05c72cf594600aba27d59d6d66581 : SUCCESS in 56m 23s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/91515786cd404cdb9fc194bd0aff898e : SUCCESS in 41m 11s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/4e7556034763449fb0830bd216788c20 : SUCCESS in 46m 08s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/befec528b1fa4c69b2ab5f2cda872b9c : SUCCESS in 57m 21s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/f02b95b3adb14dcbaee2ac6db336a837 : SUCCESS in 48m 22s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/1db95d79848e4e48b6a1f88404e03a89 : SUCCESS in 54m 43s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/ee562f65d96f4b48a01984c11873c12a : SUCCESS in 1h 00m 59s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/f83d62dd3cf6489da0603b8bd69cb86d : SUCCESS in 42m 08s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/a8bf173c8f7c41ba98ad3940292ba457 : SUCCESS in 1h 14m 50s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/38b110340d1d47b6b5abadf0d27f4457 : SUCCESS in 1h 23m 12s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/2558ae15484543299218f86f9933f855 : SUCCESS in 1h 34m 20s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/ecf86f0ef55c48839f316b5977d58de3 : SUCCESS in 1h 10m 28s","accounts_in_message":[],"_revision_number":1},{"id":"dd1592b53f0d300b589b570b254103c7870a9355","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 14:53:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d7e04989f268d79762d83d9fee2de708d30b6fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 16:58:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/880b6f49da2b4230bb41a372169acbbc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13ad7f67a6a24f2d875b994054803d7e : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd7bcd834c1b4149b469d47eab2a5dae : SUCCESS in 3m 20s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/9b93bd694fdb41d4984a2566473538b7 : SUCCESS in 1h 03m 17s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/4ecfc973870748a28b1b9b390f07e2cb : SUCCESS in 48m 19s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/f82a23d8e9a04a57b29e67e461304d14 : SUCCESS in 1h 00m 39s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/669397fb575f4564ba6197b6b0be1333 : SUCCESS in 1h 03m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3411662dd9d58fa10f132b8931d481f6c575b378","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 16:58:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dbd6b3ae9245240b22a92f967efc901c86106aab","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-12-02 18:02:41.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/6979d3729a3b4d31a88745b8ce4c2a05\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/7f94e2f45a354b9e9f9d4c3d426229f4 : SUCCESS in 4h 44m 35s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/3e912c147fc648f7940be82ac84e14a4 : SUCCESS in 3h 35m 17s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3bf7c1774d2b191b63edd7d7c299e72c7906075","revisions":{"d3bf7c1774d2b191b63edd7d7c299e72c7906075":{"kind":"REWORK","_number":1,"created":"2024-12-02 12:59:48.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/30/936830/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/30/936830/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/30/936830/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/30/936830/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/30/936830/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/30/936830/1"}}},"commit":{"parents":[{"commit":"04806eba26d1db02584a44bd4b5d7dce7a4edd91","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/04806eba26d1db02584a44bd4b5d7dce7a4edd91"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-12-02 12:57:47.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-12-02 12:57:50.000000000","tz":330},"subject":"Bump cirros version to 0.6.3","message":"Bump cirros version to 0.6.3\n\n0.6.3 includes some fixes, let\u0027s use it.\ndevstack already switched to it with [1].\n\n[1] https://review.opendev.org/c/openstack/devstack/+/930647\n\nRelated-Bug: #2045549\nChange-Id: I6c34e8af701102fb56ea8f02def8d129c4ac63dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d3bf7c1774d2b191b63edd7d7c299e72c7906075"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/d3bf7c1774d2b191b63edd7d7c299e72c7906075"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
