)]}'
{"id":"openstack%2Fneutron-tempest-plugin~937520","triplet_id":"openstack%2Fneutron-tempest-plugin~master~I3eef4fd4bbb6a56b4e22342e7fd57c8d1cc67023","project":"openstack/neutron-tempest-plugin","branch":"master","topic":"neutron_api_wsgi_default","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-12-12 07:08:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3eef4fd4bbb6a56b4e22342e7fd57c8d1cc67023","subject":"Test devstack patch","status":"ABANDONED","created":"2024-12-11 07:58:58.000000000","updated":"2024-12-12 07:08:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ab19d355dd20892b0216ec7031a2d990419c104c","_number":937520,"virtual_id_number":937520,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-12-11 10:38:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":1,"date":"2024-12-11 12:35:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-11 10:38:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-11 12:35:01.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"2bf620d38243720ba625c66738ae54bebb491d92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-11 07:58:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdcca5827476c4787eaf3f1d87f6e273ce8aabd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 10:38:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f646243f30c54288992c2ab24f1bd710\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f79153a42a8489cab669cc71452c56f : SUCCESS in 3m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/acd06ce61e004c8a94c060c4e85b294d : SUCCESS in 1h 52m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a4e31fef04a24427a026f5804187ae3c : SUCCESS in 1h 54m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/437b6123378c42498aca6c892cd90568 : SUCCESS in 2h 37m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/021ca13b45894e568ec4ca8c11d722d9 : SUCCESS in 25m 59s (non-voting)\n- neutron-tempest-plugin-openvswitch-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/784a4d3f1d554fe6952a07ad3f3bc9da : SUCCESS in 1h 42m 35s\n- neutron-tempest-plugin-ovn-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47ff1cdf473c4a5e9e5669d0bcb12490 : SUCCESS in 1h 39m 51s\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/fd7702fa8426411185b2bbd407d7b156 : SUCCESS in 1h 08m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/65ab7a60699c4acca2eb504e7cc48bcd : SUCCESS in 1h 16m 10s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/7c9d0efbb70349df913eb9130f2ceee4 : SUCCESS in 1h 39m 25s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/6ac086c1a8b148c5ae83eae567ad0822 : SUCCESS in 28m 27s\n- neutron-tempest-plugin-openvswitch-2024-1 https://zuul.opendev.org/t/openstack/build/f6abff86425d4db58c9d21d2d9e1edaf : SUCCESS in 1h 14m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-1 https://zuul.opendev.org/t/openstack/build/602333872f444af89261256b4b71b193 : SUCCESS in 1h 54m 02s\n- neutron-tempest-plugin-ovn-2024-1 https://zuul.opendev.org/t/openstack/build/de035989d68b47ccb5b50ecd6445fec5 : SUCCESS in 1h 27m 05s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/30e614ec334646d08cb37c35796f6bd5 : SUCCESS in 41m 11s\n- neutron-tempest-plugin-openvswitch-2024-2 https://zuul.opendev.org/t/openstack/build/ae906fdfe90e4907acf7a50a25496fcf : SUCCESS in 1h 56m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2024-2 https://zuul.opendev.org/t/openstack/build/1db3aebe47184dd7962fe425274aa467 : SUCCESS in 1h 49m 52s\n- neutron-tempest-plugin-ovn-2024-2 https://zuul.opendev.org/t/openstack/build/1b9011f0103d463293cb938afcbbf241 : SUCCESS in 1h 53m 31s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/4355a46310e549f0a37d3a66525e03da : SUCCESS in 32m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/347da8bf25c64f87bfb619b395b6f7c7 : SUCCESS in 3m 32s\n- neutron-tempest-plugin-sfc https://zuul.opendev.org/t/openstack/build/abec1b0e8cc540b7adf1013dc0d02f3f : SUCCESS in 2h 16m 33s\n- neutron-tempest-plugin-sfc-2023-2 https://zuul.opendev.org/t/openstack/build/c0e367cc938f4360b38ecdd89555323c : SUCCESS in 1h 34m 55s\n- neutron-tempest-plugin-sfc-2024-1 https://zuul.opendev.org/t/openstack/build/a2bd774f3ba04d578f46ec4a6efa0493 : SUCCESS in 1h 25m 36s\n- neutron-tempest-plugin-sfc-2024-2 https://zuul.opendev.org/t/openstack/build/fb15d3ee1bd644bd807c63ee699149f2 : SUCCESS in 1h 55m 01s\n- neutron-tempest-plugin-bgpvpn-bagpipe https://zuul.opendev.org/t/openstack/build/7dbbc44964664b1b932628da0019671b : TIMED_OUT in 2h 03m 13s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2023-2 https://zuul.opendev.org/t/openstack/build/16b256bf846b45f1a42beb2fa6232276 : SUCCESS in 1h 22m 34s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-1 https://zuul.opendev.org/t/openstack/build/d6757a0cc6114b38b244460ee28b3c0c : SUCCESS in 1h 32m 27s\n- neutron-tempest-plugin-bgpvpn-bagpipe-2024-2 https://zuul.opendev.org/t/openstack/build/41494e69941d4ba78fb8996197fb71ab : SUCCESS in 1h 47m 22s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/f1a0c963f2b14b198e495ed731ae751e : SUCCESS in 1h 14m 12s\n- neutron-tempest-plugin-dynamic-routing-2023-2 https://zuul.opendev.org/t/openstack/build/4564caffa4d745a7814620eaa312078e : SUCCESS in 56m 16s\n- neutron-tempest-plugin-dynamic-routing-2024-1 https://zuul.opendev.org/t/openstack/build/a3f36469f5584c659dad22adf94279af : SUCCESS in 1h 00m 04s\n- neutron-tempest-plugin-dynamic-routing-2024-2 https://zuul.opendev.org/t/openstack/build/62f819243daa4f46b0b2796ddffcfda3 : FAILURE in 1h 02m 23s\n- neutron-tempest-plugin-fwaas https://zuul.opendev.org/t/openstack/build/ccdbbd39478b4db8a0cfef8d65c9bc01 : SUCCESS in 25m 52s\n- neutron-tempest-plugin-fwaas-2023-2 https://zuul.opendev.org/t/openstack/build/49163e190cd4438a8eeaeb4a9e47ceb4 : SUCCESS in 51m 17s\n- neutron-tempest-plugin-fwaas-2024-1 https://zuul.opendev.org/t/openstack/build/707f1dad1deb440699f07a1e090ec473 : SUCCESS in 44m 51s\n- neutron-tempest-plugin-fwaas-2024-2 https://zuul.opendev.org/t/openstack/build/e8a7ec1a331c43c681d5d3941cf9b46a : SUCCESS in 52m 38s\n- neutron-tempest-plugin-vpnaas https://zuul.opendev.org/t/openstack/build/09f5c8b640f34b5a916e4f77e4182021 : SUCCESS in 49m 50s\n- neutron-tempest-plugin-vpnaas-ovn https://zuul.opendev.org/t/openstack/build/cf28ddb5f7624195932894a9e891e473 : SUCCESS in 54m 18s\n- neutron-tempest-plugin-vpnaas-2023-2 https://zuul.opendev.org/t/openstack/build/f90ac2d6aefd402c85b58389b69116d6 : SUCCESS in 45m 38s\n- neutron-tempest-plugin-vpnaas-2024-1 https://zuul.opendev.org/t/openstack/build/ac6b80dd181b47ea80ba5c737a4be212 : SUCCESS in 1h 07m 36s\n- neutron-tempest-plugin-vpnaas-2024-2 https://zuul.opendev.org/t/openstack/build/e5975bbc443a4f8c8d62f251729eb664 : SUCCESS in 45m 45s\n- neutron-tempest-plugin-tap-as-a-service https://zuul.opendev.org/t/openstack/build/55cb75cc5b6341ef8a18d644f0d47440 : SUCCESS in 1h 45m 00s\n- neutron-tempest-plugin-tap-as-a-service-2023-2 https://zuul.opendev.org/t/openstack/build/ae6917f9d0f64501ae5883d9a4aed961 : SUCCESS in 1h 19m 39s\n- neutron-tempest-plugin-tap-as-a-service-2024-1 https://zuul.opendev.org/t/openstack/build/f1f31e637a624e6396f0a38f45eecbe7 : SUCCESS in 1h 46m 46s\n- neutron-tempest-plugin-tap-as-a-service-2024-2 https://zuul.opendev.org/t/openstack/build/c15f7ba192834f4eb5b5da23f386eeae : SUCCESS in 1h 34m 54s","accounts_in_message":[],"_revision_number":1},{"id":"74e8702fe4f7473df73f87f7fc6de811d26bd43e","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2024-12-11 12:35:01.000000000","message":"Patch Set 1: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://softwarefactory-project.io/zuul/t/rdoproject.org/buildset/00858912372549598bb678dee2277bf9\n\n- openstack-meta-content-provider https://softwarefactory-project.io/zuul/t/rdoproject.org/build/3402ac79df1b4487a6a55019985edd3a : SUCCESS in 4h 34m 29s (non-voting)\n- whitebox-neutron-tempest-plugin-podified-multinode-edpm-deployment-crc-2comp https://softwarefactory-project.io/zuul/t/rdoproject.org/build/1d1cd41438d64fb7a4a328dcc3a10373 : FAILURE in 3h 40m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab19d355dd20892b0216ec7031a2d990419c104c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-12-12 07:08:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9e9e5a32d4460c42d59e0ef644c202c09c0a727","revisions":{"c9e9e5a32d4460c42d59e0ef644c202c09c0a727":{"kind":"REWORK","_number":1,"created":"2024-12-11 07:58:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/20/937520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-tempest-plugin","ref":"refs/changes/20/937520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/20/937520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/20/937520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/20/937520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-tempest-plugin refs/changes/20/937520/1"}}},"commit":{"parents":[{"commit":"c27cf6baed462feb6d527e03d28184037ff80abd","subject":"Remove all eventlet Neutron API jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c27cf6baed462feb6d527e03d28184037ff80abd"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-11 07:58:37.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-12-11 07:58:37.000000000","tz":0},"subject":"Test devstack patch","message":"Test devstack patch\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/937518\nChange-Id: I3eef4fd4bbb6a56b4e22342e7fd57c8d1cc67023\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c9e9e5a32d4460c42d59e0ef644c202c09c0a727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-tempest-plugin/commit/c9e9e5a32d4460c42d59e0ef644c202c09c0a727"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
